)]}'
{"id":"openstack%2Foctavia~739638","triplet_id":"openstack%2Foctavia~master~I3e8725eb77ea947e71d85ea406a60ed94c7bf971","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3e8725eb77ea947e71d85ea406a60ed94c7bf971","subject":"Switch from unittest2 compat methods to Python 3.x methods","status":"MERGED","created":"2020-07-07 01:53:14.000000000","updated":"2020-07-07 18:41:59.000000000","submitted":"2020-07-07 18:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739638-1594147117182-de16a12a","meta_rev_id":"dd4457f465192c65c5889b2439ab168cb5338071","_number":739638,"virtual_id_number":739638,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-07 18:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-07-07 08:08:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-07 16:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-07 16:45:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2020-07-07 08:08:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 08:08:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-07 16:45:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-07 18:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3dc04fced1d01b1c5e4522e95ff48e289afcd007","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-07-07 01:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab396fbab441482960da1fb4f6682bd8430e1337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 02:07:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa57fea98ddc4a7f8d48081138e122b5 : SUCCESS in 12m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aa8bb71713b42dca7061bd4e7bb3de0 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a906939b03d43c58232b389259ceeaa : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e24cacdb672c41f2b69c3f6861150313 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/114c44121abf496ea581c7b60d93f3b1 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fce74b899eb49ad9353ef20f95bfbdf : SUCCESS in 13m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a9da1fc57f74b13b01d1c421ccc3259 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/29a6932c439d421492a07f305db59cf0 : SUCCESS in 10m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b35dd2b9235047c79fc115405035933e : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ecd4e6c34f5a0258c32525efa18a8d5cbac65e97","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-07 08:08:30.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2\n\nI was a bit warry that replacing with assertCountEqual because it reads as if it just compares the length but it is not the case per the doc [1]. The doc [2] for assertItemsEqual also states that assertCountEqual is the new in Python 3.\n\n[1] https://docs.python.org/3/library/unittest.html#unittest.TestCase.assertCountEqual\n[2] https://docs.python.org/2/library/unittest.html#unittest.TestCase.assertItemsEqual","accounts_in_message":[],"_revision_number":1},{"id":"1d0515b744c7f31b4ba5482d192a330818af4221","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-07 16:45:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"385db3445699316ee61d6fae092ba6480bf42206","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 16:46:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6ebde4ca81c7e1be77d7ccbd2f4d1ee50186890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 18:38:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07427704e3af41b3a70d5bfb0466c364 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e3272d6848482c8c3b2747086d7cd7 : SUCCESS in 10m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14e04e0c1fda4283bd2193f754fd576b : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dec7e6973594a1798f8ebbe852a3306 : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a66eef73353468ebe18605f22a4d43c : SUCCESS in 15m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f77696e2eb42491bb8e3cbeddd74ff90 : SUCCESS in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fbda3d8aa5354ec78c02f50a9bd3b555 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3381175c3e94d75b1825676ed974048 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df84b19a493f46439f0fd35f2958b1e9 : SUCCESS in 54m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3146e3bdc7bd428dbb4a0d5beb9aefd6 : SUCCESS in 57m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1e4eab3259646e5a2a5f4beab862cc2 : SUCCESS in 1h 20m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/417d09095bb54cf3a11184e15cdc0528 : SUCCESS in 1h 43m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ba6dd7d59bb1557c90978c8292e870a06d7c94ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 18:38:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dd4457f465192c65c5889b2439ab168cb5338071","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 18:41:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d62da5076c420980c331f09edbf713 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"103653503cc641a59e6efd145567d98d7fbdb5b3","revisions":{"103653503cc641a59e6efd145567d98d7fbdb5b3":{"kind":"REWORK","_number":1,"created":"2020-07-07 01:53:14.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/38/739638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/739638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/739638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/739638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/739638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/739638/1"}}},"commit":{"parents":[{"commit":"f68e3aba4944a21eef9de2a395899f426f049e2b","subject":"Merge \"Remove amp_ssH_access_allowed option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68e3aba4944a21eef9de2a395899f426f049e2b"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-07 01:51:56.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-07 01:52:52.000000000","tz":480},"subject":"Switch from unittest2 compat methods to Python 3.x methods","message":"Switch from unittest2 compat methods to Python 3.x methods\n\nWith the removal of Python 2.x we can remove the unittest2 compat\nwrappers and switch to assertCountEqual instead of assertItemsEqual\n\nWe have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I3e8725eb77ea947e71d85ea406a60ed94c7bf971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/103653503cc641a59e6efd145567d98d7fbdb5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/103653503cc641a59e6efd145567d98d7fbdb5b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
