)]}'
{"id":"openstack%2Foctavia~739959","triplet_id":"openstack%2Foctavia~master~I76755b1dbae804e145d1b474f062fff58ec6a4c7","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I76755b1dbae804e145d1b474f062fff58ec6a4c7","subject":"Avoid trying to run jobboard conductor on error","status":"NEW","created":"2020-07-08 11:51:45.000000000","updated":"2021-06-08 14:21:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"126b6278bda1e5fb8080aa43401ff80ce4a8c83a","_number":739959,"virtual_id_number":739959,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-08 14:21:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-08 14:21:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 14:21:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-1,"date":"2021-06-08 14:21:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 10:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 18:35:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-23 15:08:54.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-23 23:57:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"60d112df5af79c976262d7f3ddedfeb27ded311a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-08 11:51:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32826c4148f2359227bf535f205214d96b343c50","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-08 12:58:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f39cc934b85dcc92cb56e3b1fb1c4fe31cd4cbd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-08 13:23:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52403933915b5cb699c340d62de0754ebc146e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 14:05:49.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa4cc9466049496faa27db12f5e49245 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec9417ea5263438c807e7d378a55b899 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a90e903074454d99bf966c9a824ef4b7 : FAILURE in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95e7abbc91848218c696ed46097ad23 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5547e6f0596444b19734e7e3717249ab : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeac93d1a2304f5ca0fa72ae133d58b8 : SUCCESS in 15m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c15cd1541f5f47f9a2fd82fe23aae1fd : SUCCESS in 7m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1e7e1817ad2747368f2654ba63ba0eda : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c1da9fef5e149c9bb5ccbd783d7bd4b : SUCCESS in 11m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7da1b20d7518457c8b1350f477b9abda : SUCCESS in 48m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2bac83b55a840d4a6086f6cb9143a22 : SUCCESS in 1h 43m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/057e9b1c9aa045148c6a7c3471f59630 : SUCCESS in 1h 03m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/adb126d96c344c0f86a80b851e63d84e : SUCCESS in 57m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f8daa5de0f746b286c29182f6f6a625 : SUCCESS in 1h 00m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b645fd2533642ef84a139f6e21c5985 : SUCCESS in 1h 35m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b70ed1549f884821b33a713d073f49fd : SUCCESS in 2h 00m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c68ef670a21458e9372aee6fea75259 : SUCCESS in 2h 05m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8eb2fbfbf4344400b8fe17ea5e9b2577 : SUCCESS in 1h 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dacd822b230d976a785ea8bfda2bbc0dbfc1a931","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-09 11:15:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a3202d84f24259cfcb4cc0751623aa52f1c5bd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 15:48:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21d681eb5cf14cca92778bc96054985a : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f48523961fbe4cc1b3b8a134c9af6758 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65abc777b4647199ec503dbfc2e81cb : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b126a3f1eae4a7c93154a5409691217 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf8c7c74137744dab040226d81ec855c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241751b1fcb54517bd1d657f4269b060 : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/90b93f28efcb4e7e9484ed0415fcd47f : SUCCESS in 8m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cae6345dcd74cd690e198b7c2ace0fa : SUCCESS in 9m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74d5dda56d4c4096b683a46a0f0d1f6d : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6417d0b13dbf413b8041570c898e99ea : SUCCESS in 46m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/191bff830c7348f184f7d50b3581f188 : SUCCESS in 1h 58m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0500d12cfd7d45e3882059a568d39aa6 : SUCCESS in 1h 06m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8ce39e5111f4850b588fa988230b66b : SUCCESS in 56m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f6e695a8ef041aa8cd52d53a6b17afa : SUCCESS in 54m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da5f8520541e45d69f6ed941adc3bfd8 : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/397570289c774da0b9f2eea350d11ef5 : SUCCESS in 2h 00m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/554fd30f8a00437e9a1890a4e23b72e4 : SUCCESS in 2h 01m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/197a650c28d94ea3b9b3cfce42bd1469 : SUCCESS in 1h 50m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/53fc4c2cc9f247dca7decac30eb58eb9 : SUCCESS in 1h 57m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"804b39403037830e1e3140e074a29146fe7aed89","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-09 21:48:55.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47318ad0f5f5e0573703c0ff80157cf4f5d23005","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-14 06:44:02.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"b0d9993a5d08f9b5052b6537696c19f926e6c559","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-27 09:03:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f4b8673a28d047219a5b2e315083ccc1fcf6a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 11:18:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf3ffb83eac44d8dafcefd19a59417c8 : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a718feb5ed5a45efb013747bc0688b73 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81df9bb8d3d54859be2bc1ada43e4d5b : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b73f51b90574e28802ba55ed191d0a7 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dad41dee0c0a4fe1966d00c2b3a1f4c1 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1457451781d460a899fe38bee952a59 : SUCCESS in 17m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f524c9891b7d474bb3c7738d3074a2da : SUCCESS in 8m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/18b983b8485545cd98708ca17729d76c : FAILURE in 9m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63246b0042d748e4be7b02cfd59e1569 : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2c94699373b46afa3e96a2c26f32ea3 : FAILURE in 9m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b38ccef3d3f4f80a56c3e3a71ea38cf : FAILURE in 10m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3a6533c65914426bd977799618f7aa9 : SUCCESS in 45m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/492cbc8bc0334dc385446bd80cf738cc : SUCCESS in 42m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29a9edfe6b5b421586b3086b4c715c84 : SUCCESS in 42m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4fdf7b4666d5495c87964e68a3ae8441 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/679c39d7d72c48f6a66c4edadd58fe31 : SUCCESS in 1h 20m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9215316457747a681910a574afbab2c : SUCCESS in 1h 04m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/844856dbbdac45e7a4da94cd6a9a3c6a : SUCCESS in 1h 22m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd8d159a97034fff9bb4c2c9b890aaa8 : SUCCESS in 1h 14m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e5492f3f48dd4240bd4e74d636cde1a3 : FAILURE in 11m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c0ea16d3d6214b669821bdf5b20c33b9 : FAILURE in 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4ede98b2ce9943485844b02494807353d54b1ec","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-03 08:02:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e0cba1c5a1dc35c5d685575605a3cbec4fc1b798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 10:23:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/537aa85b406a407d853d727fd661894b : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1e87463c7cd438b823b3ba81ef5cc2f : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5b14589f83460a903e4069f7287df7 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a5771b5e6f49fe84fb28830a891d76 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d4170f75ef24c7880ba14e2f6960288 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895803c7fafa41a7abe7b882ec5c70ac : SUCCESS in 13m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/065085a8f3ad4209b0da09c57c1157d9 : SUCCESS in 7m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/07b4b9cb871d43d68704316d8db946dd : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d596bd11dd9847a5846c2b8815eabb6e : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/252000d72fe14d7c83e355ec41ed93a8 : SUCCESS in 45m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1967cf2353fe4be2b25d2ac97bf58ef2 : SUCCESS in 43m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7478d894f4c54ee0924a4eaa1dae2cb3 : SUCCESS in 55m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1121942a681c414d9f180cc3a0f55d4d : SUCCESS in 52m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32be3515655c4560b72f3914c37fcf5c : SUCCESS in 42m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f0e94f4f0fe47019b9f93f76c235ff1 : SUCCESS in 1h 49m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf541dcd41fd474e872e2613afcd6ba9 : SUCCESS in 1h 38m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab2c36776ece4cf58eac1dec1c860760 : SUCCESS in 1h 35m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/01032070cfaf4026b2f5757233f733a4 : SUCCESS in 1h 29m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d19c7f79b0fb4a8ea81c49c6a5e4da2f : SUCCESS in 1h 16m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/89506fe592e447fe9f2f98533c1403b8 : SUCCESS in 1h 13m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e8d44109bb9f4bf08c73a07ea288b858 : SUCCESS in 1h 58m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c60469c883757705dc825ad492fdda4fec33d35","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-18 18:35:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9a4c63c35f1d56cc7a87aebd0bd78ac94b576bba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-23 15:08:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9015cef6a7d37006e732936c4463da62057d741d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-23 23:57:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3643f2deee1d62dab05bf2181db1503d9ece10db","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-24 07:15:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"211358b0cdaa1e7d38af4106744c02b0e3a1f423","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 12:40:58.000000000","message":"Patch Set 4:\n\nHi Michael, are you ok with Ann\u0027s answer?","accounts_in_message":[],"_revision_number":4},{"id":"fcf141c4318e1182b34a9368e481163e3ccc9502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:24:25.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"70f1a4418abb7a9211b78e7af6420c6099521bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:21:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8da8d4441e564b86b607c3c40185c33f : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2c041ce337f46bab66609e23ccaeb22 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a02c23a28d46e4a7d5844916fa9921 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d86e200c3ae477e840afdd30ad6d59a : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb76007288b147ef82735caa119f51c9 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57eb7d88ba004cae800bf01555a36b56 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d0451955eb4de4840a871bc3f20292 : SUCCESS in 14m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5cb1e8816e794c38ad9c71d5f5c2a4f1 : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0c59de9dfe534c838def8438365d67d3 : SUCCESS in 11m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f1b9b399eb04495e8519ac9d08a4077a : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e5049b8c4024e5ebd3cea44cb6d5ae0 : SUCCESS in 10m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e30e389269784526ac132d49109fba41 : SUCCESS in 1h 27m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9556f3f2fe874b479ad5f470746e0b94 : SUCCESS in 1h 43m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a6afe5b1bc9481ba4edb69f1e8640bc : SUCCESS in 44m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4306f88f2be3439d9adf96ff851af9ef : SUCCESS in 45m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d70a2ec73a784c0888499f0c5efc6d15 : SUCCESS in 1h 05m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2579969601cb478e9acac9f16ee79e9b : SUCCESS in 1h 28m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c39e5e2faed4b8581bb9e01b742b859 : FAILURE in 17m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3ab5b305b29b45da9b95be1a93056546 : SUCCESS in 1h 36m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/52e548e158ff4cf197557c8d26abeed5 : SUCCESS in 1h 55m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d2ed4859c0054037b8e9f606ce31f939 : SUCCESS in 1h 40m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e1bd756d459c4674a9facbe8614da40c : FAILURE in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7b42782652f0106d0dc7dac5c6461b214f1e4377","revisions":{"992c0b7b6bcf393090459d0556b11cae0dd92069":{"kind":"REWORK","_number":1,"created":"2020-07-08 11:51:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/59/739959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/739959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/739959/1"}}},"commit":{"parents":[{"commit":"bf196db6bc535de616aac4947405fedeec3d301b","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf196db6bc535de616aac4947405fedeec3d301b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 11:47:22.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 11:47:22.000000000","tz":240},"subject":"Avoid trying to run jobboard conductor on error","message":"Avoid trying to run jobboard conductor on error\n\nIf run_conductor failed due to some error like redis uninstalled/\nredis server unavailable, log exception and do not retry in infinite\nloop.\n\nChange-Id: I76755b1dbae804e145d1b474f062fff58ec6a4c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/992c0b7b6bcf393090459d0556b11cae0dd92069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/992c0b7b6bcf393090459d0556b11cae0dd92069"}]},"branch":"refs/heads/master"},"c95d7d15ca315d03832aefd01542340fddc48b89":{"kind":"REWORK","_number":2,"created":"2020-07-09 11:15:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/59/739959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/739959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/739959/2"}}},"commit":{"parents":[{"commit":"bf196db6bc535de616aac4947405fedeec3d301b","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf196db6bc535de616aac4947405fedeec3d301b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 11:47:22.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-09 11:15:40.000000000","tz":0},"subject":"Avoid trying to run jobboard conductor on error","message":"Avoid trying to run jobboard conductor on error\n\nIf run_conductor failed due to some error like redis uninstalled/\nredis server unavailable, log exception and do not retry in infinite\nloop.\n\nChange-Id: I76755b1dbae804e145d1b474f062fff58ec6a4c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c95d7d15ca315d03832aefd01542340fddc48b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c95d7d15ca315d03832aefd01542340fddc48b89"}]},"branch":"refs/heads/master"},"e3250ecbbbf902471ab04f99f614b3e1b9e62d16":{"kind":"REWORK","_number":3,"created":"2020-08-27 09:03:28.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/59/739959/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/739959/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/739959/3"}}},"commit":{"parents":[{"commit":"0b1d8dd5e76edc4b3058a36b47f6c059f970e516","subject":"Merge \"Update devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1d8dd5e76edc4b3058a36b47f6c059f970e516"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 11:47:22.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-08-27 08:55:56.000000000","tz":240},"subject":"Avoid trying to run jobboard conductor on error","message":"Avoid trying to run jobboard conductor on error\n\nIf run_conductor failed due to some error like redis uninstalled/\nredis server unavailable, log exception and do not retry in infinite\nloop.\n\nChange-Id: I76755b1dbae804e145d1b474f062fff58ec6a4c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3250ecbbbf902471ab04f99f614b3e1b9e62d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3250ecbbbf902471ab04f99f614b3e1b9e62d16"}]},"branch":"refs/heads/master"},"18151e4c313b6acff8de4f57ff37036a65a811e0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-03 08:02:12.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/59/739959/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/739959/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/739959/4"}}},"commit":{"parents":[{"commit":"4a4a2344de09d61aa253174d9a7ca242d116913e","subject":"Merge \"Add a message for the amphora image size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a4a2344de09d61aa253174d9a7ca242d116913e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 11:47:22.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-03 08:02:12.000000000","tz":0},"subject":"Avoid trying to run jobboard conductor on error","message":"Avoid trying to run jobboard conductor on error\n\nIf run_conductor failed due to some error like redis uninstalled/\nredis server unavailable, log exception and do not retry in infinite\nloop.\n\nChange-Id: I76755b1dbae804e145d1b474f062fff58ec6a4c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18151e4c313b6acff8de4f57ff37036a65a811e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18151e4c313b6acff8de4f57ff37036a65a811e0"}]},"branch":"refs/heads/master"},"7b42782652f0106d0dc7dac5c6461b214f1e4377":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-08 12:24:25.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/59/739959/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/739959/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/739959/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/739959/5"}}},"commit":{"parents":[{"commit":"e1adb335c788dbd902bfd8a00f748da4abb8b06b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1adb335c788dbd902bfd8a00f748da4abb8b06b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-08 11:47:22.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-06-08 12:24:25.000000000","tz":0},"subject":"Avoid trying to run jobboard conductor on error","message":"Avoid trying to run jobboard conductor on error\n\nIf run_conductor failed due to some error like redis uninstalled/\nredis server unavailable, log exception and do not retry in infinite\nloop.\n\nChange-Id: I76755b1dbae804e145d1b474f062fff58ec6a4c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b42782652f0106d0dc7dac5c6461b214f1e4377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b42782652f0106d0dc7dac5c6461b214f1e4377"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
