)]}'
{"id":"openstack%2Foctavia~740096","triplet_id":"openstack%2Foctavia~master~I6152838c697e12d19b27343e3a0714e55ca52d88","project":"openstack/octavia","branch":"master","topic":"improve-terminology","hashtags":[],"change_id":"I6152838c697e12d19b27343e3a0714e55ca52d88","subject":"Remove blacklist terminology in the Octavia tree","status":"MERGED","created":"2020-07-08 19:22:26.000000000","updated":"2020-07-09 09:55:19.000000000","submitted":"2020-07-09 09:53:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740096-1594288402999-64565078","meta_rev_id":"280b6f54b37a6989cfb5444e2e989d72b88bf75c","_number":740096,"virtual_id_number":740096,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-07-09 08:24:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-09 09:53:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-09 08:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-08 19:51:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-07-09 08:00:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 19:51:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-09 08:00:31.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-09 08:24:37.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-09 09:53:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a2419ee504871684a75697903e0ae1ecebe1318","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-08 19:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29420bed17a314f2c0fdaa21cff942608b03d574","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-08 19:51:49.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"9b5957c8359650f759c4bdb72921a5db18aed22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 21:42:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0f41263d4e847d2a72d457cfbc2df15 : SUCCESS in 10m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a890dd76028a4d0a82c88277468a22b7 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e3e2b463a0843f2b72c72ad9deabed2 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1b821126a9b4c0ba1e3d5f4e63fadbf : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9de56d762544b59a05b9e692a623e22 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2eb655099b94ffaa8f9342c84bf6848 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3346785e441f4383ba28a06f80ade165 : SUCCESS in 7m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/59ad3984fb4e4edfa57afea636a85bb1 : SUCCESS in 8m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5f888d88cf5947688d1f772ca6820e62 : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6dc9e1487def45f3817a9ab2e7741c6d : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb9be4c9151b4195b872678471d7ee67 : SUCCESS in 55m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ac7f1dd71d84a11ae554a3ece73fff1 : SUCCESS in 1h 13m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ca6187f1dd44917b1523213b76bf599 : SUCCESS in 1h 08m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e065776fe6e422c96b944e8a82e4861 : SUCCESS in 57m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d550eec2dd241beb3735c2c7afb0148 : SUCCESS in 1h 03m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3a6eed90d35f4db48555526434cdf2de : SUCCESS in 2h 07m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8b82f8e134449838092c16ba5bd6d02 : SUCCESS in 2h 09m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b652f888e1ea4d16b3fdbae5583fa247 : SUCCESS in 1h 57m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/190873b444174b03b1ad3e73c5219ecd : SUCCESS in 2h 09m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f95c25e2ae7131a416166828cc68dd94ed4b6a61","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-09 08:00:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cff3bc88dc2ea4985cecf83958dd03914795901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdcca35f0f843a8f5433f77fe201d6947d99ec84","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-09 08:24:37.000000000","message":"Patch Set 1:\n\nHmm, I\u0027d have preferred \"deny\" instead. Shorter, easier to spell and to pronounce, and more widely adopted as opposed to \"prohibit\".","accounts_in_message":[],"_revision_number":1},{"id":"400e551104774073cd7c956e147b7d252a2e45d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:53:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b3f1d020796f4ff8ad6c016d986ba1dd7cd7ca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:53:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/484d0a86777143a7907af83f3a34d820 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5a58f39edd4cd188fa97e447ab0055 : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a9f7419366b40ab8231d4b1621739df : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5541efa041394ad39adbdedf0663561e : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd0ee1147254b44b20560586d83e3ea : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df12a4e7ff6a4e60bfe79f61984fd038 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23bc038c51e34bb0a7226ea66b3f2b17 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba2342ceb71742f4aa7e80fceedb997d : SUCCESS in 57m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1453725a17db4c10a092b3e3fb8a17a0 : SUCCESS in 1h 46m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75023e1afe444bf79a8a11ddc1ea0ada : SUCCESS in 1h 02m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4f65d822d0b54e8d8fd51e6ed586f828 : SUCCESS in 56m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2191b1efea484838862755eb588b0225 : SUCCESS in 57m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/96129f965afa459095f5302d561f133c : SUCCESS in 1h 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8e3f82234a130868cf6b68526e008535f52d58fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718f0d1855924d9dbe31e6639a5f92c0 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2f7bbdfb57e4eb8a69e8dce467291c4 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7890f0d99929f0c2630486c7956448b1ae3d1843","revisions":{"7890f0d99929f0c2630486c7956448b1ae3d1843":{"kind":"REWORK","_number":1,"created":"2020-07-08 19:22:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/96/740096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/740096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/740096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/740096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/740096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/740096/1"}}},"commit":{"parents":[{"commit":"bf196db6bc535de616aac4947405fedeec3d301b","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf196db6bc535de616aac4947405fedeec3d301b"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-08 19:19:52.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-08 19:19:52.000000000","tz":-240},"subject":"Remove blacklist terminology in the Octavia tree","message":"Remove blacklist terminology in the Octavia tree\n\nThe configuration option tls_cipher_blacklist has been\ndeprecated and replaced by tls_cipher_prohibit_list.\n\nChange-Id: I6152838c697e12d19b27343e3a0714e55ca52d88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7890f0d99929f0c2630486c7956448b1ae3d1843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7890f0d99929f0c2630486c7956448b1ae3d1843"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
