)]}'
{"id":"openstack%2Foctavia~740432","triplet_id":"openstack%2Foctavia~master~Ica759236f395c7bec28c794d2f1df5e0b690939a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ica759236f395c7bec28c794d2f1df5e0b690939a","subject":"Alias change amphorav2 -\u003e amphora","status":"MERGED","created":"2020-07-10 09:41:50.000000000","updated":"2021-08-20 14:34:20.000000000","submitted":"2021-08-20 14:31:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"740432","meta_rev_id":"061ddae3122a8aa1c93bef7344325b64d15ef29c","_number":740432,"virtual_id_number":740432,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 14:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-19 21:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-08-19 21:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-08-19 21:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 13:54:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-19 11:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 09:08:47.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-08-18 22:01:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"8136431277668600135f0429b5258c286e71fb66","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-10 09:41:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a931f8187caf6e580da1ccd07e96e180ffe4c177","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-10 09:42:20.000000000","message":"Patch Set 1: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"cc9ecd42d3b7e075030253e60ae7be6a34654e89","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:39:53.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/5e50a2fc93fc4b0fb83257026a29b851 : FAILURE in 43m 41s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/9bf5116be4c242d1afbb829f0180370d : FAILURE in 1h 46m 19s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/b99eda67e6df475d9a94e80a5f306905 : SUCCESS in 57m 44s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51f29cda08ee4520b867a49fced030f1 : FAILURE in 45m 16s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/f22b8ce1ce954e61a8a27eb046d7bfe4 : SUCCESS in 1h 07m 14s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/1835b2b10e6447b690da5ae6e8627f34 : SUCCESS in 1h 43m 13s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a33e1b396e948aebb75e55de22e0627 : SUCCESS in 1h 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ce7432f9402032963b9b7198a6164544671e5127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 12:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81ff978f7d0447abb22e9fc9fe07505b : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d1747e5f2d5431a81a864c8050f2455 : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79eda63d40074478b8acdb1f969ee99b : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af79610bbb548a0ab4ef465770bf298 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c7e6fb8c60e4f9dbc1da9d39f5c78b7 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70ae3000bbd549bf9f8f36ec354af5a9 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874f35e5e1f34092bcf0e33001c63ae0 : SUCCESS in 19m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aa9c15c6ceb54a90a14f4563e4ab0adc : SUCCESS in 7m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/29602924d79242a2a004c980754ef44c : SUCCESS in 10m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66674aad24f744dcab8f80c86864f5ae : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bde9039ee71d47cb8414027603e82c3d : SUCCESS in 44m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11ccbffe6d594b4b8229482a789b0952 : SUCCESS in 1h 58m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2281f1e1cc4f40a49d57b14e50a711ad : SUCCESS in 59m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ce144c8aec9421088dfd96b7378b78f : SUCCESS in 56m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a1103b401124219b697249fce9e9865 : SUCCESS in 1h 04m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8b92e7a8a324b62a371127c87a553aa : SUCCESS in 2h 01m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d4e0b7e91444891812357f0f4bdca01 : SUCCESS in 1h 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8996b477b5740a19927440bcc33d746 : SUCCESS in 2h 03m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/235b5dc17f884580936559ee515020aa : SUCCESS in 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe9cd40061d7a22fdfc684ef72576c4fb9db5646","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-10 12:14:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc049ddf514cbd21ea1d92604c66f69aa62fc6d1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-10 13:34:04.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d648ebad5c1029a1d958d91596f11722f02a6718","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-10 13:34:17.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"3ae169e3dfffab847a3e141a985ec33cb6d23e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 14:17:39.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e929aa1a605046aca00a26629cf2a345 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07930c7bed9942cf9b2c34ea39e94da8 : FAILURE in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9311a1011a204833ad681df4e94b6b48 : FAILURE in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aba4ebbdf2340fdbf42088b17ce6e36 : FAILURE in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7018c8fb3c0c4427ba7a100b2b626be2 : FAILURE in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810bbe1fbc1543208562fb87ce7d1ece : FAILURE in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c0d5f3a757439997624b780ce1787a : SUCCESS in 13m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/219241f545094721981647c2e41f3264 : FAILURE in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/45b19dab446445e9a5908ba99914fe0f : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acec52f4e6b44865b1d7a6b7c2de3755 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91b1477fd54b4c5b872e534d6c14be44 : FAILURE in 48m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c16185f2d2f34224bab0ef87a7feebf5 : SUCCESS in 1h 01m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4932be746ae14e35b90700e9a82f8db3 : SUCCESS in 42m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c38332edc2c4c38881d1aabf9dda0c3 : FAILURE in 55m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f2d5a4ecfec4b9d928877151755b42c : FAILURE in 1h 13m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0151c6e71692459aacf38d4323b4a165 : SUCCESS in 1h 40m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8aab707c2c34fd78e15e319d7962979 : SUCCESS in 1h 54m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5af4693c1ad4adc9348fe704f78921d : SUCCESS in 1h 51m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e8bd018c8934f009678e9997b625b86 : SUCCESS in 44m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6ce8aafc91824f52b9cb0d0f95a0be73 : SUCCESS in 1h 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"534b8a95a8ad44eaa46423506ff4fb364cedcb09","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:35:12.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/c6928366b2ce46caa7aea5bf42533576 : FAILURE in 46m 39s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/78769da7de9f4b8e9ad2a555bcb8a871 : FAILURE in 1h 50m 09s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/284edb9e40b04700a00dbfde8f3e01e0 : SUCCESS in 54m 22s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/002a8e1752d9445f93a4585284012f72 : FAILURE in 48m 59s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/c24db492882e447181ea0475f9bec9af : SUCCESS in 1h 44m 26s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/93c46174648c471b9a4ae5c4c615e2a3 : SUCCESS in 1h 44m 28s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef4c61bdb71c4a2d81729aed55ff9857 : SUCCESS in 1h 40m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1843bcab5a22d0f2b91cce3e37581063e07bc6ac","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-13 08:42:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a55b08edda3a80e96a08cd51306a694aabdddbf2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-07-13 08:42:40.000000000","message":"Patch Set 3: Workflow-1\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"4cf737fbfa5b01cb2b69ef8e3a62668e1c815609","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 10:52:14.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c453533dcf04b26bee3a2e55cb6ce80 : FAILURE in 46m 05s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/9be8c0a4194a46a3add022009fef6e89 : FAILURE in 1h 40m 25s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/d841ab85990e4bed94bea3987c30a3e0 : SUCCESS in 1h 03m 07s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/01e3fe51f337485fad8482b74e50f979 : FAILURE in 43m 07s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/5e2e051da4f442ec997bdd7aad818b0b : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/260dca6b5da94d2fb272245286430ccb : SUCCESS in 2h 03m 35s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/a59f38d1332c4c0b834d6a8c47f08863 : SUCCESS in 1h 56m 03s","accounts_in_message":[],"_revision_number":3},{"id":"6327eb0e9e48e7a63f4a75365bb7ae0de4e3413a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 10:52:17.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/832fdc33cbb3422d987bcf01ee110c9f : SUCCESS in 3m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/602cac5c6da14494b0b48d6a32ac5098 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b59e8ef9197e4ca2bf4bfdc61f929217 : FAILURE in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8be3c45c284e45f992cbb009c8ba197c : FAILURE in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9344962dfc848dbad0d9d1e8518bd2b : FAILURE in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d2e72507545429baefc41b3b894fe70 : FAILURE in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cede528626b6488d825b26a7e6e4b706 : FAILURE in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd4c474320c41ca81d18b929401a6f0 : SUCCESS in 12m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8188fd69d0cd4a5fab03dca00d832cbe : FAILURE in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb9a1404814047a8b17e139860080a5f : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18ff4cdf4aa7460d84b51c50050c1d7c : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a43c90c600e4035a0ce5278622c716c : FAILURE in 45m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da5ad9419ab44c76b11e038c6405193d : SUCCESS in 2h 03m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/034cb11fb5634e59ba2c76f93897d5c0 : SUCCESS in 57m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ef329b2e159a4efa93547efe5a7e15bd : FAILURE in 56m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdfae9da3efe4de699c3895cd19d6ab9 : FAILURE in 1h 10m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ac2d01c89037490aa30593549fde47c3 : SUCCESS in 1h 42m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/99a3a98a65834012ba1822679e0e53ad : SUCCESS in 1h 50m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8883b60231cd4887a6c973be364e49d2 : SUCCESS in 1h 57m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eae62d2dc05540fd8aea766a6734f0e8 : SUCCESS in 1h 55m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/15fe536dc24446e9be75d928687f1d31 : SUCCESS in 1h 51m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1cf1e3fba54a2b60add92beddefe38b360f21b65","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-18 09:01:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4ea8f0622f657b0ea67de3e01d0d913d5361919a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-18 09:02:00.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"24fe60fd4b6d3e093a10adff5dfe88efdf7a61ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:42:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36334fbfd8714c57b65bb01473b20824 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20ef2fd00b004bdd9c283362026cba4e : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6b7d4b373d64306bdc0e64205fafa78 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcfe469dd27045e5886ff8f0c4efc6d3 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a85c4e993ea946418223def410eb719a : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ed81e631e8147c5945a7f08a0513726 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4286e1ef83eb455cbad81f721165bc79 : SUCCESS in 31m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4ebaef39ecfa438182d895220b0dc70f : SUCCESS in 5m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9c5aa47afaaf4824923b94d6efaf87aa : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7581ec4141784d6eae95ffe24066b6ed : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d0fff10fd224cd691da90f58e618313 : FAILURE in 1h 50m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a83a5fe0c2d43569dd82077ad0f14ca : TIMED_OUT in 2h 34m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d58e09d2be7144f6b34d93dceedc5105 : SUCCESS in 42m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a4b74f8e2f3c4a0993a38c90ee13fe5e : FAILURE in 37m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/659a550f882e4a119845cd212d12587c : FAILURE in 49m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c8a1c65333142649da94b444b78a803 : SUCCESS in 55m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1c1ca209a4d74ec096b2e44a20e21382 : SUCCESS in 2h 08m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cef29f049ed4e4e8c8a17a0e2bfb28c : TIMED_OUT in 2h 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3eaa40005f6d472396a65fa8b3a17eb4 : SUCCESS in 2h 01m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96baa04e0cab45fa9417c95087dc1d72 : FAILURE in 2h 00m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3095df56865540c2a1ad74aa2b0c2056 : SUCCESS in 1h 27m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/953474feeb884bb2b2f39691be644cd2 : SUCCESS in 2h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8a980a51893cfbb2d576179205c91ffe9d7e988b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-08 12:27:07.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"fae2d2214a0d13a5dabe87f0eb86eaf1dfd7ceae","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-08 12:27:30.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"08f2fb085b3bfa78b6b4ec5ac37dd41ee3524531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 15:07:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/789213218bef437ebe58f491374db910 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d63ad953efad45c794b4ffbd4d9a8108 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46cd06abe586432fb23b18141256107f : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0a62f149c44919acc6100e0792251e : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/244960f17f334bc2826af080ace7935f : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2feb2f51dd84603b7ed4280aa79f411 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824a09dd302f4ab9a3042348cb0993f0 : SUCCESS in 12m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/522aff8692484a43ae50574fe001faa9 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca55ceb6f7e04e5d9bf160deeb3d2c97 : SUCCESS in 7m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce9ed7f0987b443d8092a3d8a0daa1bb : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4f2746e0cee4e41946fb261242ffd07 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c64e440bad8c406fa4e9f75891bf77f7 : SUCCESS in 1h 41m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/586c6e66e4414aa4b54d123a786fa915 : TIMED_OUT in 2h 34m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dcbe0b485b87424a915f12d433997d45 : SUCCESS in 52m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/08aac5af9e374c50966b1735c0657e5f : FAILURE in 42m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c29a3446ca94061890e7416ab7e9c18 : SUCCESS in 45m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78e9f7eab2024f73bf2475e8d33d7bf3 : SUCCESS in 1h 09m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/296906d18f684b8da4a4f4549021340e : TIMED_OUT in 2h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/133bcf3331044bcb90645a8afc7574a2 : SUCCESS in 2h 18m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/853e7e1c3cb04fe79b7a4c03c755f4bf : SUCCESS in 1h 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/137e5c8b964341b5b205f66b9dd17e17 : SUCCESS in 1h 40m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f3d6a5d1e1f346bbb6f7b2ffaf52cec2 : SUCCESS in 1h 44m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/434a17b4a50c4babb60bf8751e344dd0 : SUCCESS in 2h 08m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cb7bea77246dd8c786549974a65cbfd091426ac4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-26 11:49:42.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a83afbeecc84f8ae47b14b1c970691dd8313f8e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:36:10.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec0097f8d13b4ff593b81206bfe30d8c : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3c25d2134df4dc7ab866bb4fc0fce8b : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a74e218cf33248409510fb91b8dbb6a8 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a9cb72b5084886a9b3a906107a2925 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c36b639532be4910be3595da44200d06 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d2985d3beb4cdbaeaf11192b0a721d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ccb29b84914cc7a64fc141b85ea572 : SUCCESS in 10m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bbec7df14c874203a8ee9758a8536e19 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4f19ec122369490783c2dfd0d901672f : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a3f7cf293c38487a97fd66e30f41137d : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e973c8e9cdc4671b76aa4786a1aed09 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66ea110248c84037b48ecdc6849f6706 : SUCCESS in 1h 42m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f67718327a9049eaaae0535e02786097 : FAILURE in 28m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/34e137574dc84db2a34935d08c6c1107 : FAILURE in 33m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6feebcd0f3a844108ec59394bc875dd5 : FAILURE in 28m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1034a02b2ab645fabb42929401f5f18d : FAILURE in 30m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f42f8401039461488fe683d3f0b0fd0 : FAILURE in 52m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d9778ae969d340bcb5fd9712b6589b98 : FAILURE in 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ce16271e2da4b96a81bc491a6e0cd1c : FAILURE in 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/643e1dc89b73428980591a05fb7a0836 : FAILURE in 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/88c13bfe151d4622bb33453aaa0756fd : FAILURE in 29m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a91bccd7e32343259ecb25e020440bbb : FAILURE in 28m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bbcf720bf0e44b9bb69849e65521e193 : FAILURE in 30m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"56866e02e1744f76184bfdc53adccbf622a7ae15","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-27 08:49:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"ff631557741aa72507520d3fab605d8a46c26fe0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-27 08:49:13.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"eb6105676ccf63efa7e2e2e56902d4c1e3ff6d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:17:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0eb9f624c18e4cf483f318b3d65da4be : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38fcd618c17342318847b1b40aa70316 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2308d15673e448faac6db95f2d22cbed : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3fe833f5554675918d6b7368b9a1f2 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/007d126be36943209741cf574ca71d2e : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89cef9b5993e414894521bd19033703a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2affb3a564af4b028f948fed8bc20dd8 : SUCCESS in 10m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e9a92fb926b34b2b854d3f36d375237c : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/178c1b3336814bcda22767f610a83356 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7214646b0f6c4d7a86a5b507d4b44275 : SUCCESS in 13m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe6bf5ac33c74470a38c08174d08bf79 : SUCCESS in 18m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5acf18b5bbb44e5b8e279fd4fa249f30 : SUCCESS in 1h 41m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfc553d720d149f7a16bb10905284d94 : FAILURE in 30m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d363c6d1a3774b9ea6f2fd6e9463277e : FAILURE in 1h 48m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8868a049e98947cf813c07465c6ad741 : FAILURE in 30m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56099edd8afc4143abadf519c87b1771 : FAILURE in 30m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f362b88b3f2d4ee6b6ccfb5a26b4474c : FAILURE in 53m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/33c36d17aa11489c976d8c587f98b6b2 : FAILURE in 31m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88f0587200174ac28a1664693f7aa701 : FAILURE in 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1c2542b8d7e148cbb4f4d5be2c21da44 : FAILURE in 31m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73a7b743534f474d993d9bdfd19dc3fd : FAILURE in 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bfac4cb6638e4a2c979b8ca78df9bafe : FAILURE in 29m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6ae9063533c54b479d689eb33b1f3cb3 : FAILURE in 32m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9d41cf4b044d999905d6496ac7c992e3604ce11e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-29 08:51:33.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4c7fba1da821845bcb179390262de28b894fa96f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 11:35:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bbf072fc34ff415b9df9d23a19a39dd3 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67ad28d04b244f30a412aea207d04191 : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818057e8d9fa4273bffb6840924d0b91 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb698ca32c241bd8a61394c50e48186 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b147372eaee8429e91a75d65ebc28c33 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b28ebffc1371479fa5a13156287da762 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1a79a2a39b4b31b9a6a6fb79c986ba : SUCCESS in 21m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bed884ffc8954db68a4477cdb2e49d2a : SUCCESS in 10m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/74ddc72efdd246b1bbb7b288b529e0bc : SUCCESS in 9m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5d3b52e8d9d44287bdbd87e2ad17cd0a : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7334f544f0f3472cbe9334fe204ba759 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e9623f819334187b0a0e045d7c2102e : SUCCESS in 1h 52m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40c2b04f04e948a890d2870f5d3cae25 : SUCCESS in 2h 12m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca3f8a05fd724d9c97d27794ff3a1ca1 : SUCCESS in 47m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3069558e9ca140bda46e30e9bfd76b65 : FAILURE in 43m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4388f7650113488c875920ef3e14c481 : SUCCESS in 45m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/519afea041ee453ead8a2b1ef861a950 : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8a7d92bf9a748359c13777885c79491 : SUCCESS in 2h 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b6a5a0cfecb44a79de10486313ad884 : SUCCESS in 2h 17m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d6c8856af44d3aac9614d24c0cab2c : SUCCESS in 1h 56m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/604a2f4d195b4c2f9559da285b643ca9 : SUCCESS in 2h 04m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e5ac438339a24ea59097ef8cf65965ee : SUCCESS in 1h 54m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90d28fb162ea47bfbf0b351e8dc78c0e : FAILURE in 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e8dfd3498da71ff29b97a9cb67d327cc4af39b44","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-30 10:45:24.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b9a948a02a1e8d9cac650de4fc367d35bc8f782d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-30 10:45:49.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"aba1660996950426c791ed31e62887dee305f035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:27:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba849e6f1970490abd8d6067e8131188 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33afbe815f6b43fdbc579764ef75ba08 : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aad548f69c664e7a96d6734fa321660d : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb58417cd164a51b9baa4d0451289d5 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09e803ecceb41e786a8d4b9333a803e : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ac921b64eaa4c62882ea6cf434d4400 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3157d8c97e49ba84eec03e6ba65179 : SUCCESS in 13m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/53034eef77cf4029814f4a4b132fda30 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/28af350abb1b4ed7a4dc2f29ef22f90e : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/224c88ded0864051823e048447ad83f6 : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88ceaad6396f4fe99185c6acf4e10176 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90820945b1934a2588bbd30dd1168e0a : FAILURE in 1h 24m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eaf0b96a4887484d862d8c19dd2aa9f9 : SUCCESS in 2h 02m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c447114031445988b651448afd110d6 : SUCCESS in 45m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/93957abf023742c6a22bd1678cbbcbd6 : FAILURE in 1h 15m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f1108bb268d4b148a886d5f21679c20 : FAILURE in 1h 11m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1833c6ccce0f4073b64e1c0ce8c02d34 : SUCCESS in 1h 16m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ebcfa4774c945b0b02242cd7696fc99 : SUCCESS in 1h 28m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6b3b943afba4d06ae9139189d0d6a38 : SUCCESS in 2h 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8adc94f9ca474acca84727c5585b2270 : FAILURE in 1h 51m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e94bf1ae3fbe4734b3b80ea84d6bef0e : SUCCESS in 1h 41m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fbe7c45aacf14f609da9509bbd164280 : SUCCESS in 1h 59m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/76b8c288f41b4ff5ad826c65b2a5cbc7 : FAILURE in 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f743034c9ed4f2ff8f6ee4a79813ca60829e2138","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 09:01:16.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"5017f8b468e4e3f6ead5f9533d0beeb81866bc4c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 09:01:22.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"d6db2fbbd60f62ffcb86ebafee21ea6c5e4b5fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 13:58:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/614d6b9873c848039d2eaff142d7766e : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa0065f62c8d4d6fbad9c83242bde657 : SUCCESS in 12m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35a49434c9a74ce1bbb6c026be411756 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a49bf713a5436db5300f992aeb5b67 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd77113be71f4e8cbcad1fcdb8708016 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3214cb6f8e8418c841354f5b08c3062 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5176f703e6e54a298fb409f2c12cdafb : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ebcb4ac11a45cda6991a89eb453472 : SUCCESS in 11m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9e7096b5e9604737bafe529a83d909c2 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/186c48035eb7448d9e4e3419015f5378 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b9f41b07f8a34d06b8227b5555d7a7c9 : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f997e4a4e11c43d1b82dd050dd04be00 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07510e1790834292a83edd5fc7a62d08 : FAILURE in 1h 32m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3aca12395bf9419390a1c6a548c9ef2e : SUCCESS in 1h 43m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43ebce24f0aa4e5d83d1000f53caa6cf : SUCCESS in 45m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3c2c527ec6a446268ee027370b44e1b1 : FAILURE in 40m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecf63027692644df8eaeca664a9b501e : FAILURE in 51m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae8637e9f9f2426b860657e8cc931cb6 : FAILURE in 53m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a82c79959c5f4a10a27ee12871cd04e2 : SUCCESS in 2h 04m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bac6169f04cd401685be9d4d97eac28a : FAILURE in 2h 04m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0bab535634384219a029255468e14399 : SUCCESS in 1h 50m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1a4ffe80c6ad46d4bf0a8dd9b39b28fe : SUCCESS in 2h 10m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/22ce94f99ff04b60b814e60529442883 : FAILURE in 1h 48m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/58e471780a20403986fb490447e9d426 : FAILURE in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6405063318e542457f8e8e29f5ac3e1f563dcdb9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-17 07:28:25.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"30a45283f5e8697407dd9454ef79deb417a491f7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-17 07:28:31.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"75f47d0c2c67613ab2fd74639fda731d08da0c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:37:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5d9780d7c4a74f9283a4548d9c07de60 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20260e914d914298883b7bd570d39589 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95d8feca396b43d498e6ed89465c4cd6 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84672cd51c6b48f79bd7dd7cc9e4d6ce : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95d060283f23414f9531a8a8e9335586 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d1fe276b1342d9abd10a7bd4740cf5 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d17b6b110a54e37b0a4e78375c5ac23 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf37ab329ba14b4ca9ff0fcb75f188ee : SUCCESS in 17m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2a95b822384d4770969da08f051196fb : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/12cbd0996ccc4c27958a3eba7a5fa407 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ffe210b1e294742bb788921c33328c6 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6880411fface4c30a203df610d271193 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40f0d52bd4974e0caa688029df084b0f : FAILURE in 1h 29m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6eb0a05c7d342c1abdfcfef1d37899b : SUCCESS in 1h 49m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7502e69e83d646a7bf79b2ac69ca8b28 : SUCCESS in 39m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fe928021920c42c18ac7c860e0e11da2 : FAILURE in 40m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6efaecd00ace4a1eabf5293a9ea227e3 : FAILURE in 47m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0a9b97eef6148369d9514a97c404e3e : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/53c997504a9247e983eb0456cf3ca562 : SUCCESS in 1h 39m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/073bbbd9db5541a089e704303b6fac54 : FAILURE in 2h 04m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23c292afdc35451c93539707cdac96aa : SUCCESS in 1h 50m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a43557ba3c21450e8707935d400be263 : SUCCESS in 1h 52m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b236ea281d644c78a9bac093746ecbb8 : SUCCESS in 1h 56m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4f47dc3461984ccba03340d78d4f48f4 : FAILURE in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b1ab293f2aa89df815c14320be4430fdec007d4f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-17 13:39:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"735c230deb48999eb4f960ac26a37635007ea3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 16:24:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/341659ce3e754c779c2fea00f08ffc68 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36a721172a524acaa86d8a5c3d78155a : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ccbc6bfd0ba42eebd1c1f41267275a1 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a10f5f37740e46a0a5a4ed505d7ee868 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d53d5b93f84485bab7e29f611cfc4ce : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/856a1421871645b1a603ace7063711c5 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24506b1c3d1845e68c51ca5acd02d55c : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95587d18d73f4ffaa4f56ef1f9368c47 : SUCCESS in 21m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/909440816546470bb38cbe6efcac9d7e : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2983671227204ee38a3565cd83d64434 : SUCCESS in 6m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f04912d8bf240b5933882f3d23f51c6 : SUCCESS in 4m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28c3b3e8096a4463ae584615e5f2ef05 : SUCCESS in 11m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f7d2aa1ea9c4365b2dd043c7fcf47e7 : FAILURE in 1h 18m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53c3c62c0010408fa835e5a0df2f5422 : SUCCESS in 1h 54m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0465453f0b149e6a8c225e051914cda : SUCCESS in 39m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3d8292a895cb43a2a80d8e61cf743fe7 : FAILURE in 41m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5efab7f03f742a5b6513f3ca862e69c : FAILURE in 48m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2963327f95f41729970d17056b15841 : SUCCESS in 1h 01m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d583ecdcfb3403786f284f7103c02b9 : SUCCESS in 1h 57m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b9a3b16834904b7ab4f369d2f983d12b : FAILURE in 2h 26m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37dedf855dca467a969369dd6c4ce176 : FAILURE in 1h 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b6842f38372b46f98c5a75b25390f39e : SUCCESS in 1h 51m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7b642dfc61544f9a8a9a400b734e66cb : SUCCESS in 1h 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/81e50215a7ef4b5ba84ea734c98a7f5b : FAILURE in 38m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"73d31cdd27460c486d0e3bdaa6ed2825c39b00d7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-19 08:47:32.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"504dbe713d77279b23fd47bdf7f7f75d1345cb41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:29:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18398cb7636b419a9d70d3312b440942 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07cbe2150e65487db422affd52fd5fb7 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a02217a61bfe4931ab8f50da2fc5c8af : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448438f2eff54472af392533f921eef3 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4a54ebdfa84417a968c357a0e117ab5 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be0cf65103324c8d908463f9b5716804 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f842935b936435abb1c9e84c6d667f5 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9eba9182a9427f98630e62454cd0a7 : SUCCESS in 16m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7bab90df1bbd49098db9243f6ccc6f58 : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d5a07e6beefc4e90987440ccb8be12d9 : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d605e6085b7743578c538b512fd39de2 : SUCCESS in 4m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/183fadc8bbaf4d33b9f14de7df48890c : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84b01996f62a4d6e8d12c7b1001bca15 : FAILURE in 1h 44m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b970cef95d74ad4a3aaa178a03f4b1a : FAILURE in 2h 19m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/869f4c1292444a6e9a1c8b4bef407b6b : SUCCESS in 53m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/84c5e539b6264af0a284583c566b3e46 : FAILURE in 33m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78c3b97c776f4b6fa5fbadb25fffcafd : FAILURE in 53m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14093713ff96472f8c3c7b4bfde7bbea : SUCCESS in 45m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45b74c0c77db49c7bd6a83c1493ca558 : SUCCESS in 1h 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f294c9b2abaf48abb3f90f47d9e7b82f : FAILURE in 2h 14m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22c09495524a48cea16c3f75ccdd75e0 : SUCCESS in 1h 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ffa22fd91a40475f89b63f95b8e73088 : SUCCESS in 1h 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ccf54665d1f9467193391026533617c3 : FAILURE in 2h 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a256df38e63f47d19dc1482e56bb0d1f : FAILURE in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ef2972bbda2f0fe00cfc7ed65d868657f0dd51dc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-20 08:51:07.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"6808169473ec1d3b0306f477c75a3be1d26757f9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-20 08:51:12.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"b1603dac50b5573133f0937cc78b3a199e3523c0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-20 08:51:30.000000000","message":"Patch Set 11:\n\ncheck experimental","accounts_in_message":[],"_revision_number":11},{"id":"a89909af2b40db46251c78c243a62887d6658daf","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 11:32:43.000000000","message":"Patch Set 11:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/38d74c035cf148ca986c52f023866596 : SUCCESS in 2h 02m 03s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/739e9f7a13cf42349c03228832b48071 : FAILURE in 1h 42m 16s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/0b78f6d7e4a14fe3aa4c45a8d8e64019 : SUCCESS in 46m 19s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/723acdd6402b43fd8c4159d8bce12f6d : FAILURE in 54m 11s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7cdf090e97b8403ab5f364afe7780475 : FAILURE in 48m 08s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f6425a2b58643489f09329bfb274fc3 : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/eec1db818a424dfe9c4ce4817a430885 : SUCCESS in 1h 46m 40s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/802a29db374745bdafe8ece5d4eafddc : TIMED_OUT in 2h 33m 05s","accounts_in_message":[],"_revision_number":11},{"id":"5de3028dc6129494497bb17c70e8442e23ec507d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 12:21:57.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99b999500b9a4434974d2a23cd301c16 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc4ca48948d84ae6ba7689f3c73e4d90 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43a798b2ccbe4a3baf89b3aaf669105d : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ab4433906974a00a1cc8e5676f40e71 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60547e3d002047dab2ce63a073da15e3 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ee3d365bfdb44c4a4f0c9fe5fc24441 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abf24a6b622d4f0bb43637c6b57ac4fd : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/423e0b685139484992c6707439233880 : SUCCESS in 12m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5d171fcdbd8447428c87bc29c3c3f566 : SUCCESS in 7m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ba40210c05ff4294a0ccfae5d9bb16be : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c8673ce463a4d95b2a5e3d1721a85e0 : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/235f13832a7a4f5cafe53d049411b066 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b3714836ef8440d8a7866d30410564de : FAILURE in 1h 54m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86c8cc138aa14533b60bd0690999cb2e : SUCCESS in 1h 42m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74a8980953e64184a5bfe6e214fdc70b : SUCCESS in 45m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/66809630b27f496d842717c6ea2b64b3 : FAILURE in 36m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c25270d5eee7484b97942943878348b4 : FAILURE in 54m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dbf5ae534f3340f8bf7307ea3e894f69 : SUCCESS in 1h 04m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dc9db73746dd492a8fb9b2403e65dd00 : SUCCESS in 2h 03m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1a86b5d71694d8ebc5f323a01b08f58 : FAILURE in 2h 11m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f95ed564a7774b6bbd8d5a73098109b2 : SUCCESS in 2h 09m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fe44454b3cfd42a2bf3180ecae0e06ce : SUCCESS in 2h 31m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/350e5796a07749baa1a851c0727c8ede : SUCCESS in 1h 52m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d361932305324960a1f4df68cfd11c49 : FAILURE in 40m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1f1caf9b0424969db684fd19ea0d0d07d490b363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-25 10:07:50.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"65e57097ac6246e6f548256e7c66e54523e5e4a7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-25 10:08:03.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":12},{"id":"69a5dd124c9499b37cbfcb0c943312f0292c48f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:18:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b190537bcf84085a1541cf0a56f99eb : SUCCESS in 2m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88661fe4634b4584838ac9a15ce73043 : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e4b0546692f472a9a3a751384338acc : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/015e07d368c9478f8b41641eaecce34f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c801240e61243fba23221ba321ccf22 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d04ded29dc134c32afdbba78b8857c8e : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5af7d7209c4d450a8db6bbcf1dbe039c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a88fe0c60c48f6a9373435fbdb092d : SUCCESS in 11m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6c24d3e0f4cf42a28df4d57428366fa8 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4e5f78b3a25945d9a60b0a7cdeb7f701 : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c97eadaa2ac94251baadbfda6b00c305 : SUCCESS in 4m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c91f6fb238341b0ad77745c183eb6bb : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11525ce204834fb2b53689646e5a09bb : FAILURE in 1h 44m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7f8fd94ecd149d4bb6bb090e7bad345 : FAILURE in 1h 57m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26e1478f14f94942a57337d2f8a833ca : SUCCESS in 42m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c67fe201787b46d6a21130867502d680 : FAILURE in 37m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6d556ac8c3b44f7b40f5d36f1702ebd : FAILURE in 52m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eb12f6bc52ad4573b870da8969cef0b0 : SUCCESS in 1h 03m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5638e9749da8455388c4a86fd437768d : TIMED_OUT in 2h 40m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/24e5bfbabea241fda892f3616af6e17d : FAILURE in 2h 07m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97246f4c89c94c3f94015d601c603aee : SUCCESS in 2h 10m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9ba446938a734f629f95509d113751bf : SUCCESS in 2h 07m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9898b8b2a8024464a7c6ed7cae746cd8 : SUCCESS in 1h 38m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/506a62aab8564b39b89bc244ada4cd12 : FAILURE in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"08403925bed1f4a5503290f9d8651323683f77e2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-26 12:21:53.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"bceb23635662ca650d83a799b255a7397bb877c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 14:58:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f50ac19e02814935990a321d37771262 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b88dec373a543f09bb9a18dc28426e6 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73c2c6552ca045f597248342f621b81e : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37af00bd04364bdb9b62557bd4ab52fb : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aef115bd6e9449fb9b28467dfa3c127 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd5ad9a53844a89818aa228a08260fa : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fccfb74532c7483dbd34662ab559f15b : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96baa1505d6b4d57952c7c238ed61a1a : SUCCESS in 12m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3b9972bb2b0942e489b0c936bda9ba8f : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/852dbf9fed53455d8bb925d939ad9698 : SUCCESS in 10m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da447debb3ff40e580602e12cf695606 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36b512e7232b4bb0a6c64d0aad2fcf93 : SUCCESS in 9m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06fb6f9f493647efaef69ca8f02e68be : FAILURE in 1h 44m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f91de88c019a44b4958716b91a37a5b5 : SUCCESS in 2h 12m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33a957800a584fb7a218498d3a6ef210 : SUCCESS in 45m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/168593b142714f5b90de73d0d9d9e078 : FAILURE in 49m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aad2b7f07dff46bfbe3724c68688dc65 : FAILURE in 1h 02m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f09ee8f617a94c74818de87025dc6107 : SUCCESS in 56m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c45d1efeca84d2cb3c55f827ccb1148 : SUCCESS in 2h 04m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/267be687cf8345b3aeec03b0b5ff73df : FAILURE in 2h 29m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/058a26fee9fa4681b0d6c90a3065284f : SUCCESS in 2h 03m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7720b34b0bd5460f8ffd868892e80fd0 : FAILURE in 2h 21m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6c954cd039c04110ad6ae18550282be8 : SUCCESS in 1h 47m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/48cb4d9b036b45cf923f090b21c8b4ab : FAILURE in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b225e311dcce22145ede5699050ec7cc1f49f9b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-03 07:20:09.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"6e99f8fe2a17e704046762500188a5f033fffb8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-03 07:20:20.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"95de45552fddb6e9ff7fce4059a3565f35499a3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 10:54:36.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/157ae123545849478e7c8fe7fd26da0d : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bfb27fbf51740579990de3688617b7a : SUCCESS in 17m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0437a572ab3742d58381b2236f39e6eb : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73b657315864c75a19e1b5c3c774936 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fc53c56032c4391a9dece6542312bf1 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f2f041f81644881b4af45bba30800d7 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fce02b18486f4157ab61441025a59266 : SUCCESS in 15m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358c3a5b44f642e7a1f0e4842ae749f0 : SUCCESS in 11m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b35e4671c7f041c4a2256768c4eca51e : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/06ccae4403e74c5aa5901a16da426763 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ebe0aec9d6ac4988850392e0fb4cd34c : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92cf3fc1af744046a4f74788f6c14241 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2482a434c6fb45b8aed8032fe9e7475a : FAILURE in 1h 28m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50e86bad31e246b7aa6a423245322afb : SUCCESS in 2h 02m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49bfec09b3ac418cbf9f6a0236ca607c : SUCCESS in 46m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2addde532f9242ea858014e48e4d901c : FAILURE in 42m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9dea3f72e454b1f8ddb2ae56017e3cf : FAILURE in 47m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/432b34c5d0b74cb38d6158a0bd8bf277 : SUCCESS in 1h 02m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2368f2e7031e427fa0946b83daf84f84 : SUCCESS in 2h 16m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd960fb20e674118b4622149588ced81 : FAILURE in 2h 09m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bb5c905be718407aad7dbf5d9e42627f : FAILURE in 2h 12m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/99b429716b6a45fda15c07f7a2db2b8c : SUCCESS in 1h 49m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b1ac0277bab14b9d860226d29f2b4178 : SUCCESS in 2h 12m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/51a49e5c1ad74a2ba6965b776d1edbc1 : FAILURE in 33m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5f7573ea4a50fcaaecc332e9e1bd71a2ae978eae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-03 12:01:45.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"d6ba94c1be01d72be7df813005d377f8b20c4ef2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-03 12:03:19.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"e6e5af844402e81d04c6e91e0193bc5c644a4ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:01:55.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fef7eccd908243d5b650244a28ed336b : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9196b0eaf974c119cac228729843f73 : SUCCESS in 13m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34a3a17afac14e25ae781788f5b704f8 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e34a7b57a44b568cb1d107a92a04ba : FAILURE in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/973bdd124b654ae1857a777cf4c7ed03 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8294cac1180f44f494bca51bc1e475a1 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e782902d425d40c5b56a3542b17a47ad : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf0cb09abea4a59bb016d5228925007 : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9f1439e5247f49b5a7f91e8bfd7578e5 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5c4db5050d7145e8a602825145fc6436 : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98d7d9e340104b2ea45b3a9b23b6b8a9 : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a6467ec80614cd19933559be4644303 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b50771e1cb3e4126842dd6bbf67177de : FAILURE in 1h 40m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ed3af0af976466b8befe81904496f91 : SUCCESS in 1h 54m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ac7f49b2ebf4a0ba019d436c28d82da : SUCCESS in 47m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/689a336f1ff542069c8a3ea7109e5134 : FAILURE in 39m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19ee9274c7fb4095abeb80916854b97a : FAILURE in 51m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01a3785ec039456fb1e96e7b3fff9182 : SUCCESS in 1h 03m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad4cc3fbec474f558599b49ed494ed96 : FAILURE in 2h 21m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eeca1b946d504ff8a9dd634dfa9cb048 : FAILURE in 2h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/28d073f8373d42a48ae3e5dd2a51b8c1 : SUCCESS in 2h 05m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d20157ffb92b4efaa3eba0fe7ffa1cd6 : SUCCESS in 2h 22m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6bf5509346c0457a8dfb6a43f81aa16e : SUCCESS in 1h 50m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6b91289c3f4f47f692bbfea2be62d118 : FAILURE in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ab86b20ef9733cbba93102a0775b29c3ff1d0560","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-04 10:26:46.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"0c14d8435aa0b77658f08182ee08fb03a5dc197d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 12:55:46.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d2e952950d584c5fa8999f634241d870 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb97e95be213498f865cde6fdfc43961 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d7a89b1c0614bcda61f2781a29c7dcf : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c00d2abfdce4763bfbf654890662bc6 : FAILURE in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ee90d2e7ead406696ae84209d814bb8 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f7d40d2a374952b628e4aeeb064e39 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5208cd9db7ef4ebcb18b7eebfe72960f : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfc60160a704e8996f58e1a0fa8cabd : SUCCESS in 13m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/21beee068c484716b362751b5e4ac533 : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6bfede0beb484b41ac9c051eabb24e41 : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d6751fcc52a54cd08b48a715f7716a71 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/048173cede4842bea8a989f98cb8822b : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98b95de020234dd19c1ac4b662cba72b : FAILURE in 2h 02m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0359e431a2e48d089910f460237a779 : SUCCESS in 2h 18m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1bf94849ff942a088b13cd9662c3185 : FAILURE in 1h 32m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ad5887ad6e04b54b078386a74859ad2 : FAILURE in 32m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/834ec63dcbab475f98bab3e2546c6962 : FAILURE in 45m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b9f1c6ccbe44ebf89f8630187eaf58d : SUCCESS in 1h 01m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8174522ec3bc4cdc99be828763635e3e : SUCCESS in 2h 02m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d23ecd39fc4e46828a0f9a348e19b315 : FAILURE in 2h 08m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e6b44bd9bfe7400da0538f4e8d01b40e : SUCCESS in 1h 42m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a7e51330276f4af4bc3958800b0d8311 : SUCCESS in 2h 22m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b8825b8b611e47098485c259249d70fb : SUCCESS in 2h 03m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/049582e108c54aea80e176a80092d6c1 : FAILURE in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"d9c9e97b45b50123a3250d25bfb087a318b55985","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-04 14:33:43.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"7f350eca74ad5f1518bb0cedca64fa04ced6ac03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 17:25:12.000000000","message":"Patch Set 19: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b81ca2f580b49da93e39c68684a9db4 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe0cf56e5fc54749ae1ebc9b571c9358 : SUCCESS in 17m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd1f1808cf1476fa03f5b0d90699ac7 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6a86014d604af69956ba77c31754af : FAILURE in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e845cc2780ca4300af53acbc40b27a7f : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c64f14f876b411ca81634077d33fad9 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6146ad0548b4ab6b5b16edf051ced35 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50eb67b841f6417187d7502ce4fb5d66 : SUCCESS in 28m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a8efedb179f4e53a2ca53537cbcc32d : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/72de4705c7154ee3a69d43cc8f1b919c : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33890f0e772348cea2d17c596936ad1d : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/404285c9a8c14e7b8b3b9228bded6363 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be22d0affdb848c4a48228a85bba233b : FAILURE in 1h 14m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f485a76d6ec0481f98f8a62fa0a0b194 : SUCCESS in 2h 16m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07fd2e054e94437381f2a894b983c038 : SUCCESS in 1h 06m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8e4b1d5ad12414784a8191998021a9b : FAILURE in 42m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67afda33ea9f44979300d66fa246a775 : FAILURE in 54m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85ca5b3b34ed4d82b787eea4465909c7 : SUCCESS in 1h 02m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/46a845f045674445b663748c4f47b9e3 : SUCCESS in 2h 08m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8eeb0ef6849141f99e7125f579fcda4c : FAILURE in 2h 20m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d7bc44417094e63b8f47ca4b486759c : SUCCESS in 1h 57m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a717fcb1e6d4aa8aef699025746c5eb : SUCCESS in 1h 54m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7b9f0c33e8d74d73b9f2123b6849d26e : SUCCESS in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fc1856c2d8f64cac980fc18c5a1eebd7 : FAILURE in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b97f013343e583fabdb4e5a7d4a7a6ee0ea93f11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-11 12:50:15.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"177798b5c53bbfbdc59dd83974062f6c900e2bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 15:17:35.000000000","message":"Patch Set 20: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f03cf7be6274579bd4d66fa4e036f21 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8c3a6c8516f40deb30842788214cd6c : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82c5815833a14ebfb07e1934bb4375aa : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe30348f5294f408001220c4be18748 : FAILURE in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21010b720b7c4d838b2ae428ad4c7a33 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba43b99154e149f7acb3e1b42626f79b : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74750963848840499bbd8a615a763caf : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2ea16e4d44946a184004a9e7ad95afd : SUCCESS in 11m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8530ec01c4194bf28ed53ea775b7ed92 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8ecb01afd7445b0ae76ab3c399e798f : SUCCESS in 13m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4afec20ffe34ff0bff3e956e0588bd6 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b6bbd204a1346f6ac1b0aa2e1648cf2 : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f437f4fc2c8e4e0dacbe36fddad6fb8b : FAILURE in 2h 13m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ad0feb9bbba4387b98c063b8044f529 : FAILURE in 1h 48m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2402b10f29ef4664a1ea1edd9c7a858e : SUCCESS in 46m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ae4dab49de544a60854e13e371425cd8 : FAILURE in 44m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5a68a6bd9aa49c58311f35fb4d9cb9e : FAILURE in 50m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0628c879b094adebc248b19d945d035 : SUCCESS in 1h 00m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df38d902e8d3457fa64032aa1fe07d29 : SUCCESS in 2h 08m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00210482bdfd49f1ae16dc873aa3c04a : FAILURE in 2h 22m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fb7c96b8a47a4cdeb6f5be629b95ef04 : SUCCESS in 2h 07m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d3c17bdd742340ff8bae375743437529 : SUCCESS in 1h 50m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fbc0d4bd256f412b9c5e4c675e20c512 : SUCCESS in 1h 45m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/28316e6db01e4a278a78646d6c847d99 : FAILURE in 37m 44s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d6f938b2a48c7e8d2dedf998d24bffaa356b468d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-19 12:49:15.000000000","message":"Patch Set 20:\n\nI faced some strange issue here - some amphora api are failing with https://storyboard.openstack.org/#!/story/2008377  when executed several tests.\nIf the same test executed alone, issue does not appear.","accounts_in_message":[],"_revision_number":20},{"id":"9d1021fe0fbb4f0d7702594532789fd352c809ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-20 07:38:42.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"475b2bc5cfc0c413cda0b18ab3a9f99c7fe70c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 10:02:28.000000000","message":"Patch Set 21: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed9c98d8becb41a5a0c929d1c80a9478 : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be9d36642d9e4ab5a7410ecc1beeadf0 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43c0c9565a7b4ab8ac023550b81b2220 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7406771954a4b089dfc73107098a7e4 : FAILURE in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9329098bc7214b2185b61708fb483128 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/143615639ab4492fb898c0ad35de375b : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2f70158c85448ad81d51643d5c5133d : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d497d01e5547039c283d5930032486 : SUCCESS in 12m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9443c9c26d894ae7b703ed69a05f3107 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7f30eb90f49d4191839aa685315e3dd3 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ddb513a24fef4ea78c777b0ee101e695 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f334d6d6a7e14b3c847877b4dcce81f6 : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0d8cce3d5b345a793fe1b886ce9a741 : FAILURE in 1h 51m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5947bb4b0b8404a8a21be9bab356dca : SUCCESS in 1h 55m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fbb51ef5b22846cb8c4d2b9e9f345c94 : SUCCESS in 50m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f2592b4ad4d54beaa71f328a99aa2066 : FAILURE in 38m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/369da839f43642078bedf9ffa0378152 : FAILURE in 46m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5cd8f7cb9d6467a911c865a9c722f0b : SUCCESS in 55m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/948094690f4c4a27af98a047ad3bb8fc : SUCCESS in 2h 09m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/357afebd46884a54a3ac7eeba4c8a28f : FAILURE in 2h 22m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2373b40fb2fe40ba977c4d60d569beae : SUCCESS in 1h 46m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/23c8290109ee4bef87e35ad397db1769 : SUCCESS in 1h 36m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3980b4450f054875b4818a6fdbfa0f62 : SUCCESS in 1h 41m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6d9d2f36522b4928945166328d583130 : FAILURE in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"75825a131aa134480c91faac5e3c7f0d7432d9b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-10 08:06:42.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"9b02abcf3ea43900ea496ac0dca22f921ec115f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:40:52.000000000","message":"Patch Set 22: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4be3d3c8e9f747c0848ab59fd250f88b : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d43f1d9e08dd43e095ec5e5bea455a34 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9132fd1e55f44e7bb4af839eb0cedde6 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf12e703c684d3c98a1da1937442496 : FAILURE in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/998fefd7275e4f67a111d839ed855a03 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/627c0f763e6f48d79e4fe4319c16380e : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/789914c1a18a4293a99bafb0f7836285 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8728cd12cb4cf3b643dc1d6a42fa90 : SUCCESS in 11m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06bd8cb038284e1b948a86fa64b6972c : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c7a45974c3d438cbf83cc5584a9575f : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/644cffaa94e9436db9328a572db73783 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1df71c33f11426291d6ecb0af4fedbd : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc652f7b7b894eba860c8f3b19484824 : FAILURE in 1h 34m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/690e3723c5ba44de91f3456e98a5e11f : SUCCESS in 1h 54m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b73424a150fb429e9315bc4144f9ef10 : SUCCESS in 44m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4c1afaab2baa4028a1e580d2425c2d3f : FAILURE in 59m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17b1dacc831845e1991dafcee59c9ed5 : FAILURE in 45m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/71642772793d4351bcb3d81aea9b8c9a : SUCCESS in 1h 04m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b1fec3a18dd04cda83dfc4f9571918f6 : SUCCESS in 2h 12m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9847b7b0dc084691a7c298fda9a70d86 : TIMED_OUT in 2h 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fc102917788485082f32f5aab9ceaf4 : SUCCESS in 1h 55m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce86c500a4ff4b588732c7ea018b7b2e : SUCCESS in 1h 53m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0dc2701222b641e1a8ec7929080598ad : SUCCESS in 2h 08m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c2470357b1b74654aff47bf21f8a5a3b : FAILURE in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b99c46ad13a61551848af55cf54d55110487ab77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-24 09:40:35.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"82b508bf2648e696f2d11edcb157a6fcdbc5ba57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 11:56:09.000000000","message":"Patch Set 23: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3dc29ceeaa044c4e8cfd1659d340b7cf : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85dd2ac5acc34a9bb7fa0a099d21c143 : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/351b6a61a86d47a39cdcbfea6b159930 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df67bf781f043c69bb18c06ea21feaf : FAILURE in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/381abbb23fa646b68786522dec68b7ce : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e1da3c4e75d45129609f5d0bbca7268 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87fba73ddb024b3fb53395b67d1871ba : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d93f25e19e4ace9463ea9e90b9bf91 : SUCCESS in 13m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aa954935bed343b693ae747c29ef5244 : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/560652a2072049cba0c680ca47ed63cb : SUCCESS in 10m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/150dac7f8701433e9f0ed53d07862654 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82d02bf5545c4f5495f4e619870d3318 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98b54de12283401dae4dd78854f18945 : FAILURE in 1h 52m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c81809968657446d95b3321b1239d6c5 : SUCCESS in 1h 43m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/79ec2172e3ea4f5bba1e355ca79d5b1e : SUCCESS in 53m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/724538d7ed5b46ebb75860ba4dfee6fa : FAILURE in 56m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04d5e88e54a7419489d7e03a2799614e : FAILURE in 50m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa025ef0f4c24450945eabfff39a5791 : SUCCESS in 1h 18m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fd2903d25c3477cbcaaa094aea6759e : SUCCESS in 2h 13m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c23397aa19824760bc60815e87cd5c9e : POST_FAILURE in 56m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ef932badbce4a398f9ddc677a30396d : FAILURE in 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/464c4b05199c42b08e8e7af094c2bc1d : SUCCESS in 1h 49m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d2727863790948da923524de19598fe4 : SUCCESS in 2h 07m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dc433e3a208e4808a922d177256722a5 : FAILURE in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2ac5a3b933150e0c498dd2d3da05d1104ef9df93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-04-20 15:33:14.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"3533775d1eeea0073c49b9be81453ba1d3f51830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:52:13.000000000","message":"Patch Set 24: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da5dab8180794dce87ac6a53aca1d5a0 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e75455d07e87481690cdb1da0220dee2 : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04327138c43f40e3b7fe236af4e1a0e3 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9286d9e01e1243f59aab8f0adcd53a83 : FAILURE in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30fbfc3614f547abb6ced36bd2d9c171 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc012efc9da844c7b25a829d63147c83 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0505fcac2b34b06940a22211260359b : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2988d8922ab24f359e8e0ac2f3533990 : SUCCESS in 15m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a2f7c04f9554c82a078fff9727948b8 : SUCCESS in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b57273c4e59e4df9af74198d36b9b505 : SUCCESS in 7m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0a1fa40c5154d1f842ab777423e95b1 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a9fb2dece4e499899db0b3b45921658 : SUCCESS in 10m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0db69e80a2e249908eb502a3d83b8ef1 : TIMED_OUT in 2h 12m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34cb7dee9de44c14b6a58f4a13f9c884 : SUCCESS in 2h 11m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef7fa80596d94aa0af7c0272fbb5ce04 : SUCCESS in 53m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/53768a138e7142d2b52c8b0ef6ebee9c : FAILURE in 54m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f598e784b7624297919ba4be7a4b3b06 : FAILURE in 42m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/13cae69da0c749359bdd00cdee02b628 : SUCCESS in 56m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3599ba0d83b4010b5635cfef6df7503 : SUCCESS in 2h 05m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cb309143befc4625afe75932a9b099f2 : FAILURE in 47m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c4306a61f54a5bbec2717e52365597 : SUCCESS in 1h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eea765759d834bc594b1ca82d776f57c : SUCCESS in 2h 07m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0d795f2e55934ee79a441ad4a86ce3a5 : FAILURE in 2h 09m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/69b5f90ea4b24f04b2eb29714e4fb43e : FAILURE in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"eabbbe7cead3531eda4090b846fb86e98ca1ad14","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-21 09:08:47.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"43cede62a1a949359b052ac06d4f156cf190e1ce","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-21 09:09:23.000000000","message":"Patch Set 25:\n\n^ pep8 fix","accounts_in_message":[],"_revision_number":25},{"id":"89eedff303d25f1e9b19cafedc1800fb0f243586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 11:25:20.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a168a5c91424e9496a36acf74f50eb8 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/262d1f4d0d804ad6a6a3daa74cbca8ab : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34bd3954dd424babbc26c6e2d315dffd : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed72695781947238c20df67c06da86d : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2348f7fa1ba4bcab4605ed809be07d5 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40885ef4e58247e48cc2ca3e41ef9582 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8df455ed6844d73b1a51f960ed49944 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c66003e368d407b8ad43ebf70bbd1ce : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a17910adc2fa496986550570afd7318e : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d1a80bee74fb442b8d378d7971302c58 : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc37b2cd572545478f5a8d7a1399f0f6 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65279daf334246d0b9535a2a33b8ca76 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e863553a504d40739db55868bc473401 : FAILURE in 1h 51m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d9e197db7de40a48be217d6e7895315 : SUCCESS in 1h 44m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d7b4f05f3f648239b705c218a89e3be : SUCCESS in 43m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7dbb8218a6a34eeda0c4c65f296fcd7f : FAILURE in 43m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7868e94bf3194dd48c4b153dfadfb0f7 : FAILURE in 44m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/96e6b156d9d743bda1a97e15411f7014 : SUCCESS in 1h 00m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec242b1acecc4e44af7f62eef2ba5a34 : SUCCESS in 1h 31m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/479f27a9ddf940c49f4a731c5b6ab253 : FAILURE in 49m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0580dac1732e4bf4bc48fc571758f388 : SUCCESS in 2h 14m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a7c0d3cdfc6f4b179341ef4f59879e75 : SUCCESS in 1h 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a3e19f98fd3e458ba6a26b70a7cb3ab7 : SUCCESS in 1h 46m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/44121b13c950484790eac123dcd86af1 : FAILURE in 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"3448b7cd421153219325d2a7ba448f19ab745e1d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-23 12:56:31.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"8463619369573d0469c94e084841c6954d05b61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 15:14:17.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/09cb63b4808b485aa6013e23742fe032 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773443198a974bd6a19f01153a93abcc : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a630b1fcbf247c2bab9e0dc954e073c : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928f1e16da904ff5a7c046491d15b1e5 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/270e033ec6e248d69a008134acdc819f : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31b136344c4e4401a995e0bea217e3f2 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e759c89eac7a44c78098d598d54098d8 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abdfb7dcb4354577b0fc836b96d20503 : SUCCESS in 12m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36722be486374b83b0e57438e8ff2bf1 : SUCCESS in 6m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ea1c5cd3d93f43999fc991f62a214976 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6a07453c744048f3a7dfd9256702f0ec : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67e416b2e4114b3090712069b33c78a9 : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/495cdd0ba75740d488c6b8a80eb2c4ed : FAILURE in 1h 58m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f900fe162aa4a49bc6bbf7d095aab23 : SUCCESS in 1h 56m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f44b26e24ec491a84e783bb5bbe7752 : SUCCESS in 44m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/00ed9e707bb64b5ea7a33bda26e6113d : FAILURE in 35m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1d7b23d1f054dd9983f780312964db7 : FAILURE in 46m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f01c7b276d9342cca79ad5a1932f5e8e : SUCCESS in 1h 09m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c971e8675874976b3544d2dca692b90 : SUCCESS in 1h 44m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a3788dc5a7b4b4a83c0fa420956f29c : FAILURE in 1h 09m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/260d65ec80514d55a17d8443fa11f978 : SUCCESS in 1h 50m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7ca23cee662d48b2ab3480665132e35b : SUCCESS in 1h 29m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/09ca91fb29fe4b8ca9fe43de92290b3b : SUCCESS in 1h 47m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a10f0879fee84bb685533b87d95e2ec5 : FAILURE in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"b34730b2cecddc374b5d9afeda35c1d2636a8800","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 19:54:29.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"c22c7defdebc05cda1cc17237783da234dc86b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 22:04:53.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ef29af847824091b4788e6bfa6a6551 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2aeb1a6b3c9403687ee00d18580a886 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/148c813bdfb547e7823761b600c183e0 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c2b1238f504e9b95caae255989d85f : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f74723635d9e456490cd63b1bb0ae932 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b96013fe0de34d0a8b78de14410c4e6c : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26078851bc8a4a7eae63d2f0d6810ae5 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6903b25a03843e3afa61e9856dd9c41 : SUCCESS in 12m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fbbfc65bae3e47c49fcb6d29c6e3b519 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b75294dacf264604b3705a80a3f90d4c : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/554df918e9374316a53e154a65477b7b : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9bf7bb55ebf4163ab7dfea11b0355a3 : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf88c137661347eb95bd21acbb8fd3dd : FAILURE in 1h 59m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7503ff9dd5054770ab99877b18ac22dc : SUCCESS in 1h 36m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1033fc12408747c5bd6848af79e71f8d : SUCCESS in 50m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45cd8ec09f8c4c3ebe97bf41328d1b52 : SUCCESS in 37m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33567c94f51642cc939a6b9879f10329 : SUCCESS in 55m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c6a7df054ac24bd4878b323d08351065 : SUCCESS in 2h 00m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fed95f40afa1440e9e8a24ec536f9b47 : FAILURE in 1h 02m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5575ba1a9eda479ab7c761f5da465581 : SUCCESS in 1h 54m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c14641afafed4e0e851caafa3afaee82 : SUCCESS in 2h 07m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/87c14befec3d4b6ebdf64a753c5357ed : SUCCESS in 1h 30m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/75c6cab5de0d40fa823d8025d0a95f4d : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"c5bcd00e6455b594bb6ec3e3c97352c6033b9e71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-30 12:55:02.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"ae1b1c9a50595d52e53d272d28a5761acaeb7221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:56:26.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12c0e9805e324698b436b74d5c97f583 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fa10fbcdd174663a4bd406355e160ae : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3939d0e5b05b40efbcaf6fe08a388b4c : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998c1cdd601043ccb7bd16d979dbe605 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16c87c670fbd4c218164282455b6ce12 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e19adde8a724d35ab0c1ea4b8056b0b : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bddff601a7e4cfa843f43a01c66fc32 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397f4fa8abee4f1bbd6eee6dbe35fcaf : SUCCESS in 12m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ac1105fbf10c4f669c2975497e5785ef : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6c2f9bb4e91a4361a6071075bb8af634 : SUCCESS in 9m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47081eb3a6234d98a374d426cead36ba : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8998a795ae2842f48ec06d2c7683a939 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8c4d52a5bc9458d854d56d6aee22341 : SUCCESS in 1h 55m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b2a25039ad24db89df6b5a8ebacdf83 : SUCCESS in 1h 34m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/487aa96c460d4aff893252ee9c46f692 : SUCCESS in 1h 02m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/feda9876142d4a2cbec48cc930c89c63 : SUCCESS in 31m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05bb609888974926b964f492294e0a27 : SUCCESS in 1h 01m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/08a81355ad5949f39ac29b627f7781bd : SUCCESS in 1h 50m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66d610f37a364538a60f92cc80739cfd : FAILURE in 40m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be9fb109a1d94be887e3cec86017c99f : SUCCESS in 1h 44m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86522f3647074a0aa358e1e2b1167243 : SUCCESS in 1h 44m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7cb863950ac34ecb82767bd7010a1971 : SUCCESS in 1h 18m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5a69122abbc4410e8b80a7059e518800 : FAILURE in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"6eff3eddf73f51557f396e748de2482c796398a4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-01 14:22:16.000000000","message":"Patch Set 28:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":28},{"id":"3ee6ab3224fb1f8f87b883c5e8b04bcf8511fe0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-03 09:52:10.000000000","message":"Patch Set 29: Patch Set 28 was rebased","accounts_in_message":[],"_revision_number":29},{"id":"8e056f9974b27de8902d453b7e5d2332ab354191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 11:54:24.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7eedd65adc5447fe976c636dd0272840 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07b3e2dc38b64585b5b5e97f4cb55d0a : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6838f4fb6cb74709a6d6c55e5125f461 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343e6c5b45324a5bb7817b5972f757b4 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a6bbff1c33b42bd881f1816241ee76b : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5100f20a7894d61941717374e2c7635 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70c77d27a8274cc08890673418eadcc4 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d5e7aa30174024acaae661faeded3b : SUCCESS in 10m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/336a7bc31283470f96d0b54acc2738fe : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31cfab8a60ed410d8cef70c8a33ae776 : SUCCESS in 9m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b26939f08a984d77ba6c21b544e24c99 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a42ac53c7b9c472f9c3f5c7b13228633 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1b4d9632713437ca8f46905022715e8 : FAILURE in 1h 27m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2beca1b82ce40c39034053923c6edd0 : SUCCESS in 1h 42m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2b3de1b92ca4d61b4f2ded2e4bf6276 : SUCCESS in 40m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62a57dce19564c4b8aeb675c3dc8386d : SUCCESS in 33m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc40dc5968e243189c54b1fc67f35f2b : SUCCESS in 1h 00m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2120870f7d434b08974f52c1dd65d114 : SUCCESS in 1h 55m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8124ea1b37604b8188392eacc2e1d2ed : FAILURE in 45m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4fcbe20cb670405ba2fce70e31ad925d : SUCCESS in 1h 58m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/925eace22221417b90ad6e9abb6899c5 : SUCCESS in 1h 27m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b639df2b98914940b3360b2df2e8c130 : SUCCESS in 1h 47m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d3ec7ca0a83d42149e53b8867d903ba5 : FAILURE in 21m 12s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"c0df7f467296dad633c88d07b3117085f7361628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-06 09:07:11.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"6c78ce3c75bb9dca0038abce7132e8a826ce0f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 11:15:21.000000000","message":"Patch Set 30: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a85c049a7f243a992c13c83fd923baf : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61ccea48fd8747668d11c7ba5584f971 : SUCCESS in 14m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5435bda1ba7848e69478ff44211566e6 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8e10e9c4b64676a7dada10b1fb0fb5 : FAILURE in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/062348f04d554c619b4b0ee2bd6e8f43 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0466efdaa5ee41eb8e05936fecd61a1a : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aee9ca6d514463c857ef1b3f0f5bcf1 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d680b35c65945469f2e37057ed5125c : SUCCESS in 17m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf713007e6f44703aed67382799a2c39 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6cdae6a2aab34ba9a663de80df00d902 : SUCCESS in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a774f5425d8b4814abb05f13a5fc75cb : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84e8d060dc6a48c299ff8af103840966 : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/974e2390a78a4985808f49ffb5a43eaf : SUCCESS in 1h 41m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04609868682a45b5b06f43339aba60d9 : SUCCESS in 2h 01m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/960b20b6964c4b9881f3fa65b03b027e : SUCCESS in 1h 00m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5abec323121a43dda1b5b83771a2199c : SUCCESS in 46m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c275cd2e72347de97657666ba3e77bf : SUCCESS in 52m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6bf157776fa14b3a9153886a3855db66 : SUCCESS in 1h 22m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2668c6db4be74a1184536daf5b82c7e4 : FAILURE in 39m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c0f4f438e6a645668651c0c34edaeb6a : SUCCESS in 2h 00m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96120ce5c7cd42cd93a6f6c1e73b6758 : SUCCESS in 2h 06m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/464e5f787428433d9da4bcdca5c1c083 : FAILURE in 41m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/89f25e7474ff4ddab7210b6bc624e57e : FAILURE in 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"f8e8440bb0e270eb99561dea9cb4f5c92c58c349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-06 11:18:04.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"27f5511201258d89bf036151514f6b590e72e012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:33:57.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/59954f2b9d7a4b63a8735da86a33452b : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f528819730244e64b823eb1704e57221 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dcae51830394e42a0ed3b5f1bb25a77 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65871646bda845d99ef6790717ad3154 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be3df7bf5af643a89ba81266e696a71a : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5367676804ea401189028f380d6f5ff4 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce62d29dc0e41ffbfee9fc1669e288b : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646f6c827e214c76a3500ce621de4528 : SUCCESS in 12m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef2a79ef53e546359afb5656930ca44d : SUCCESS in 5m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5c41b81e4b4749a2802aa8fdc3458fba : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ae58a87abef4b6d838ff01c09158299 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/924ba2a55984460f83dac947629cfa5b : SUCCESS in 9m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fe231ad586284187a6d4b1ae581c0c68 : SUCCESS in 2h 03m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd1cffbe05504a57a3e4ab9bbea1ccd0 : SUCCESS in 1h 28m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3eb085cd27884c44b3fee6cd1800cc27 : SUCCESS in 39m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c863017d88944c69a54436a6c3cef97d : SUCCESS in 33m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57edcd9841cc4e0ea05fe47a537e167f : SUCCESS in 49m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a2ad5087c3743b79ba423d086a80870 : SUCCESS in 1h 19m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbda2314e0fd4ee581cd0c6e695b4cb2 : FAILURE in 42m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/86b3cb09e058443496ea6de13eca8f74 : FAILURE in 1h 59m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b56bbe4a2de44b079778d0e4d103440d : SUCCESS in 1h 49m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f9bf129138d949d8a2c97e5d15f1e0dc : SUCCESS in 2h 13m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b1a3918600a549a3a33f71e0d03747b5 : FAILURE in 36m 48s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"26848eae5cb803a3406051b7a366bb884dcfb18f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-10 17:41:15.000000000","message":"Patch Set 31:\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"dc52e033437aabf69744ec2837d8d14cbc409a46","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-11 08:14:51.000000000","message":"Patch Set 31:\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"e2e8ad80c198b2c636ce994fcb59156cb8d9a59e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-11 08:47:28.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"501fbfcf05a285a75341f983a40656350d79179a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 10:54:44.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/419c165360ad4f45ab7c1673aacddfcb : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79ebe9005eab47219def9de5bb727227 : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c89550c9f4464a23827e5768e645a325 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ed4eb06d844e749fb98e12bce3ec93 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/526bf7428835498e801e1f2422ca4559 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd1ea131f1a14f3ab054fe9de058c7d4 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7270518f5c684270a3fedb4ca814810e : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71065770e86e41d481bcfcd242e28684 : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bde646e1d39f4c8f8d414debc789637c : SUCCESS in 5m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e79ed4f22f4f440b83a513d4c7f18644 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/72e386ba8a544032a0f245573e5b6acb : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/783029553a6f4704b9532953e4f4c063 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/535d971b3ccb4c4fb322aca441ed2cf1 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90170acbbdc54023abf6f79ddec3b907 : SUCCESS in 2h 00m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c8858d9582c4cdeb1452b5d5950dbe4 : SUCCESS in 1h 41m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/94b5abbae0e34f699ba5b250846bbc79 : SUCCESS in 41m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eed7483cb86a4af784aeb34e853b039d : SUCCESS in 43m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b17346ad2a44fba84825f2508742c61 : SUCCESS in 1h 02m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07258b4f46034a72b9d0811cacd11a30 : SUCCESS in 1h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c25898cbc6d843879a8cca26247158ce : FAILURE in 44m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f125b0ada23e489786c9e6a50d1183fe : SUCCESS in 1h 36m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8fb77ec005f142968c8f2ca02e9617e3 : SUCCESS in 2h 02m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d55257d9720e4759ae3e3b29c7218ca6 : SUCCESS in 1h 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/74516772f7284413835594e2a1623e7e : FAILURE in 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"454f5fa084595f9f3defe6088503aa9ab9954cde","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-11 13:30:28.000000000","message":"Patch Set 32: Code-Review-1\n\n(2 comments)\n\nJust one question","accounts_in_message":[],"_revision_number":32},{"id":"a76c19efb8b9b6d23ea329f4d78b3ef98dd23e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-13 09:56:34.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"dfa4d8faeabf421d4612464a3db9c431bb596938","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-13 09:57:28.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"4652399208e6394576fd8a3103f97ccde11b51f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 12:34:41.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/907b5de400ed484c9c2ccb17436b0baa : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40bdaaf1689c47d3a742c847bc27fede : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/972690fb24e24eb5a3c4e323fc227eaf : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a24be1eb63fc4bd0acb11dd20675895d : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f51f242dd4b54d1bafe0aba7207107ba : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d95b5452ab5341129bfd3d62b5989340 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c5a63ae385b4cce80f3e29615ef9370 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac92494b301744b3a81261713c0ecd52 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9eb7cafcf9146a69966de463d80e6d7 : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/59442e6805ad4e3484daa86e543d7db1 : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8a9ed7310c24e919f988ac8be427032 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/22e85fe594da4fa4b28d034679eb7426 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/716cc39325694c94b892793919f3d3cd : SUCCESS in 6m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/580b3302538440499e0a463baa116ed9 : SUCCESS in 1h 20m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa9f4d2e6e03465aa55b6a6eee48ddf6 : FAILURE in 2h 24m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9648b6af5a01414fb66953c56864317e : SUCCESS in 39m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51bb34681a764af390bf68e2d5ae4893 : SUCCESS in 41m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6e2f899e7ef4216a9ec3cfb87c9e104 : SUCCESS in 1h 12m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61786baa22eb431a8f0b4ce8c4459c52 : TIMED_OUT in 2h 34m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fcf1d586efb240febd99fa65c39ba46f : FAILURE in 14m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/354132ea1f784846904be83273ce5c61 : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4b02595cf8b64cdaa3df0f91ba00aa99 : TIMED_OUT in 2h 34m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/15bec18158334579a89762f4256ca17c : TIMED_OUT in 2h 33m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9cd49a5b0fb7469d96d45e278fc76abd : FAILURE in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"1d3e87d21c852777e8988e4060d66b2b3f55f2ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-31 07:51:51.000000000","message":"Patch Set 34: Patch Set 33 was rebased","accounts_in_message":[],"_revision_number":34},{"id":"1de12cc43346001457a98ffbabcaa9583ebb7d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:28:20.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/814ee724d95f4572beceee6b658f7cf5 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43e374aa092d4c25b23d77ca9a89dd45 : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/124b495f5b834d24ad978d434943a209 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad3d53664c754a29ba53a91c2349ba1c : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a71e69f1de34c86b613673db90abb95 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d688bed7b0d54630a925021272d9e1fa : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b1a8212f58147c7974bf939413cdd99 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046d5f46ce93488487257d26459b0142 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36b9abe6cccd4f3980c5871686b583a5 : SUCCESS in 6m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4dda6488ba5a44559b55cfd1c758eb7b : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e68268ca52ea40b088bf24110ea5902b : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aaca69e35d1d4d52a878a4960645de4e : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cccb24d17e6495ab412fbaea730d216 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea9894cc5f3745879b23af6c179e5cf9 : SUCCESS in 1h 48m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4102f5ceac84a73b2acb09b5a144896 : FAILURE in 2h 15m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/577da35242f4483d96aa83d2dfbf3c3f : SUCCESS in 34m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8829038315c4edba26aa5a5cd36e133 : SUCCESS in 32m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/038e4d29e40f4da7a5d5b55b010f7400 : SUCCESS in 56m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f2b5fc2554946d781cfaf6fe50b2a20 : FAILURE in 2h 08m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23e4cd55796a4c00ae1dd6e9989f3e49 : FAILURE in 15m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b1d4d6f50a7b4bc79dd037b1cd869950 : FAILURE in 2h 31m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3fdddb2145748ed8ea8628ebc9bde84 : FAILURE in 2h 18m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e7a58af87bc449f6bcafdaf42d90a116 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1b7db1fc55364e11b5762f27070480cb : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"67b1213e5026746a6ce6c0d259c42f4deaa84c0f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-03 14:56:29.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"c116db9deeefe4f16c0f5631e3c78995fde9245a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:54:09.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12cecf60e7f54fe4887d0dc70588ac5b : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05a58df384ef4715b51b2e5584d9633b : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dea06cb015249d6bd61afc6e7019890 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0082713f81a546b687acdeb16c2e358c : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/928d776528884a169c3465e07e611a34 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c40c8122bd54e699392ec51853c16e0 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b23f5749a8e2468e8a6aab9cf8fa9039 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658b919b4bbe4e44aae305766cc69bcb : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c971b37c6224ca79a880df67bb3b344 : SUCCESS in 6m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/281c20a0e72c43bd84a1cbf7539736c6 : SUCCESS in 8m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91b962edc7a6454596703f31b80088b1 : SUCCESS in 9m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a02c72f5a326474896f439a1f5cd0979 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39489f66b3174158a5361f8a75b9374a : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27490916dc154ad5aa836b71e096b7ff : SUCCESS in 1h 55m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14b36a285a5c490691bbe56c0767b1e1 : SUCCESS in 1h 17m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e083b46ae063450d84ee16d5fee5d7e2 : SUCCESS in 35m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b589e758d5e4887923f1fd69d4e55db : SUCCESS in 33m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/81bf6e84efba45c5ab02c25e84a17187 : SUCCESS in 49m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0cf14595c696430791a4e4496b91562b : SUCCESS in 1h 30m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de6ef4f66cf640e3a357edd5274d497a : FAILURE in 16m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc6ea262185d4f80b6aaf395133b0cbd : SUCCESS in 1h 22m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/897f843345a34b009364e54ca80ec504 : SUCCESS in 1h 50m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/43d582035a2e4fabb73cd5c1f02c200c : FAILURE in 1h 19m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6349c9063ed74a06abcf6496c95191f8 : FAILURE in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"046a3067bb277a4d0f7b6b2303db20e0e980d448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 11:48:16.000000000","message":"Patch Set 35: Patch Set 34 was rebased","accounts_in_message":[],"_revision_number":35},{"id":"2de0759687ae76c6acd5c072811aa05fa149b256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 13:47:46.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a038daa252e24d12a6d70548575e2033 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/592bc8443c704efbb3e68c7435685153 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bfa93bf56f9406f8c79221bb52479ad : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09727b2e90cf4d91a585ae0a05bcea83 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74f275be962849bbbcb9c69fa645a139 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc9497ad0175427b8e82cce49b20f64d : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88f83be1707a47f28430977504e2f6a6 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f25437bd6c49fba0f8348c020cc209 : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d36f271ec59b4044b059755538a071d7 : SUCCESS in 5m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/72caf61e703249269c4aa59430ada039 : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c2370cdded04e2a95546efcd192c6ea : SUCCESS in 10m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aedb2e50740c408aa4480699b0b2a4a8 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a22befb0e3841a5a147257f0397fac7 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbbb183c5f0244979e7d27c3a8bb3889 : SUCCESS in 1h 48m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77b50a437dd646f2b0dd043ae3d32ba7 : FAILURE in 1h 34m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/272646e127a6471d8733b12e07a766a7 : SUCCESS in 41m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c02d0d816f145c695b918781391c6be : SUCCESS in 40m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/833782516ca347e1adecaaf5d34d6194 : SUCCESS in 50m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bdd1506f5e4844b19e82edb0448b1d72 : FAILURE in 1h 29m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/149848fa25134d889f455c7f2ec18015 : FAILURE in 16m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a09eabf1e9741438d4a49cd453dfeb6 : FAILURE in 1h 55m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2dd83a85f32d4bf2b983307b5f2c6fa5 : SUCCESS in 1h 29m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c070ba3e43aa4c78bc73d899467e4e91 : FAILURE in 1h 21m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2c67c525fb084ceaa18ed488a07e8509 : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"516e69e227a5c9c18b018e306d64f47052ac319f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-09 08:11:53.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"32aaafcc612be3248ab8856edd3b63aab541e320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:36:26.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f228c92ea71431792775ad871f2f593 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc987c88a2614abbb5f91ae80b7cdeba : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c582aa6163d42a6862dbf7aeb584db7 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84dcff84bee4e338e362cd5676da88a : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/868c23972ffd4ac38c117cca6db357cb : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fae2d1b3faa4c4e9d011331407f4c45 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6281e34924824c14ba8bf676337ce5d2 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb72d47463bf4e4a8fb00abdb4bca515 : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f70db1145244a1eaf0cf665d1a87ea4 : SUCCESS in 6m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf3974f60a87437fb66c5c38d5019cbc : SUCCESS in 6m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/858edbf9cbf445018981d6702bf7afeb : SUCCESS in 9m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9e029b39e18483dae3d7acd9556d94c : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2aeb2bd93ad64809b36c9cf6a3e1491d : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3324e695727a49128abbc95dd01360eb : SUCCESS in 2h 02m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ed97d32045249bdaf79f298286d3f7e : FAILURE in 1h 35m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d311c1e11c9457987d49b0825d32238 : SUCCESS in 34m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5344ac0effad45779b9a5d2a3d52459f : SUCCESS in 34m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce8fc32413f94afd9a1b3ab01b05de3b : SUCCESS in 53m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/121b4a326b0c4a01a72ab2994512a33c : FAILURE in 1h 38m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/950f15a08f15416fbe37675e8c3751b4 : FAILURE in 17m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4194eb61f4ed445e9dfc2c0fc65be496 : FAILURE in 1h 48m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82c2f93333d94727bcd238b9095f5db2 : SUCCESS in 1h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/90caa38d04e1440c8ddc116ee7c53376 : SUCCESS in 1h 30m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/10d231447b274cbd9a3b77419a8fc42a : FAILURE in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"611761b43bc9244172c008189b041f4403ff4fdc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-17 07:58:01.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"862589da5a834fc56e8e2028b03ff2b4ce040172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 10:21:01.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6dcf875fd1d342cf84f8a96e4c7b65b3 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2434804b79f4414ac524414cd90ab99 : FAILURE in 12m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/050b69b16fa146aaacc339a110fe5d16 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cfcf1cfc247412e931d81e08546ccae : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad27a6e03fa243c7b04a5aef5576342f : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a05e3d09ed04ac68f5307de1d47f28b : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc50230871964291b11560e4c0a5aac4 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44721d9c78d4ae4af07f3d09a01f3a9 : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a9a1a2c6f924f2b897253c0b05baf75 : SUCCESS in 5m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f00718b4cca4db3a009781e9912f9d8 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a8e5747c319e475a88dd727f34e121e0 : FAILURE in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/faddbe4a20ed44f3bde3dd2028a7de51 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f0eefec593b44b4845c20b7389560e5 : FAILURE in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb8755f55e88470598dc2c5186c2571d : TIMED_OUT in 2h 13m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fbcfff3badd4f988136bfaf5096ee19 : SUCCESS in 1h 27m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9dfadc9119b142dfb86b683fa976b794 : SUCCESS in 43m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6934fd30771a453295061df5fbd5b4fd : FAILURE in 40m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6574335d616b4ed489897a5e53d97cea : SUCCESS in 1h 10m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b0aea5ef2d254f79b66864a53ba08047 : SUCCESS in 1h 29m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1cead423465c422398036a5b7838df6f : FAILURE in 16m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e8d7e0e33194f9a91a0b4fe74ff7ed7 : SUCCESS in 2h 02m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/80f9164254ba43d9a93df5737187f961 : SUCCESS in 1h 42m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/44cdc4fd2c184ceeb83a8a3f42a18036 : FAILURE in 1h 30m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/38b4c7d20fdd4974ac9baf2447bc0800 : FAILURE in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"59bfd0d13cc8def73f1bfb933b89891e4058e59e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 02:44:30.000000000","message":"Patch Set 35:\n\nrecheck\n\nSaw some remnants of the sql failure Greg just fixed.","accounts_in_message":[],"_revision_number":35},{"id":"8c046b583cace6e448196558717222d1e599181f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 05:34:23.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15cdcbe16862465597b4d858660883ab : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca9075161e104c46960a8dd442ed3def : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deba9866e4494db980f3a60aaf930b73 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2111fe98378c49fda49efa5b7141cb4a : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/729e17bd029746b4948795b225e6e1cf : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa1c5a86f244c009256d349c740fdcb : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c26ebdb99474494facd00a9c20ee6ce5 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4ac093b4014d038a6b67b117b114a2 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/992cbe131a844b9c8c36127220b126a7 : SUCCESS in 6m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/142f7aea66394bf29f86746c077cc89d : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/896adc128c9746f6b2bbeab5b0984b07 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef2b34b09d7447a9b31832026e584793 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8dba6439a1f43d291776c7f8069fc40 : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/378bc31abdc043e9961e0a169262e6c3 : SUCCESS in 2h 06m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfdd5fd3e34941cbbec8dc23a28502dd : FAILURE in 2h 29m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d7c032796644a6d8853b52b870ecd7a : SUCCESS in 38m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3b5b11a1eb045729324d91c4ab6f6b3 : SUCCESS in 37m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f32e2d071b04fb9a8c9e108c2d856e2 : SUCCESS in 54m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b40e56f102e74b3eab1476995acdf0e0 : SUCCESS in 1h 49m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c07dffedcee4f339734ecd8192ef68c : FAILURE in 17m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ec3d64d7cc98499b86a37ab5115ee1a5 : SUCCESS in 2h 19m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8d473560dff24cca828611a5d46a1eab : SUCCESS in 2h 17m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b638562d8d21479e9454b2749fad42f6 : SUCCESS in 1h 56m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ce08ec3507f247088c6a79c72927db7b : FAILURE in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"25a2f6d7970e1dc294403d80ef2fc84562514deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-22 07:17:50.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"eb74d10eaa8292559e9821ceba587028172737e6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-22 08:31:32.000000000","message":"Patch Set 36:\n\nPatchset 35 octavia-v2-dsvm-scenario failed with:\n\n{1} octavia_tempest_plugin.tests.scenario.v2.test_ipv6_traffic_ops.IPv6TrafficOperationsScenarioTest.test_ipv6_tcp_SI_listener_with_allowed_cidrs [134.232904s] ... FAILED\n\nCaptured traceback:\n~~~~~~~~~~~~~~~~~~~\n    Traceback (most recent call last):\n\n      File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/octavia_tempest_plugin/tests/scenario/v2/test_ipv6_traffic_ops.py\", line 329, in test_ipv6_tcp_SI_listener_with_allowed_cidrs\n    self._test_listener_with_allowed_cidrs(\n\n      File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/octavia_tempest_plugin/tests/scenario/v2/test_ipv6_traffic_ops.py\", line 514, in _test_listener_with_allowed_cidrs\n    self.assertConsistentResponse(\n\n      File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/octavia_tempest_plugin/tests/validators.py\", line 351, in assertConsistentResponse\n    self.assertRaises(\n\n      File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/testtools/testcase.py\", line 489, in assertRaises\n    self.assertThat(our_callable, matcher)\n\n      File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/testtools/testcase.py\", line 502, in assertThat\n    raise mismatch_error\n\n    testtools.matchers._impl.MismatchError: \u003cbound method Session.request of \u003crequests.sessions.Session object at 0x7f61e086f160\u003e\u003e returned \u003cResponse [200]\u003e\n\n\nI proposed a fix in https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/797387","accounts_in_message":[],"_revision_number":36},{"id":"6c5b94d744e4df3edae1ecff11000eb1388845c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:31:33.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9000cc30021f4233a985ca0108c8a665 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecc9d4d4457b41848018b1fb09171a46 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f70fa78e8c8b45d18b487232f7294bd3 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a4ecc1096b48ba9e6eb2ed235bb547 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1a4743f7274243a12d82e040b8c43a : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/068d77b874f945dd900774d9b1fb26f4 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/113b6be17bec419ca7daaabdfe485ff1 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54518933fbd04c108d52deb515099e0e : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03daeba313344de4b6ef1fc44459065b : SUCCESS in 6m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6bb74a1ff49b49dda55eda8fe22f754b : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7b152781f5f744eea4f1939231a8b9a1 : SUCCESS in 10m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29e6e17e31f545098c8d07516d5decd8 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1a822f0fb734edc8d3c291f3ee378d8 : SUCCESS in 11m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/604e93ce556c43feac8a886bb58325ca : SUCCESS in 2h 05m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70fffba74ee14fdfb544fa21a924bde6 : SUCCESS in 1h 41m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef39d5de69f0471495c9ed4a38a33ac9 : SUCCESS in 48m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c72976c0162434f87c23cc435afcbd7 : SUCCESS in 45m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97957075eff94fcc830051440903af32 : SUCCESS in 56m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4035d157fdc940e1aa7b5512a40a371d : SUCCESS in 1h 27m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddb08736253548aaa6a01412a7e12bbe : FAILURE in 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/586013aa44864e0eb1452a045e9fc27e : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1dc5c8f94cca421c826b224e742c5bb2 : SUCCESS in 1h 27m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0beafdf5d8ea481f892993aebc880e6c : SUCCESS in 1h 22m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4809a1f3801e45288fd1f011c3d833c3 : FAILURE in 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"68e2a331a6dc90f43d42291d3517c6a734bf4033","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-29 14:02:22.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"12fe47e7cdbf269116a88ffc06cb2d3d11f0f487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:20:03.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8facc90864504fc9961fec1e0171ee9a : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2052fd02b524434e8f5c45e25737c0ba : SUCCESS in 12m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fa87371a33247a79404ccb187f9189e : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e911696d86b4ceaaf6601506d9449a9 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49b5be349cb247bb93ac6b2423a967ca : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a9e5aa1d5a47a1bda703e169d9bbcf : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf33d8edd6f546bab19515c48212008b : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e0173ea59c43cc8790d1f3c5862317 : SUCCESS in 24m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eb1b5741ae44271b39e0668fdb5f735 : SUCCESS in 7m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e903df1032e4471098d3c03363c3d509 : SUCCESS in 7m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e2368871f1e74a4e9e4651cccddd7500 : SUCCESS in 12m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5fbac6b6633b4f5094d5eb2c04dd56de : SUCCESS in 9m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c172e9f3f884e469f6102e5de3467a9 : SUCCESS in 10m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23a870041a104a25b19f03f7ecd60cb4 : SUCCESS in 2h 08m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb3efba725c64e41beece7195d25425f : FAILURE in 1h 29m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0d759c931994a4eb575e038b2a0e70b : SUCCESS in 43m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5629f36588c64759af292371e2c49a59 : SUCCESS in 41m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e4ec1d5e2c948afbc925439d6fdf272 : SUCCESS in 48m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/87828bf905ce443d91a0b7d0f6083cb2 : FAILURE in 1h 39m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f75b13423b04d189e829726b5b6f9e8 : FAILURE in 17m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/546669290e744a2c9b8f8c31669abf34 : SUCCESS in 1h 40m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfb5ae40e874424585fe30bd3c52a512 : SUCCESS in 1h 48m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0ac7596a06b04b9596ba1cb404e0c0c7 : SUCCESS in 1h 24m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2ae2d1137c2d4807b3fdff466faec763 : FAILURE in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"9aeb9714ab7d5217e7a408b0b9bad6aad70d4de5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-01 09:13:08.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"87f062dbc3640da19d620f7b5e4cd691e2580da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:46:04.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42122b76157246d992f1f7dcb178d840 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b04a75c5961949a3811b725652d4c1fa : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f3a3f6a89fe47c09fed8fccf0a45110 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a42a896e7e48b7abcd5280714ecc59 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0490882686e47cea4efbd78d4b926bf : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2bb8cd48fac48758fae6dd3fe390b72 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1b4ee393617474eb8a31de1c7f0b2ee : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0641b67ff404ba2ad5e41483290620b : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5d0a261478241e48b49e3c7c145bcf0 : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7e47986c884d488cb9ab3873758ae271 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b35f360c172d4e609a60659431f4d309 : SUCCESS in 11m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7bdcdd166afc43ad8997f236d947dd92 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/737cb29f00664280ac20c9ecb628289c : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1d01c9d7884470f97da44098e48ae6e : TIMED_OUT in 2h 13m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed832cd114a2423ab0fac8f8fcd93d67 : SUCCESS in 2h 31m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd72be1df07c4cde9b1e12974126361a : SUCCESS in 46m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a68b64bbe11149e687b6f3c357f274fe : SUCCESS in 45m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/476b5aa125c3449488f9002aa1cdc56a : SUCCESS in 1h 15m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf6ffbccf5c847618f82dc34b832b2dd : SUCCESS in 1h 40m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ce9342da40d4cf3aeca3487d4ba399d : FAILURE in 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d851e7c6e24678babd182f037c5fe0 : SUCCESS in 2h 07m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9e8c1b6cff41455c8747b395bf06c414 : SUCCESS in 2h 12m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1384c7522d224aa2afdca5f534c22ca3 : SUCCESS in 2h 07m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/daa8002e4c784593a3e541ebe542ff22 : FAILURE in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"10b0caaa773a7fdcee9dcfb1127c9c6fe1a1245d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-05 06:59:56.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"f7e664408cedf4657456f19a85953982b2ef346c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 09:24:30.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9e0251ed065947119c01feac895c66a7 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/953e7c1aead14d74b8f68683b744307e : SUCCESS in 12m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d14cacf4994b9482579341878ed114 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b270f041cb4d81a816b2bc783ccd21 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61dc5f96f988485ea78067b6d11b6f86 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbc9e02d07d341ee978df7da5c1800ff : SUCCESS in 15m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e741a8cad46743e9b04c8f7d6fa7d415 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7700de0846d4ff2ba02ffebdeecc726 : SUCCESS in 21m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5052ca9ea4de493facdf3760672533e4 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea964930dd8a4fb99f940c1554ae47df : SUCCESS in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d32a0d81b8534338802dd142ec6fd726 : SUCCESS in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b5a86139bce747e1897b6f5461ad7f2e : SUCCESS in 15m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fba41862451427a90eb45113935513c : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48b975a18d894c6da601291bf28b39bd : TIMED_OUT in 2h 13m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1f54607fb844241be31ffc8153c79d9 : SUCCESS in 2h 06m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83ed072ba35c4f9590345fdbdea14f81 : SUCCESS in 34m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf9dcbe6236e4ea180d6237c045ba5ed : SUCCESS in 36m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6261758c7d774ff5b2e8ba8b6ec1cf1f : SUCCESS in 48m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce7ea21aab9f496f990494f1b378764d : SUCCESS in 1h 52m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8aa32b1b7c1b42a7baee40f5b4c5c021 : FAILURE in 18m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/720ed014b5c24c40a528747af2fd460e : SUCCESS in 1h 58m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2ea240bdae3d425c8ce366a7fff92d45 : SUCCESS in 1h 58m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/293116329d9c4616a71761afe508181e : SUCCESS in 2h 03m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ccf1fcf1ef8948f9ae761300340baf46 : FAILURE in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"e13cfcf653ca064ff263b0d7a95c06d3d3628d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-07-08 05:37:38.000000000","message":"Patch Set 37: Patch Set 36 was rebased","accounts_in_message":[],"_revision_number":37},{"id":"1bcf56ace5eb597a571c54d5e13a7d345263e492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 08:09:56.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e204a20e1d948628c4a414d98d3507d : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebe4b143d0f3418790c6be54b071ab25 : SUCCESS in 11m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae07272fc7e4ac88218ed18ca782872 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a95fa4b72914056a0f2fce0bb5c4175 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7de9742e1064e6ea4404c94ead78d72 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/826a4bdcb3bc4c059b3753f1645d97fe : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b559d47a1444f25b091ffe8c544b21e : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c910dd64650a40f284a146efe6536aea : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/939029ad89324304840740c1607af7b6 : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b41d704e2ea47a38fc27a01da864c08 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99147706c1b84ec3ab8cbb8543c496c3 : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b474954d20843578f55547860f3100f : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/782a4f4b5d8c408ea09d39ecbccde8fb : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/533ce66edd50415a8481b6c38926733e : TIMED_OUT in 2h 12m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3dac8944849445b2b40828e001a1d183 : FAILURE in 39m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/298085c3a8f442f3b00d8d99c168e99b : SUCCESS in 58m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0172aac37234f6ba63e08912b5f725f : SUCCESS in 47m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cdd1096812a946008435e1e2e9d55f09 : SUCCESS in 1h 04m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8ea99d28f954073b6da9d295a944f1c : SUCCESS in 2h 30m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5388441d96b40bdb86e2fedffd1c248 : FAILURE in 21m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b7d13502b6a645539760716c282e3dab : SUCCESS in 2h 12m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/969c4b197b9b4950ad8547894d41137e : SUCCESS in 2h 18m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fc8e1f4a6f2740f6a649659e84e98da7 : SUCCESS in 2h 16m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f230f080b824439fbaa7e8578cdc00fd : FAILURE in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"9b9eb2f530abffb387e7a168ea50d2a1b620706f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-07-19 07:57:09.000000000","message":"Patch Set 38: Patch Set 37 was rebased","accounts_in_message":[],"_revision_number":38},{"id":"0a0934a2b88504814a9d8cab42458b3911e56f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 10:05:28.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e9a8562297c04a37b05cd23ff129f43c : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/368701dcd2fb41c18746650e9e1bdb83 : SUCCESS in 10m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2df0dd1e75af4ed1bc3e49655ec8d0de : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1290ade627a4201bdd0fe8e3184b974 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c459cd90b55047e5a11d90963486376f : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf0f52d84dc145e08ad121b2abe01d65 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c14f4ae9bb14a92addef032921a9101 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dcd325a5d874c8b9afb430d72f3842f : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a96686f4297b41c98058d5a739f3930c : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/60de052bdd554a4f80db07331c7a059a : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f550bd7a6ed4aaebae888043a9f313b : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b4b6824ea11a427ba4aea98397e65d77 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa7a256bd27c4b6d9ce998ef7277f320 : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/641db26e126e4707b57b3e78df68afb8 : SUCCESS in 2h 06m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/264005a9b7b14b4385c3a71b54f40cc6 : SUCCESS in 1h 57m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d192be9ad62a45058ced3e2900acffcc : SUCCESS in 45m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a85220ee36342bdaab974a720186732 : SUCCESS in 47m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/398c668bcf8247588c1023698c70d022 : SUCCESS in 1h 00m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce4b6330e62040539269a87218d8e4c5 : SUCCESS in 1h 25m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/764375590ce84f0d85d3c243687095cf : FAILURE in 17m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/abe4718b75844c959f22cd3c8741e5bb : SUCCESS in 2h 06m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a24176c4fbc94599a6452b88f34dc0be : SUCCESS in 1h 51m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/83e4dda718c54f4f86ac6047c37b4f09 : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4b7e4dd2a58e4eafbb6530f129d28b4e : FAILURE in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"da08fa17407d2376893a981a2978bb32f29b1886","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-18 13:08:32.000000000","message":"Patch Set 38: Code-Review+2","accounts_in_message":[],"_revision_number":38},{"id":"38c9fa2af9b4a14b1458947feee93c9ef52ef7e4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-18 22:01:29.000000000","message":"Patch Set 38: Code-Review-1\n\n(3 comments)\n\nI have a couple of questions about these changes.\nWe need to make this switch, I\u0027m just concerned about some of the changes.","accounts_in_message":[],"_revision_number":38},{"id":"d4873bc7d78f618b7a3297eed4ff2e16e6e47029","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-08-19 14:18:44.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"9d67613fa0e161dbccaf629cd93211d027bd1ea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-08-19 14:19:24.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"fe905ed9e62dbc52a189045f52b97dde0df0fde4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 16:39:56.000000000","message":"Patch Set 39: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9b7077d302bd49428a067297e4d11768 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5039602984764af581a1cecf3bf4af81 : SUCCESS in 14m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55b157a5db3f4063bfe733581a8f779e : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d10327d388c4026b46ea4186264aa8f : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54c8b024f81e43b6a1089ed049ecde15 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76cb475b2294628943df9bfb8912c63 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54941d02b2454cec849ccbbeb466803c : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b8ff1192664dfa84903a2078c1ee90 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db5bbae048b4400abe3b4cebbe73e972 : SUCCESS in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c092a0a5c7974e3fa09a3bbd70279dc0 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bcb9b4dfb1e34ceda664699895c87c02 : SUCCESS in 10m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/52b0de25486c4d4ea54db95be5e950eb : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/729379dd04e34abf81754ea2582b3618 : SUCCESS in 10m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/affe8e599b1b4f239db36e3310df10ff : TIMED_OUT in 2h 15m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d23ca496d1ce494384325dd54b425ec8 : SUCCESS in 1h 27m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9389dcdea30847f2b0a11ebd0f3348e1 : SUCCESS in 37m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af04279b5e404c5fb240b6e820473b0c : SUCCESS in 35m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6f6095332fef41bd8813776a5a66390a : SUCCESS in 49m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/98bbb35c7a4541a2aa1c60b4bda67300 : SUCCESS in 1h 29m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8404a14938ea4ae98bcc5450c756d96b : SUCCESS in 2h 01m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c526bb461a2f44e7b16fcb2fb2e099e8 : SUCCESS in 1h 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5835a8133e224ca697fcfda9624bf85b : SUCCESS in 1h 24m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a19976f082a34f9b93c4b68a89e609d4 : SUCCESS in 1h 30m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/64c0a7e5a5ea44ed8c0816d62e867eb7 : FAILURE in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"8209d3349b0c0f7defb691edff3aca76571ac7dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-19 20:48:43.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"78614dc7dcb2d242da180f2915fdcb7f7ccff482","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-19 21:09:22.000000000","message":"Patch Set 40: Code-Review+2","accounts_in_message":[],"_revision_number":40},{"id":"c5fa72587a317b5c54af3dfd99a43a200f5efb85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-19 21:34:35.000000000","message":"Patch Set 40: Code-Review+2 Workflow+1\n\nLGTM, Wahoo!","accounts_in_message":[],"_revision_number":40},{"id":"fe98c2979c16e4295181db7ea6af7d3aaab67786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 23:19:34.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ebb10b1ec5a04448b95ab5a9c83d6847 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23aab7a295f14c42b6df348a0ebcbfb4 : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab9ebd40bdb48aeb9b1e713752af2d3 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd77eb57f68146848a35684be976a294 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4935dea073ac44be81f720b81ade9370 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b70b1c5ee1a24628a309096c6fd345d6 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b57c260ab8d14120b9daa6fa4e42b6c5 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c6c4656f234806a3d2ca92f534f735 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d5192090904e51a92db5715ab8f486 : SUCCESS in 7m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/306a4c4baa0d41eba60341a1a937547c : SUCCESS in 6m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f1eeb9bcced540e8b49c95dcb3108be0 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/80308d70fc39406c87ae50792894dba1 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bbbe4c80b9864ad4a7d5c18a115861e9 : SUCCESS in 10m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8ffec87c64b448a933e9188035a2ec8 : TIMED_OUT in 2h 14m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a01f5af6314542d881e33bb1ca4c0a86 : SUCCESS in 2h 28m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3ab06104405410f9175ec6916dfd6f1 : SUCCESS in 45m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be95814742e443838a78a38eb7579d87 : SUCCESS in 42m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3346a818efe4b5794d483c6501eabf4 : SUCCESS in 58m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c77787baa4c145b7b33c0cd02802a738 : SUCCESS in 1h 42m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8ade537a52049ce8eac36fb6472eb21 : SUCCESS in 1h 47m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/832e0470e49c46e39a1aa1c3d75da7ac : SUCCESS in 2h 20m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b07158f7e6584ead8764ab0d17368c51 : SUCCESS in 1h 57m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ef8aae8e26ba46e596fa8fd4c9d636c1 : FAILURE in 2h 15m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cfed8b4703524a03ae892b95edbd85bf : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"0f84ca55e148883b2882047a211ac16d2fb5e319","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-20 07:24:18.000000000","message":"Patch Set 40:\n\nrecheck\n\nnoop-api timeout","accounts_in_message":[],"_revision_number":40},{"id":"25ce0b4b183f09930a1ffc93e34718796db18ac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 09:52:13.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a56a36e60b6348b9bf30dbc65b7af381 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6244290a723049fdb7ef125d4f2591dd : SUCCESS in 13m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da372679c41f423aaa56bc6ad5186e1b : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb9828d81a2d4abfbebad2cbc0c6befd : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd5f17cace64904ac33b7cd5b73c710 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/414c548289b643a591f120cf38eaf2a9 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd889ba52ba140f3b196f77f5ebf5fdf : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08970f89067845e8864f01ddc3dc5a24 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e200f0196c94b0da622b638cd5c05d9 : SUCCESS in 7m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1b8c2c032d084cd1b3e3ae635c0f03fe : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eb5243ad3e0f488b9f83f4094599e0c8 : SUCCESS in 11m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/abab5ebc84fe4bdab4d1c4f1df252c54 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8486906938a44c8199d3c50bb7b3cb8c : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/380ff5badfa346a2a529e6b4cfc27f97 : SUCCESS in 2h 15m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb49e3fe74f943fea0cb754986dfdbe9 : FAILURE in 1h 52m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fb418675b8e4ec19dc3039d9e39b685 : SUCCESS in 46m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c942c7e23344f02821c292736fa2a92 : SUCCESS in 43m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5393e6832200436dbb8d1f9b4b7f6032 : SUCCESS in 1h 05m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/298c40a1cdec452f8d9e6e0b60e01a85 : SUCCESS in 2h 10m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/328fdb521fb6402e9e4996830375c388 : SUCCESS in 2h 25m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/049b19b232b24127bda0b1d0b970be70 : SUCCESS in 2h 17m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10e3964fbdc644fe8aa5f67d082afd7c : SUCCESS in 2h 20m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aea0d5ad64c142e1928568bc0d00062d : SUCCESS in 2h 11m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/15f70bb23796481aab44131904136d65 : FAILURE in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"e0873690183f7d1c30a2a69bf32ca55567ddf9b6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-20 09:54:44.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"c52808c7b4c84fb0fd9dc793aee47584412ba3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 12:20:35.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ff5975b5dcc4e9f9a5ef8b75dc3f9e2 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35f8a9bf22cc431097e40e4be72a7feb : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d11ab4eacf134950a2195e6af1dfe5ad : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e16d4353d34e9698645dca20b43d1b : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd49ff1c391d45c99cb37e988da56718 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e23614fbe1464ca99396c33c64c34471 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c39a9ed411b94c7d8d40eed47a3b9622 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae538829bbd40889ff394b8f4494c33 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1c6184597364953b3d89ea3778e75bc : SUCCESS in 5m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/83a5de903b7041e79934edae4dbfd344 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b291b55419f84ebeb756b3ce6133a13a : SUCCESS in 9m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7eac4cb9723c4994830b968b55490b98 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dac72b038e774f619db55fe90c3c5090 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86c16f319d704679b9ee21c647c956dd : SUCCESS in 2h 02m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e18efff3bf048818e2ab014b800fbfa : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30a991a26661422cb8ad78f8cc110057 : SUCCESS in 42m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19bc9bd54c1f47afb4afa4f11824a4b2 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3aebae0791624316839ff51f317e1568 : SUCCESS in 1h 01m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9083bdf478b4c5daefb8f58acedd0e1 : FAILURE in 1h 51m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/877f31f6c5664ea593f86f1fec4c68e2 : SUCCESS in 2h 04m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/835597b1c84244d98c77bc0a866ad159 : FAILURE in 2h 08m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bcae108821b447d8842a53b13b9ca068 : SUCCESS in 2h 22m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d98dc14ea70c4d4982ea32aedd99d4c8 : SUCCESS in 2h 04m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c925ef032b324dbbad649f4ab865f3f8 : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"58c17a1ebdcd9865bd065558c981c27b68c01029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 12:21:10.000000000","message":"Patch Set 40: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":40},{"id":"b9b90ac5ebade89d6fd2888ac6a7418e2350b90d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:31:16.000000000","message":"Patch Set 40: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bd3a9c9884e441b489d83a35c4de03c8 : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b9c635b20214ddbb171b13330ef6924 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66673a17e2c24ea289f491b0ad69ce63 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6487c543d8814352902074db3d7a0185 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c95d42e3016475385e4c975824a49f6 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0738699f59db4acf9ef4dd830efcf726 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cfa20448f884d72a8cbe6423c2cbb40 : SUCCESS in 6m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ef0c5cb2b91414b87bcd5de63eb7c79 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67b8a8f87e74456a8dd6af7afbd6ada5 : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3ab3da8d31e46ecad7176e92a0ce811 : SUCCESS in 2h 07m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/573d2e3cdce54334ba2b7e1291312a57 : SUCCESS in 1h 52m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ecd51cbd65947ffbb9eba73916a24f6 : SUCCESS in 1h 01m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47ed304c483d4abbac1b98dfe99781ee : SUCCESS in 46m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3dab7a47db844af826c507f145099fa : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":40},{"id":"2ba1059971b1c30010c7931f85ca6bd2987c9c70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:31:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":40},{"id":"061ddae3122a8aa1c93bef7344325b64d15ef29c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:34:20.000000000","message":"Patch Set 40:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246de005ee414b69a34a120619172e27 : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0454ddc7a07447cd87d9e9dacbcc5f11 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":40}],"current_revision_number":40,"current_revision":"1aef468ec819802bee3cf43f4eeb1fa52c5b01c8","revisions":{"28cafe91bcca4ee0843a5f6babb14f1521befe94":{"kind":"REWORK","_number":1,"created":"2020-07-10 09:41:50.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/1"}}},"commit":{"parents":[{"commit":"54b536974799e9f31d2b708f84d0ca995e5580de","subject":"Add experimental amphorav2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54b536974799e9f31d2b708f84d0ca995e5580de"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nFor testing grenade job.\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28cafe91bcca4ee0843a5f6babb14f1521befe94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28cafe91bcca4ee0843a5f6babb14f1521befe94"}]},"branch":"refs/heads/master"},"5920f5716b9461da8e7b69c8a883980386e629a6":{"kind":"REWORK","_number":2,"created":"2020-07-10 12:14:30.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/2"}}},"commit":{"parents":[{"commit":"54b536974799e9f31d2b708f84d0ca995e5580de","subject":"Add experimental amphorav2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54b536974799e9f31d2b708f84d0ca995e5580de"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 12:14:15.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nFor testing grenade job.\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5920f5716b9461da8e7b69c8a883980386e629a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5920f5716b9461da8e7b69c8a883980386e629a6"}]},"branch":"refs/heads/master"},"037ba0fb686331012318a86fb08fe03cc1db891f":{"kind":"REWORK","_number":3,"created":"2020-07-13 08:42:20.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/3"}}},"commit":{"parents":[{"commit":"54b536974799e9f31d2b708f84d0ca995e5580de","subject":"Add experimental amphorav2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54b536974799e9f31d2b708f84d0ca995e5580de"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-13 08:42:15.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nFor testing grenade job.\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/037ba0fb686331012318a86fb08fe03cc1db891f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/037ba0fb686331012318a86fb08fe03cc1db891f"}]},"branch":"refs/heads/master"},"5e693ae1eeeb84c1bf6aada043835f8fd0a81d02":{"kind":"REWORK","_number":4,"created":"2020-09-18 09:01:37.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/4"}}},"commit":{"parents":[{"commit":"c3ad32d1a8b542adceaf095bbf0e5229cccb3316","subject":"Merge \"Switch grenade to run the octavia smoke tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3ad32d1a8b542adceaf095bbf0e5229cccb3316"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-18 08:56:57.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e693ae1eeeb84c1bf6aada043835f8fd0a81d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e693ae1eeeb84c1bf6aada043835f8fd0a81d02"}]},"branch":"refs/heads/master"},"57c6122d3a045af65f73d8e7b4a10032b5ec55c4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-08 12:27:07.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/5"}}},"commit":{"parents":[{"commit":"e17c05afdaf8b4f8858af88b4b1e829d6fc33668","subject":"Use run_flow for failover functions in v2 control worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e17c05afdaf8b4f8858af88b4b1e829d6fc33668"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-08 12:27:07.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57c6122d3a045af65f73d8e7b4a10032b5ec55c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57c6122d3a045af65f73d8e7b4a10032b5ec55c4"}]},"branch":"refs/heads/master"},"7d4ca6393622934072b130236786a058c9e7fb9a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-27 08:49:05.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/6"}}},"commit":{"parents":[{"commit":"a9a817e428a5ec71c478a48b4e29378b9daeb410","subject":"Use run_flow for failover functions in v2 control worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9a817e428a5ec71c478a48b4e29378b9daeb410"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-27 08:49:05.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d4ca6393622934072b130236786a058c9e7fb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d4ca6393622934072b130236786a058c9e7fb9a"}]},"branch":"refs/heads/master"},"5a27e98026f5c3312aef510be9706fe5302a27b3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-10-30 10:45:24.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/7"}}},"commit":{"parents":[{"commit":"14246048d117102f24dcc6155169dd730401b413","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14246048d117102f24dcc6155169dd730401b413"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-30 10:45:24.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a27e98026f5c3312aef510be9706fe5302a27b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a27e98026f5c3312aef510be9706fe5302a27b3"}]},"branch":"refs/heads/master"},"da5b160fc70ee90a534afbeb2f9c5fe04e718f77":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-11-03 09:01:16.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/8"}}},"commit":{"parents":[{"commit":"4d639f803e79f6cc68ff13bdf90091a81bf3c207","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d639f803e79f6cc68ff13bdf90091a81bf3c207"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-03 09:01:16.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da5b160fc70ee90a534afbeb2f9c5fe04e718f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da5b160fc70ee90a534afbeb2f9c5fe04e718f77"}]},"branch":"refs/heads/master"},"8b66db467d95cf0783b86abecfede3e3adbe1947":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-11-17 07:28:25.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/9"}}},"commit":{"parents":[{"commit":"bb1ad104d21ca34d4bbc2cfbc248614de381773a","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb1ad104d21ca34d4bbc2cfbc248614de381773a"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-17 07:28:25.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b66db467d95cf0783b86abecfede3e3adbe1947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b66db467d95cf0783b86abecfede3e3adbe1947"}]},"branch":"refs/heads/master"},"13daef25727b85e0be693ffcae0bdb5fff26fdce":{"kind":"REWORK","_number":10,"created":"2020-11-17 13:39:24.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/10"}}},"commit":{"parents":[{"commit":"bb1ad104d21ca34d4bbc2cfbc248614de381773a","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb1ad104d21ca34d4bbc2cfbc248614de381773a"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-17 13:39:06.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13daef25727b85e0be693ffcae0bdb5fff26fdce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13daef25727b85e0be693ffcae0bdb5fff26fdce"}]},"branch":"refs/heads/master"},"20ec31a3b97f697f32ffce042d5c2f7e95aa7536":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-11-19 08:47:32.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/11"}}},"commit":{"parents":[{"commit":"bb1ad104d21ca34d4bbc2cfbc248614de381773a","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb1ad104d21ca34d4bbc2cfbc248614de381773a"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-19 08:47:27.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20ec31a3b97f697f32ffce042d5c2f7e95aa7536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20ec31a3b97f697f32ffce042d5c2f7e95aa7536"}]},"branch":"refs/heads/master"},"306b4464c235bd2b47971f29445c089100261fe6":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-11-25 10:07:50.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/12"}}},"commit":{"parents":[{"commit":"9d5de15642719e04d7f7a91eb9913a5e614ca1f2","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d5de15642719e04d7f7a91eb9913a5e614ca1f2"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-25 10:07:50.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/306b4464c235bd2b47971f29445c089100261fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/306b4464c235bd2b47971f29445c089100261fe6"}]},"branch":"refs/heads/master","description":"Rebase"},"89dc7106f95cdfe43c525c7d0e0139d156e1bfae":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-11-26 12:21:53.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/13"}}},"commit":{"parents":[{"commit":"1a154839c2dfb5a9c89963b27a96f0adef1038ee","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a154839c2dfb5a9c89963b27a96f0adef1038ee"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-11-26 12:21:53.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89dc7106f95cdfe43c525c7d0e0139d156e1bfae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89dc7106f95cdfe43c525c7d0e0139d156e1bfae"}]},"branch":"refs/heads/master","description":"Rebase"},"c3a57de21b93889cbcb19723ec274f0271909a4b":{"kind":"REWORK","_number":14,"created":"2020-12-03 07:20:09.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/14"}}},"commit":{"parents":[{"commit":"1a154839c2dfb5a9c89963b27a96f0adef1038ee","subject":"Add amphora_id in store params for failover_amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a154839c2dfb5a9c89963b27a96f0adef1038ee"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-03 07:19:59.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3a57de21b93889cbcb19723ec274f0271909a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3a57de21b93889cbcb19723ec274f0271909a4b"}]},"branch":"refs/heads/master"},"3362ee69d36413f769ed952442eedd075f3ac98e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-12-03 07:20:20.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/15"}}},"commit":{"parents":[{"commit":"9d38db9065b27bfa2ef91f955da32c56c4b466a0","subject":"Merge \"Add amphora_id in store params for failover_amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d38db9065b27bfa2ef91f955da32c56c4b466a0"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-03 07:20:20.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3362ee69d36413f769ed952442eedd075f3ac98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3362ee69d36413f769ed952442eedd075f3ac98e"}]},"branch":"refs/heads/master","description":"Rebase"},"0d39910eaa065202d5014ce717f86a064e9a0ae3":{"kind":"REWORK","_number":16,"created":"2020-12-03 12:01:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/16"}}},"commit":{"parents":[{"commit":"9d38db9065b27bfa2ef91f955da32c56c4b466a0","subject":"Merge \"Add amphora_id in store params for failover_amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d38db9065b27bfa2ef91f955da32c56c4b466a0"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-03 12:01:24.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d39910eaa065202d5014ce717f86a064e9a0ae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d39910eaa065202d5014ce717f86a064e9a0ae3"}]},"branch":"refs/heads/master"},"94afffe48ff01f54be32af5a2fd76485579915ac":{"kind":"REWORK","_number":17,"created":"2020-12-03 12:03:19.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/17"}}},"commit":{"parents":[{"commit":"9d38db9065b27bfa2ef91f955da32c56c4b466a0","subject":"Merge \"Add amphora_id in store params for failover_amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d38db9065b27bfa2ef91f955da32c56c4b466a0"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-03 12:03:07.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nDepends-On: https://review.opendev.org/#/c/763319/\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94afffe48ff01f54be32af5a2fd76485579915ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94afffe48ff01f54be32af5a2fd76485579915ac"}]},"branch":"refs/heads/master"},"81264e5d9d563646a6af06f1d21acf3172a730ca":{"kind":"REWORK","_number":18,"created":"2020-12-04 10:26:46.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/18"}}},"commit":{"parents":[{"commit":"9d38db9065b27bfa2ef91f955da32c56c4b466a0","subject":"Merge \"Add amphora_id in store params for failover_amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d38db9065b27bfa2ef91f955da32c56c4b466a0"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-12-04 10:26:24.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81264e5d9d563646a6af06f1d21acf3172a730ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81264e5d9d563646a6af06f1d21acf3172a730ca"}]},"branch":"refs/heads/master"},"dfee00815cee1e26ad710e95f1c62dad0c9a519d":{"kind":"REWORK","_number":19,"created":"2020-12-04 14:33:43.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/19"}}},"commit":{"parents":[{"commit":"9d38db9065b27bfa2ef91f955da32c56c4b466a0","subject":"Merge \"Add amphora_id in store params for failover_amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d38db9065b27bfa2ef91f955da32c56c4b466a0"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-04 14:33:27.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfee00815cee1e26ad710e95f1c62dad0c9a519d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfee00815cee1e26ad710e95f1c62dad0c9a519d"}]},"branch":"refs/heads/master"},"be31e4c535e5c6439f4bbd344da8faff37c76c1f":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2021-01-11 12:50:15.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/20"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-01-11 12:50:15.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be31e4c535e5c6439f4bbd344da8faff37c76c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be31e4c535e5c6439f4bbd344da8faff37c76c1f"}]},"branch":"refs/heads/master","description":"Rebase"},"430b318eaeee6baa4207b83dc31d281ae277ecf8":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-02-20 07:38:42.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/21"}}},"commit":{"parents":[{"commit":"ce2e0d74258bec3a5ed4983462acee234c67864b","subject":"Add test coverage for SCTP health checker script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2e0d74258bec3a5ed4983462acee234c67864b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-02-20 07:38:42.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/430b318eaeee6baa4207b83dc31d281ae277ecf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/430b318eaeee6baa4207b83dc31d281ae277ecf8"}]},"branch":"refs/heads/master","description":"Rebase"},"8c4c000e826276a20dc06043c39f0246b686f26c":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-03-10 08:06:42.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/22"}}},"commit":{"parents":[{"commit":"41ff3ee3f1a9bbd1f82697c996116e40209825fe","subject":"Merge \"Add support for scoped tokens and default roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41ff3ee3f1a9bbd1f82697c996116e40209825fe"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-03-10 08:06:42.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c4c000e826276a20dc06043c39f0246b686f26c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c4c000e826276a20dc06043c39f0246b686f26c"}]},"branch":"refs/heads/master","description":"Rebase"},"4042634395748d346a5cdb5bb00dae1b8de0dbb2":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2021-03-24 09:40:35.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/23"}}},"commit":{"parents":[{"commit":"0513319f3d6e0d8ddbb6b386cbfe01829a68aa4e","subject":"Fix LB failover for amphorav2: set security group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0513319f3d6e0d8ddbb6b386cbfe01829a68aa4e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-03-24 09:40:35.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4042634395748d346a5cdb5bb00dae1b8de0dbb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4042634395748d346a5cdb5bb00dae1b8de0dbb2"}]},"branch":"refs/heads/master","description":"Rebase"},"8168cd7730bae860bdb3f021370740451618357f":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2021-04-20 15:33:14.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/32/740432/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/24"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-04-20 15:33:14.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8168cd7730bae860bdb3f021370740451618357f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8168cd7730bae860bdb3f021370740451618357f"}]},"branch":"refs/heads/master","description":"Rebase"},"1685a02fe2b08a1f0c295adbbecee08b1a189236":{"kind":"REWORK","_number":25,"created":"2021-04-21 09:08:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/740432/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/25"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-21 07:39:32.000000000","tz":120},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1685a02fe2b08a1f0c295adbbecee08b1a189236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1685a02fe2b08a1f0c295adbbecee08b1a189236"}]},"branch":"refs/heads/master"},"284003e84e4b9f5ebcf9851c9b8f17789b7f4391":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2021-04-23 12:56:31.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/26"}}},"commit":{"parents":[{"commit":"de352cac203b8e545adf1318ce53b99dc0bd404b","subject":"Edits for \"Operator Maintenance Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de352cac203b8e545adf1318ce53b99dc0bd404b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-04-23 12:56:31.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/284003e84e4b9f5ebcf9851c9b8f17789b7f4391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/284003e84e4b9f5ebcf9851c9b8f17789b7f4391"}]},"branch":"refs/heads/master","description":"Rebase"},"2545d9b3ef3447d7e95abf089c710969117e741a":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2021-04-29 19:54:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/740432/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/27"}}},"commit":{"parents":[{"commit":"c55e68a3bef94b481e23c2530b79826092a88db9","subject":"Merge \"Disable spare pool jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55e68a3bef94b481e23c2530b79826092a88db9"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-29 19:54:29.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2545d9b3ef3447d7e95abf089c710969117e741a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2545d9b3ef3447d7e95abf089c710969117e741a"}]},"branch":"refs/heads/master","description":"Rebase"},"12af350fd804a059a2bcae78bed8e4d755957b10":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2021-04-30 12:55:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/740432/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/28"}}},"commit":{"parents":[{"commit":"39e7ef872c8e83a39860cfa8fd3b72e6db84134d","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39e7ef872c8e83a39860cfa8fd3b72e6db84134d"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 12:55:02.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12af350fd804a059a2bcae78bed8e4d755957b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12af350fd804a059a2bcae78bed8e4d755957b10"}]},"branch":"refs/heads/master","description":"Rebase"},"a2550bcc2fe260ef7354f0a61ef230d92f14dd45":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2021-05-03 09:52:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/740432/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/29"}}},"commit":{"parents":[{"commit":"7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-03 09:52:10.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2550bcc2fe260ef7354f0a61ef230d92f14dd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2550bcc2fe260ef7354f0a61ef230d92f14dd45"}]},"branch":"refs/heads/master","description":"Rebase"},"a8a30fb5f25f0a1352d05a6fbdb33796b887d232":{"kind":"REWORK","_number":30,"created":"2021-05-06 09:07:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/740432/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/30"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-06 09:05:28.000000000","tz":120},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8a30fb5f25f0a1352d05a6fbdb33796b887d232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8a30fb5f25f0a1352d05a6fbdb33796b887d232"}]},"branch":"refs/heads/master"},"855e283143200e41f6fc79b15d18d18f86b28177":{"kind":"REWORK","_number":31,"created":"2021-05-06 11:18:04.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/31"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-05-06 11:17:39.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/855e283143200e41f6fc79b15d18d18f86b28177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/855e283143200e41f6fc79b15d18d18f86b28177"}]},"branch":"refs/heads/master"},"ba86f3d91e43ffc76709f8966503d0f7b85934ac":{"kind":"REWORK","_number":32,"created":"2021-05-11 08:47:28.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/32"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"akamyshnikova","email":"akamyshnikova@mirantis.com","date":"2021-05-11 08:44:33.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba86f3d91e43ffc76709f8966503d0f7b85934ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba86f3d91e43ffc76709f8966503d0f7b85934ac"}]},"branch":"refs/heads/master"},"426c9969551966856442a4f72d87b23b79cfa8b7":{"kind":"REWORK","_number":33,"created":"2021-05-13 09:56:34.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/33"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"akamyshnikova","email":"akamyshnikova@mirantis.com","date":"2021-05-13 09:56:21.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/426c9969551966856442a4f72d87b23b79cfa8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/426c9969551966856442a4f72d87b23b79cfa8b7"}]},"branch":"refs/heads/master"},"60a70bbbf88c3681a346d829b1cb94792ae8f717":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2021-05-31 07:51:51.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/34"}}},"commit":{"parents":[{"commit":"0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b","subject":"Drop Bionic support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-05-31 07:51:51.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60a70bbbf88c3681a346d829b1cb94792ae8f717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60a70bbbf88c3681a346d829b1cb94792ae8f717"}]},"branch":"refs/heads/master","description":"Rebase"},"c12502765ae7e9d5392f5afd18ce8851e2cca22f":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2021-06-08 11:48:16.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/35"}}},"commit":{"parents":[{"commit":"e1adb335c788dbd902bfd8a00f748da4abb8b06b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1adb335c788dbd902bfd8a00f748da4abb8b06b"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-06-08 11:48:16.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c12502765ae7e9d5392f5afd18ce8851e2cca22f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c12502765ae7e9d5392f5afd18ce8851e2cca22f"}]},"branch":"refs/heads/master","description":"Rebase"},"ac8900ddce58fba3a0201961973451047e660e5c":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2021-06-22 07:17:50.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/36"}}},"commit":{"parents":[{"commit":"1d77e93f2889627cb87f1aed3259b83e91e83239","subject":"Merge \"Fix new errors with SQLAlchemy 1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d77e93f2889627cb87f1aed3259b83e91e83239"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-06-22 07:17:50.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac8900ddce58fba3a0201961973451047e660e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac8900ddce58fba3a0201961973451047e660e5c"}]},"branch":"refs/heads/master","description":"Rebase"},"c8f56f2e7ef96a099225df712090f45ffddd3024":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2021-07-08 05:37:38.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/37"}}},"commit":{"parents":[{"commit":"e647f6d71a8479b2ce5fd8c2efb27b97c1f54505","subject":"Merge \"Edits for \"Basic Load Balancing Cookbook\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e647f6d71a8479b2ce5fd8c2efb27b97c1f54505"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-07-08 05:37:38.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8f56f2e7ef96a099225df712090f45ffddd3024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8f56f2e7ef96a099225df712090f45ffddd3024"}]},"branch":"refs/heads/master","description":"Rebase"},"438e9478741c416b3c87dd447d1aea2bcbbc2e2b":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2021-07-19 07:57:09.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/38"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-07-19 07:57:09.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/438e9478741c416b3c87dd447d1aea2bcbbc2e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/438e9478741c416b3c87dd447d1aea2bcbbc2e2b"}]},"branch":"refs/heads/master","description":"Rebase"},"8451f3a9c6b4413ac22058ef68b65d78db409572":{"kind":"REWORK","_number":39,"created":"2021-08-19 14:19:24.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/32/740432/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/39"}}},"commit":{"parents":[{"commit":"b89c929c12fb262f59ba320a37f2a5bf4109df98","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b89c929c12fb262f59ba320a37f2a5bf4109df98"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"akamyshnikova","email":"akamyshnikova@mirantis.com","date":"2021-08-19 14:19:12.000000000","tz":240},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8451f3a9c6b4413ac22058ef68b65d78db409572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8451f3a9c6b4413ac22058ef68b65d78db409572"}]},"branch":"refs/heads/master"},"1aef468ec819802bee3cf43f4eeb1fa52c5b01c8":{"kind":"REWORK","_number":40,"created":"2021-08-19 20:48:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/740432/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/740432/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/740432/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/740432/40"}}},"commit":{"parents":[{"commit":"b89c929c12fb262f59ba320a37f2a5bf4109df98","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b89c929c12fb262f59ba320a37f2a5bf4109df98"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-07-10 09:40:46.000000000","tz":240},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-19 20:48:39.000000000","tz":0},"subject":"Alias change amphorav2 -\u003e amphora","message":"Alias change amphorav2 -\u003e amphora\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: Ica759236f395c7bec28c794d2f1df5e0b690939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aef468ec819802bee3cf43f4eeb1fa52c5b01c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aef468ec819802bee3cf43f4eeb1fa52c5b01c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
