)]}'
{"id":"openstack%2Foctavia~740860","triplet_id":"openstack%2Foctavia~master~I7236326948ce152242c1aea7a6ffd5f4848dc52f","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7236326948ce152242c1aea7a6ffd5f4848dc52f","subject":"migrate testing to ubuntu focal","status":"ABANDONED","created":"2020-07-14 03:39:01.000000000","updated":"2020-07-28 05:42:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8eba57e7f5db5e64cf877ea84c606f96b072824","_number":740860,"virtual_id_number":740860,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-14 05:39:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":-1,"date":"2020-07-14 06:40:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-15 11:32:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2020-07-20 16:16:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 05:39:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-14 06:40:54.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-15 11:32:32.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-07-20 16:16:10.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"5568a25d271e00b5dc7a92ed29ebfc70bd452eaf","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-07-14 03:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"778103f9538f5acf0b1bf8680cde7df02e1f0f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 05:39:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/311f308ce46946a781d10b29d5dda618 : SUCCESS in 5m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2fb4ea4eb58d4159a2b0a440cba813cc : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6fe1b9fd06e4df2a22c41a838a1da4e : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edf4b86258bb407dbaf41f0eb78e06df : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece7aef6c74442d68eeef5265d629921 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf87047b92344f7f93b222a93ebdf66c : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84a5db9d97554e2f952b26d410ba08de : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7678fa4b1d412dbc12bbbeecae0bdb : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7514d905f13543b2a0f385158159a1a8 : SUCCESS in 4m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/099abb8caf1e4e478ab2213f3ad8f1f2 : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/35612f6840344ef7b79bc207afddb12d : SUCCESS in 10m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db231efdfa5747cb83dfcae4b64ea456 : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54791f470211466cb6299c9a841964f3 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33bedb1ebc804b07b4b79c6d19811049 : SUCCESS in 1h 53m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9747d6fc40a44f858214e04cda7ecdde : SUCCESS in 55m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a62681ec8e2442a2b581d0b4167e4c0b : SUCCESS in 41m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4920c6750e3048ada0889d1ddd77e7d1 : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/17ab68306ec749d187f142a048b6dca9 : SUCCESS in 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/99c4c35ac5fe4b519a107d92ddfed23c : SUCCESS in 1h 51m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b8f2e7b517d4443582c6921535c891c5 : SUCCESS in 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/38349d1ca51e4ee28ecd090757a87067 : SUCCESS in 50m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96f14c52f6b56a991f405ba064b7bb8e216e86e1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-14 06:40:54.000000000","message":"Patch Set 1: Code-Review-1\n\nyour commit message is not describing the change, please update.\n\nAlso, update directly to hacking 3.2.0.","accounts_in_message":[],"_revision_number":1},{"id":"91f30c00d433f7b3d3942ac7d72eed8c338156ae","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-07-15 11:32:32.000000000","message":"Patch Set 1: Code-Review-1\n\nyour commit message does not match the contents. please update your commit message,and hacking 3.2.0 is not out yet, could you provide more infomation?","accounts_in_message":[],"_revision_number":1},{"id":"591e0965eb77d715d769b2cb45b70b0aa34e5437","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-20 16:16:10.000000000","message":"Patch Set 1: Backport-Candidate-1\n\nthis is not migrating or testing anything on Focal, please follow this ML thread to help on this work. - http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015783.html","accounts_in_message":[],"_revision_number":1},{"id":"a8eba57e7f5db5e64cf877ea84c606f96b072824","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-07-28 05:42:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d14c62ed935a6a47bde8f893529fbc06d7323c05","revisions":{"d14c62ed935a6a47bde8f893529fbc06d7323c05":{"kind":"REWORK","_number":1,"created":"2020-07-14 03:39:01.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/60/740860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/740860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/740860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/740860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/740860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/740860/1"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-07-14 03:37:59.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-07-14 03:38:55.000000000","tz":480},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nChange-Id: I7236326948ce152242c1aea7a6ffd5f4848dc52f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14c62ed935a6a47bde8f893529fbc06d7323c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14c62ed935a6a47bde8f893529fbc06d7323c05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
