)]}'
{"id":"openstack%2Foctavia~740892","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I64172d6995959cf377364584ad9a2395f9ec0605","project":"openstack/octavia","branch":"stable/train","topic":"failover-refactor","hashtags":[],"change_id":"I64172d6995959cf377364584ad9a2395f9ec0605","subject":"Fix missing params in amphora base and noop driver","status":"MERGED","created":"2020-07-14 06:27:23.000000000","updated":"2020-07-19 19:50:12.000000000","submitted":"2020-07-19 19:47:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740892-1595188071412-56289871","meta_rev_id":"6434953be912757d6f01ba76ad49a846b80c7926","_number":740892,"virtual_id_number":740892,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-07-19 16:16:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-19 19:47:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-16 12:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-16 23:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-16 23:49:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 12:24:23.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-16 23:49:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-19 19:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45d9fefac8d6677648d1f8fad01d045df0ec6ac2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-14 06:27:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"711558cb40c0fa2cb764e0acabb774e97be9f1d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:11:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/763afe8365bd4d57aa51bfe59dd54baf : SUCCESS in 11m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f020c2603744ec0a4d9b665298e04fa : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b57856202649619586ed2b358d659f : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10e9efeae2c24de59615f2629cb6d18d : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f0e8cfccaa8423c8dde9fe421474d54 : SUCCESS in 8m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9b25f3ad5954ff5ba1fdea6c1194271 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7274436cf714ddcbceb1e9711e8d025 : SUCCESS in 14m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d8a744741e3a4380a7229b9e722a688f : SUCCESS in 10m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0dcaac694d7b4d629f3e50e3b53d5fec : SUCCESS in 13m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/746f301a90c14fea81da737e23e8fcab : SUCCESS in 11m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa8eb85e9e464f87bc70e86c3364b082 : SUCCESS in 9m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66f84613aa604531a2ed77fa9f4755f4 : SUCCESS in 47m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/814c3559df864ae5bcaba9b3e34d7446 : SUCCESS in 52m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bc7c5f484e946fca1e551cb4b1bc214 : SUCCESS in 1h 57m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/857ecd72bba54f58b448ad74654653a3 : SUCCESS in 1h 46m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a6daddd2b7094886b738a73010a7d5f0 : FAILURE in 18m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/788b66e780c74612bc045f369afe0e86 : FAILURE in 1h 00m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f87c9849cd24142b49f8670e39bffac : FAILURE in 53m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10a4566c9cb544a6a6d8f9f2317d74b7 : SUCCESS in 1h 14m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e02820bcc8944d618e45ac0fbb111702 : SUCCESS in 1h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/afcb9196cffd49b28aa3369a3a484f7c : SUCCESS in 1h 47m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e66c68bf56cf4639aa7d2a66b47b9fe8 : SUCCESS in 57m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a2487329ea9043249f2e5eb229afc22d : SUCCESS in 1h 12m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7d38d6f14e5d45cfaafed6829f9f6597 : FAILURE in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6e8fa8f7ceeea2a019191b0f51a4ea97b6dd3e1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 12:24:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"95cbd757d8ad0232273bde9f8e209f366676ddd0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 23:49:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"04f4a57dbec211d79845196ae6b57fff0dbf71f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:40:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50dc85dd4f97b123e2bfca739d65bdce14c484d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 18:34:28.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"50173378ea165d0cb04897e3f5165ee439d05440","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-18 03:26:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5497153b7a04627ff3011def6aaec30eeeaef0e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 05:00:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab4beb942c984ce19696b6d461fa6250 : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee5646c9b94a4ecc951b0d2297cef767 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5df730b54b847a4a3fd63e0786d4742 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66ace8e9121542029911f388178e15ef : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca016ba04829463991935398c4f7a9b2 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec801164e3ca42f887f6ec93c920f6d7 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df3483a54224beea7ac81a5e1733dde : SUCCESS in 9m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b15259a4727742c2a63673fdc893a8dd : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee715f83cab34e178d6615dc7610d069 : SUCCESS in 6m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d61f9fdd4a0b4fac8655bb697c69e10f : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e26637e772c740ae9fc81399c4523696 : SUCCESS in 6m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9617b5f2304b482fa6981ceb7fd145c6 : SUCCESS in 42m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/baf9354608dc4f76865a964fb2d9d78d : SUCCESS in 43m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/148dfe6cf5f1414ea13bafa9b25730f9 : SUCCESS in 54m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6bc1a66cd4224f4cbcf5eeb38f5328fe : SUCCESS in 53m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/65f086b162834af8968bf5e57ea5afb9 : FAILURE in 8m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84701e5037db4800821b56966ef28ea8 : FAILURE in 29m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a56d34ef7d8747cdb3a2276d6a41dca0 : FAILURE in 44m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58a453607ec34e26a88f2dfd15ad1a67 : SUCCESS in 1h 27m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e58279e914544b7b410dd7a1e17a948 : SUCCESS in 52m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a549a1e3f6684514ad963b3c228de8af : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed8e5d2096494040a2c12eaa75865918 : SUCCESS in 51m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/14a7af1b09a647d08650e4b30e49df40 : SUCCESS in 31m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6cf89a9d259148e29d46788800ff92e1 : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e70b17b74a2fa64e24598c0000fd3cb543befc83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 05:01:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74a82c8605a5b2e3f21db8be29c7d590659b4fe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 06:36:06.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"0edfb0b8271e8ae71f8bba460e8eaf354295c50d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-18 22:16:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc78c13ce540e932c6d4122cb82b7c5f12ec4252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 23:56:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf6b9c48f9dd41f4a1cd54ac2303cc74 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68178a71211f4edf961dea8a524de498 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b45e87c89042198f12c6e1b6af5d5a : SUCCESS in 8m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed100bcd603d44bda98ea8df90a1545c : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90e2d6fbf040481cb9fa04f1376b8be9 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2e90fab6b56440198a16c11e90c32f1 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757fffa7f6a14e7aaa2030489799ba43 : SUCCESS in 9m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f115667b0bc34ceb9dbdd9284066849e : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4226523be24a4a2bbcec11aa7e7d0422 : SUCCESS in 7m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd45a7251d064177a1d4937c607c61f6 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d943fe914d864b3fa045cf3663dc40eb : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9eb3190d0e648b1a79bedd9772b90e8 : SUCCESS in 41m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4e7bcfc5950344b9b2d3031208847f2a : SUCCESS in 45m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de22a1c20af54212983b90427e974212 : FAILURE in 38m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3dc59100a5d0464da1dc7d210ac05f29 : SUCCESS in 46m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5af25bd1be044cca91b3f89a1b8197c1 : FAILURE in 8m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a5dd22882d54997a45be4a66abdac87 : FAILURE in 21m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/15df68fd1b434805828bb08dffb832e3 : FAILURE in 55m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7186784546724d2b8fd9a285a669d05e : SUCCESS in 1h 34m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39561d63083b4e1987075be5c15f22ea : SUCCESS in 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b207ff3f6de8458093bd8b62db1ccb02 : SUCCESS in 39m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a04ad03932d4d8490216690770f156c : SUCCESS in 30m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/12c49edceed04853ad0c626c227f802a : SUCCESS in 32m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/005932c584dc44a5b4b5bd05f48d78d2 : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9ff03ea34db9bd949d6ad9c3ac9df8a0b97200b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-19 16:16:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"348b1a4b3c6bcf90a60a428794921169c9688ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 17:56:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27490fcbffb2414ba4797b02b4465783 : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a97ac4d316854e11825b348c566d3fb2 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302f6748aabb44dfa5b130c701c49ed1 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1e373d6312d480690b4a0b00f6f4ec8 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef378f13d0cd4392b5a8d9d461e0b20e : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a16cc2b5b5a46ef8490274a8ed77d1e : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ead967b9f134ac1b570d278ff226242 : SUCCESS in 9m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5435220c9e4843bc97b500d00f764ecd : SUCCESS in 4m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3079ab5cc79543028e5e5897a1aa543f : SUCCESS in 7m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/160be3068d2c4c2586c424fc96647145 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e23a8813c68c475388f0f33fe4502f0a : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e606d89de294694bcb702ca1503977f : SUCCESS in 37m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/eb78cf0081c04f10bdaeb7ab4ff62f73 : SUCCESS in 46m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7674a7965a14a9c97ef9f8df38a64eb : SUCCESS in 54m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f006a180de014364b2aaedc55b72b612 : SUCCESS in 56m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/24d4fbbad2fb4467bfe4f56c1c7d4c8c : FAILURE in 8m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1830a84977b84db089cfb307999f4934 : FAILURE in 21m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e0e42d0e1c6a48cabcd159af00c7d701 : FAILURE in 40m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/15ed25c2066b45dea8f00ff34cb8dc6f : SUCCESS in 1h 32m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0dcb042010e4669b105423c016d47a2 : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b67eb4b61c484e408e80cab347533c79 : SUCCESS in 59m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db6a6b2cdea14d37bba8235501678b71 : SUCCESS in 33m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/766c4273e6f040d891602c0337e05f51 : SUCCESS in 39m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e9e1cb026b824476a6fa911a73ba76d6 : FAILURE in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10fae761dab8bc718ea9d358ac30d65f5dfd8a9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 17:56:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c59456100e9621bc07ede8100472e8b8680c2d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 19:47:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"74f841ce8911d9e763499ad6b9bc7f904f70aa16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 19:47:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66c85cf15a7d4b72819440bc45b26c15 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3457ad40895245a8ad4614d4c36a3acd : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3136fc5a41c1485b82c5d90465ef572e : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a41388581694b8d806be7ba4f9c05ec : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9d4df56648a4b86bed15aa15be17e5d : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed678fccad1f44a081a8442057515152 : SUCCESS in 9m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69625d4246b14314be0f46f859c2df7d : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e22159e1207c4ce5a29ac091947d92fc : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ffcb79267d64e7eb3477991b713d8f0 : SUCCESS in 45m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d0edff96f0e24190b03c9f4b36070eb6 : SUCCESS in 55m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/794d98c1d5b24664bac8ea14470a90ad : SUCCESS in 57m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a5042c26ec1f485292e85477e785c69d : SUCCESS in 41m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/097213c9d2d94b2fa2add042cdda04db : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a98e82fe967f0424b32d2d85f0866a5eb86ffa86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 19:50:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a71a8a2d72426b8c31b92c445db8fc : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc47caa1a27e0c7876637715d62f2520d27f7ec7","revisions":{"dc47caa1a27e0c7876637715d62f2520d27f7ec7":{"kind":"REWORK","_number":1,"created":"2020-07-14 06:27:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/92/740892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/740892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/740892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/740892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/740892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/740892/1"}}},"commit":{"parents":[{"commit":"d285f517d2842b0cc87de997f941214025c7e979","subject":"Add missing reload method in amphora noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d285f517d2842b0cc87de997f941214025c7e979"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-24 09:55:11.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-14 05:53:34.000000000","tz":120},"subject":"Fix missing params in amphora base and noop driver","message":"Fix missing params in amphora base and noop driver\n\nRunning amphora failover against the amphora noop driver was raising a\nTypeError (reload() takes from 2 to 3 positional arguments but 4 were\ngiven).\n\nChange-Id: I64172d6995959cf377364584ad9a2395f9ec0605\n(cherry picked from commit 41c628a084002017d2003926cf0e25ba3ffeee0c)\n(cherry picked from commit 176791e675141b37ddf139a9d9f2ae6bb814fa20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc47caa1a27e0c7876637715d62f2520d27f7ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc47caa1a27e0c7876637715d62f2520d27f7ec7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
