)]}'
{"id":"openstack%2Foctavia~741147","triplet_id":"openstack%2Foctavia~master~Ia20caacd31ae5a411981a33d9a7f2a6330902760","project":"openstack/octavia","branch":"master","topic":"centos8","hashtags":[],"change_id":"Ia20caacd31ae5a411981a33d9a7f2a6330902760","subject":"Add non-voting octavia-v2-dsvm-scenario-centos-8","status":"MERGED","created":"2020-07-15 10:31:41.000000000","updated":"2020-07-16 13:09:46.000000000","submitted":"2020-07-16 13:05:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741147-1594904755888-f4e3b21d","meta_rev_id":"753b0d5705836fd6bcf694e81dba43e330165bc3","_number":741147,"virtual_id_number":741147,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 13:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-16 11:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-07-15 13:24:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-07-15 17:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-07-16 11:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 13:24:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-15 17:05:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-16 11:12:37.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-16 13:05:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f32f2e3df2bc5ee04d93f42a566211877837ea2f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-15 10:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13428e8c75a4566a831de8f738bfde1b81552d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 12:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ab0a63382084370a8db8e91cbeb0ec4 : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14e6757cc33c4accb3153b691e6a99df : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10fdc87229d448899f36f15e778f4adf : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36670de5a6104f0094fa6be11e8556a0 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a2efe7099b4984846bc2c6a18d94d4 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4732261c534a34a6e2716657e984b1 : SUCCESS in 18m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ff61416a38574775b3f443be36a938db : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8c610c585e434d428ef0d10a1f4ecc3e : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5eab9961eed74a54b72b1e31b45369e6 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a38641b3c17b4996af4e504520314fc3 : SUCCESS in 48m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3cf9ab800fe48f3adf77744c77cfb21 : SUCCESS in 42m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/272f08a851bd4697ba4f9fc2e1fcc671 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/04671b29961b45c499c055fa9971a73c : SUCCESS in 1h 02m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc0116aaa63e477e80ea55b35c0ee4cf : SUCCESS in 1h 00m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14127cfd7e864930b4b31032fd56b34f : SUCCESS in 1h 34m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/451804524dea4cc0859a4bb2878655ca : SUCCESS in 1h 22m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/240a0ed7bbdc45c993438cc9aa1aa1a5 : SUCCESS in 1h 16m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e3e18d875cf49d99a53c09f12aa2959 : SUCCESS in 1h 19m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b426044ebb704b58bb1fd801c2f7f3d3 : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/817e671c85ea449faa67bd183bf5c3ad : SUCCESS in 1h 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b85639565223ffe116128b03c27a716b895f7bb3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-15 13:24:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef185d2182c2e8fb30a94970a56b9637f79c0495","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-15 17:05:23.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"83a7f725def65e6b4a223250b20ea71cbea4d4e6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 11:12:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05e51512cafb78e07c7fa07f7c8edbcd8657ba9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 11:12:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fda6cb19e79dd2cca36b0ecc3ef513e26673d974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 13:05:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c38fb10de01283ec54887b03f3061c267b997ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 13:05:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba35b1c4c8a24d4188722616a811c641 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff14dbc02454bb3a087116881c6969f : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb8bdd84c70942aca8f2877810b5de0a : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9ed4b90ea684668acad3352b0dddf95 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5739c4f34c14ce1b1754ce5d89d91a0 : SUCCESS in 12m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29985dbbaad943aa896b1ed10e2c0388 : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ca42fc6ea9b45b8892fb2b5268f19ac : SUCCESS in 48m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec0cd4b0f6cb46d6ac1a901fea9593ea : SUCCESS in 53m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/382c15ab880447f08b2bf016ae7a1b2a : SUCCESS in 32m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ffaa0cdfc3634a37b88150e61036672e : SUCCESS in 30m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd81ef13cb0c49fab003ed9edcbc30e0 : SUCCESS in 31m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c3235d72b954f56937c1fc04e14ccb5 : SUCCESS in 1h 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7e0bba5826ac54614fabbc14567e2b59dd1cf933","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 13:09:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8131f425d064c20b8cc4d919bcabfad : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1072263b73810133ff2c0e25e4b0a085e8f60f02","revisions":{"1072263b73810133ff2c0e25e4b0a085e8f60f02":{"kind":"REWORK","_number":1,"created":"2020-07-15 10:31:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/47/741147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/741147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/741147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/741147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/741147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/741147/1"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-15 10:22:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-15 10:29:53.000000000","tz":120},"subject":"Add non-voting octavia-v2-dsvm-scenario-centos-8","message":"Add non-voting octavia-v2-dsvm-scenario-centos-8\n\nRecent change [1] improved the reliability and lowered the job duration\nsignificatly (as low as 41m39s per a recent run) [2]. This enables us to\nmore comfortably (less wait time) and confidently (less timeouts) run\nscenario tests on CentOS 8.\n\nShould this job prove to have a record of being stable for a reasonable\namount of time/builds, we can follow up with a patch to promote it to\nvoting.\n\n[1] https://review.opendev.org/#/c/738246/\n[2] https://zuul.openstack.org/builds?job_name\u003doctavia-v2-dsvm-scenario-centos-8\n\nChange-Id: Ia20caacd31ae5a411981a33d9a7f2a6330902760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1072263b73810133ff2c0e25e4b0a085e8f60f02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1072263b73810133ff2c0e25e4b0a085e8f60f02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
