)]}'
{"id":"openstack%2Foctavia~741619","triplet_id":"openstack%2Foctavia~stable%2Ftrain~If5b6411a0b7c75441a406234c2792ea68d35d0fe","project":"openstack/octavia","branch":"stable/train","topic":"story/2007421","hashtags":[],"change_id":"If5b6411a0b7c75441a406234c2792ea68d35d0fe","subject":"Fix listener update with SNI certificates","status":"MERGED","created":"2020-07-17 11:28:41.000000000","updated":"2020-07-17 20:14:59.000000000","submitted":"2020-07-17 20:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741619-1595016662275-b2062de1","meta_rev_id":"6d2c880472035b9f40b67b196e8f14c47614f2c4","_number":741619,"virtual_id_number":741619,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 20:11:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-17 16:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-07-17 16:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-07-17 16:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 16:41:11.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-17 16:42:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-17 20:11:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca79df123c656d8712f1ce164dcfbe165da237c8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-17 11:28:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b45a35eae21e4affe49c3557576ea4bbd56f12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 13:12:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c20f33ddf9384601b1a9c111cb3dd462 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/681e867cd2c940dbb2e92efc50910c3d : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc47e624720146cb91acac87b6a7df5c : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3212f2bcbccf4c9b8fca2d1d2cda7d7a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc622c2843434834a96450d86cd0f688 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/364e1d4f619a4eec8060bcc004153019 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bce3c14dd6d4f5691628104779b69b2 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cbb509cc61942a7992ce0f4b7eb7ab2 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e1089121184b4ff1acf3e655409663b7 : SUCCESS in 7m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d2149b7c45454fc898bcf45242c6c1ff : SUCCESS in 8m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/80305733f68f4db6adc5c77629711cf1 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fab993fc81584e6d914e6b5bac2cf628 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04ab6341c5ae40cab07a960c46c823a0 : SUCCESS in 47m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/80a98d7af093407890326a31ac3df89f : SUCCESS in 48m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c3f706943f745bd983843fef2f5bca9 : SUCCESS in 58m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/577820fd30974debaa47997f80a09cfe : SUCCESS in 1h 01m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3ade7b8493914d749c580b2da4d6b7e3 : FAILURE in 13m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2c4a53aa6004b3189715368d35372f7 : FAILURE in 47m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/48cc9e903f0e4fd581ea0c0fc8cf88fb : FAILURE in 1h 11m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34f27218c7b343b097f72230d05821a4 : SUCCESS in 1h 36m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2dd0ed2eaea946f1b958021bbc75e592 : SUCCESS in 37m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0355bfed9b7547f99db31272bc929317 : SUCCESS in 1h 03m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/244dcf15f44b44c98afc057d5d002266 : SUCCESS in 40m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9b46c464aba64bcda7418486e86b0599 : SUCCESS in 44m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2fa7d21d26474192817ace999ac4ea73 : FAILURE in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73ddbe85d4ac709959299f14a2aeb112790e0fb1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-17 16:41:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bda90908c84648a94df322b1f6ba1a0ddb05cd26","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-17 16:42:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"66464c1a6a822f1ebaa58a1f4ef1d6e6e031e84b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:42:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6c2feffd2715605b32ae7856f701ac464e8a29e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:11:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c24fd94639f312658a3f708c4228944f2dd79fae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:11:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76343dbdedd74f568caafecf39c52950 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9837633fa6d041ae98101354200a457f : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1705b8f610ba413db53bc04c1de9a798 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfbb15afbb9941228cd9bdac699851d2 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7542a8f8db434180bce4de17b4719ae7 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a30efde6bf5c42abbbb40bb30dc2c25c : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d640eef21f48480b91bd82d3862d881d : SUCCESS in 5m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/141272658968470fbdd545282b74078d : SUCCESS in 8m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/729f468a6d604d91a177e30d3e748974 : SUCCESS in 10m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aedb5b1dc0e44c2f9f9780875f7be01a : SUCCESS in 43m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2ab9443d945b40c09e18d15606a3b500 : SUCCESS in 49m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c83ae911be49bf983758753ebb680d : SUCCESS in 53m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1af03bd6998641bdab88e84cc2cfe444 : SUCCESS in 1h 09m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/442c03f831b5457b9928366c93157406 : SUCCESS in 1h 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"feb267dea9f803b7f46781ecd596a356646d9f0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:14:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e339c110eb4b7ab3ec54409f163e08 : SUCCESS in 3m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/035bfcb5f37c4deb8f4b068d67c7c99a : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9dd09450863a155490f9bbaf4b0dd1f23a00b02","revisions":{"b9dd09450863a155490f9bbaf4b0dd1f23a00b02":{"kind":"REWORK","_number":1,"created":"2020-07-17 11:28:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/741619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/741619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/741619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/741619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/741619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/741619/1"}}},"commit":{"parents":[{"commit":"2a18b6cc996e4c64d650c516071c57358681af3b","subject":"Fix neutron subnet lookup ignoring endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a18b6cc996e4c64d650c516071c57358681af3b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-12 20:30:31.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-17 11:28:28.000000000","tz":120},"subject":"Fix listener update with SNI certificates","message":"Fix listener update with SNI certificates\n\nSNI certificates were not being set in the database on listener update.\nA listener GET would not show the certificates in the sni_container_refs\nattribute. Also, the API was allowing set of SNI certs on non\nTERMINATED_HTTPS listeners.\n\nTask: 39042\nStory: 2007421\nStory: 2007430\n\nChange-Id: If5b6411a0b7c75441a406234c2792ea68d35d0fe\n(cherry picked from commit 9e070e6e6df6468a0b8f792cb089de0d9fc180b5)\n(cherry picked from commit 7c6fef82ac3afbd6e47d3b2ad6341fc1accb8c54)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9dd09450863a155490f9bbaf4b0dd1f23a00b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9dd09450863a155490f9bbaf4b0dd1f23a00b02"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
