)]}'
{"id":"openstack%2Foctavia~741967","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I670087163b265e7098af35063572d6aa9d068bb9","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I670087163b265e7098af35063572d6aa9d068bb9","subject":"Prioritize policy validation","status":"MERGED","created":"2020-07-20 15:25:06.000000000","updated":"2020-07-25 10:59:47.000000000","submitted":"2020-07-25 10:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741967-1595674693149-9adae2ba","meta_rev_id":"dc1aec51b67756271f2262386202cef23e307e6d","_number":741967,"virtual_id_number":741967,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 10:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-07-20 15:28:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-25 09:36:21.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-07-24 12:12:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-25 09:36:21.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 15:28:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-24 12:12:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-25 09:36:21.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-07-25 10:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f4a51031084b297cea6383543d4b03c5cbf9140","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-20 15:25:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"61922933cb2677f5068ebf93e3df208aa90d5adb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-20 15:28:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02260274d2cd83ada3913a3d3b43074ca310abfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:31:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52de5c7fa58b45e19dd42d22ba182d3f : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ad7bc77dd064915a6d60afcdfb3ca52 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1e1d74b64940dfadb754360c56e400 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d98d2345cbe444eb301e190f9af6155 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18477d8812a94fbabcab7bcb47fb92c2 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f07f9bb24440d2887461a72c9e9fe3 : SUCCESS in 16m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b81088b8d3c74c0d8810ccc5f780c638 : SUCCESS in 6m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/97b88bf72d2748709c20921db591a173 : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/184173b8f2cb464d8ab6773d82c512d0 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/832843cd5a064f1296f4b9c0c1c71bd6 : SUCCESS in 47m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99a2239ce40c4c559bd51c78f54eb3dc : SUCCESS in 53m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/538d97af117e4891b561a8e72306bea5 : SUCCESS in 34m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d9a2bb06113b428c8409465b37a457d7 : SUCCESS in 51m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2791cb5c6bc423d9289cab3e83dee58 : SUCCESS in 35m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3dac832869984dc5a9c6bba455ba93e1 : SUCCESS in 49m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/897590e0f9b84b1690afd8b93a93a6b2 : FAILURE in 39m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a056c031bcc14f91b2a8decee00c8c6c : FAILURE in 1h 01m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/adb95920241e47b28b677d511033855f : FAILURE in 1h 02m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7953d67dd45b5bfc3e37fb6f19f4d883a8e6719","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 12:12:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8fcaab9a28be0b06c69918436ef7bdff9f7aa04","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-07-25 09:36:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c52c422dcf91a1af15b730afc411cb30fd1951dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 09:36:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca9ffe5fde6acb28b77c9dc8e270dc0938b5eea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 10:58:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45cde444da074a38a654443fa1e4845a : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b740fd89e640068c0e140795bfa6b8 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/115b13a7534e4da4ae234ff0b7d6415d : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86602366638a45d4aba0092b08b617b5 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a020366b9bf546a98feb3219eae64b39 : SUCCESS in 12m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c6c905ffe914154a4415f1f27d1835d : SUCCESS in 10m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/079aa93dcb694793a3a6cd0807e8c706 : SUCCESS in 46m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01255a1bf31a4021b4f31df4a4379597 : SUCCESS in 1h 19m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2e8f6d999394758af48e82c8a9423b0 : SUCCESS in 1h 08m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ad343fe9269c483d833a3397db5d018d : SUCCESS in 1h 05m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea88759dbd58470bbac4d95486fcecd6 : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3cdc01d5f6fb00d1fa27a333cfa0a47616d9eec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 10:58:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dc1aec51b67756271f2262386202cef23e307e6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 10:59:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597a5b4499124b708de6b7c33d4491da : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"274393fdd4f157d7c6121b0e2265831f9ff4ca2e","revisions":{"274393fdd4f157d7c6121b0e2265831f9ff4ca2e":{"kind":"REWORK","_number":1,"created":"2020-07-20 15:25:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/67/741967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/741967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/741967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/741967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/741967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/741967/1"}}},"commit":{"parents":[{"commit":"35c971f49447c19719192761bb30ab9e735cb8af","subject":"Merge \"add the verify for the session\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35c971f49447c19719192761bb30ab9e735cb8af"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-19 21:17:29.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-20 15:25:06.000000000","tz":0},"subject":"Prioritize policy validation","message":"Prioritize policy validation\n\nThis patch makes sure that we validate RBAC compliance before\nother validation tasks.\n\nChange-Id: I670087163b265e7098af35063572d6aa9d068bb9\n(cherry picked from commit 8394633635af1045131b1ba5eeeafea8aa98416b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274393fdd4f157d7c6121b0e2265831f9ff4ca2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274393fdd4f157d7c6121b0e2265831f9ff4ca2e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
