)]}'
{"id":"openstack%2Foctavia~741975","triplet_id":"openstack%2Foctavia~stable%2Fstein~I670087163b265e7098af35063572d6aa9d068bb9","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I670087163b265e7098af35063572d6aa9d068bb9","subject":"Prioritize policy validation","status":"MERGED","created":"2020-07-20 15:52:07.000000000","updated":"2020-07-25 11:11:54.000000000","submitted":"2020-07-25 11:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741975-1595675442877-38a1cb2b","meta_rev_id":"1812e7dfde9ce231f29c8c5d88cf477494046c35","_number":741975,"virtual_id_number":741975,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 11:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-07-20 16:44:45.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-25 09:36:48.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-07-24 12:12:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-25 09:36:48.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 16:44:45.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-24 12:12:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-25 09:36:48.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-07-25 11:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c756059fff6ee6adcb6e4840b0d1ebf4152086bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-20 15:52:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"3f33e1df33e60d7d215660f7ab7b6e2a639db742","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-20 15:58:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d7172784ca83ed73ed043c9e5298270eadb6712f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-20 16:26:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b189ec6ed75e18eb6ce13313176c1288ea9fd9a3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-20 16:44:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b26141758bbc5cce7f57a929d3a89c5b36f72db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:34:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a411d4e3ac6a4948b495d360805b5bec : SUCCESS in 12m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f21fbbec3c44872be4f838e90dc2e56 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18dab9c7dd8e4170a4a8c4ea401593d1 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c95fb39d536454bb682479f8f4dc036 : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/658fcb2c21904dbf8ce75d0959d2d26c : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eb0510f9bd9408d9cd57a3218a32bf8 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1fb64e52fd4558bec63c1c13782a34 : SUCCESS in 7m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dac510e4faf74741b2299fd9a88a279f : SUCCESS in 1h 20m 56s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4af39d9fc1d948418bac6314c242e79f : SUCCESS in 1h 21m 03s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d55b288484af40bcb87fcfc66bc9e241 : SUCCESS in 45m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c030b95245a74bf4af925975ac0260f4 : SUCCESS in 1h 26m 56s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b90ee53bd75f444ba44955f0e7ee8204 : SUCCESS in 8m 46s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/de50401db6af42b3aabfbf702da02593 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2464c4aa9af14dd285082da499fd37d9 : SUCCESS in 7m 50s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa296eb4ea25495c9a4009b971293c05 : SUCCESS in 1h 26m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7269a19115c44501867376884e37e95d : SUCCESS in 40m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec441c0be73c48228f54b2d2841c70df : SUCCESS in 35m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/043ff0901ef34fc1891aa6039ffc363f : SUCCESS in 45m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be7a6d755639469bab81efe83344cd7e : SUCCESS in 56m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3014c9115f1d474a91e6a801dc742a0c : SUCCESS in 54m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/55c8fd0d0066477db6ac785aad1cf8cf : FAILURE in 12m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e80e1eacba1443291e2c6b233b1a929 : FAILURE in 31m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca9b5600926149688b72f4d694583987 : FAILURE in 46m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f3220bda1aa424db46383340caa07ee : FAILURE in 46m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f59c2f3ffae453587d62582e0525e54 : SUCCESS in 1h 19m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85f88614fe234dd0a02bfb3926d6d932 : SUCCESS in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a42311a403732c21b7a8067769c1820a2432b4ee","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 12:12:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"47e552759a4798d57c96bc17277fea040999801b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-07-25 09:36:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"354d52d2d51d5bbb7bcb02166db358872a4ed199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 09:37:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2229935f16eac73ee4259a9762ce34148bde8d95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:10:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1e14c4058254fd0b1278b0603478a71 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd251afe46e46b19053865ae5c52671 : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b8587b3c62e4bdcb8ab9d4e2f575ac3 : SUCCESS in 6m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/17fc5dc36cfd4769a672c45486a0d168 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/500e92a88e7847ecadf54ae964eeac12 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2deaa7ec8940c68318bec25c3f2703 : SUCCESS in 7m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bae7ff39e09642569746af59f622309f : SUCCESS in 7m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3cac8c070855484b8f5f9ebb7c18d73d : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a30183656640479a9c388036776dccdc : SUCCESS in 6m 50s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e45ab5fc3de646a0bc4df66784b91980 : SUCCESS in 56m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0b9df954a50f4eec9d8f6a47425dde3a : SUCCESS in 1h 02m 02s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ed43f7498f2e41b3aa97f8e44721e02f : SUCCESS in 1h 13m 10s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4dd8c7b4873c4566972c57c3b8d6a7a7 : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1858837ab77b4e509e2fbd22c2751216 : SUCCESS in 40m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/edf1e26acec248cea86655646433c91f : SUCCESS in 40m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf40001d71314b2d9e560a3c5ff0234a : SUCCESS in 52m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fb90be58de21400d887fdfd19ef066ea : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":2},{"id":"414ecea88fa68c0cba6043e0f8d51b74d27e8e86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:10:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1812e7dfde9ce231f29c8c5d88cf477494046c35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:11:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bd37cf240743dcb387b41577ad6aef : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb1a594f5e8c0d314c5c0224fa987bee10662eca","revisions":{"501c102f1e88a4d4ebffc3069a14a9916ed4d452":{"kind":"REWORK","_number":1,"created":"2020-07-20 15:52:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/741975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/741975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/741975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/741975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/741975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/741975/1"}}},"commit":{"parents":[{"commit":"d2bef9a47c6497b479d91471fbaec60d993e86ac","subject":"Fix batch member create for v1 amphora driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2bef9a47c6497b479d91471fbaec60d993e86ac"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-19 21:17:29.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-20 15:52:07.000000000","tz":0},"subject":"Prioritize policy validation","message":"Prioritize policy validation\n\nThis patch makes sure that we validate RBAC compliance before\nother validation tasks.\n\nChange-Id: I670087163b265e7098af35063572d6aa9d068bb9\n(cherry picked from commit 8394633635af1045131b1ba5eeeafea8aa98416b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/501c102f1e88a4d4ebffc3069a14a9916ed4d452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/501c102f1e88a4d4ebffc3069a14a9916ed4d452"}]},"branch":"refs/heads/stable/stein"},"fb1a594f5e8c0d314c5c0224fa987bee10662eca":{"kind":"REWORK","_number":2,"created":"2020-07-20 16:26:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/741975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/741975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/741975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/741975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/741975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/741975/2"}}},"commit":{"parents":[{"commit":"d2bef9a47c6497b479d91471fbaec60d993e86ac","subject":"Fix batch member create for v1 amphora driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2bef9a47c6497b479d91471fbaec60d993e86ac"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-19 21:17:29.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-20 16:26:14.000000000","tz":-420},"subject":"Prioritize policy validation","message":"Prioritize policy validation\n\nThis patch makes sure that we validate RBAC compliance before\nother validation tasks.\n\nChange-Id: I670087163b265e7098af35063572d6aa9d068bb9\n(cherry picked from commit 8394633635af1045131b1ba5eeeafea8aa98416b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb1a594f5e8c0d314c5c0224fa987bee10662eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb1a594f5e8c0d314c5c0224fa987bee10662eca"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
