)]}'
{"id":"openstack%2Foctavia~742206","triplet_id":"openstack%2Foctavia~master~Ie82396bacaf9c8fb011560ebd3c5ffb9ffee19c6","project":"openstack/octavia","branch":"master","topic":"story/2007943","hashtags":[],"change_id":"Ie82396bacaf9c8fb011560ebd3c5ffb9ffee19c6","subject":"Change default enable_anti_affinity to true","status":"ABANDONED","created":"2020-07-21 14:17:33.000000000","updated":"2020-09-22 05:58:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aafd278df64503e600013e2d8bbfa4bbc8aa7e4a","_number":742206,"virtual_id_number":742206,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-21 17:32:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2020-09-04 09:42:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-21 16:19:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 16:19:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-21 17:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-04 09:42:05.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"5cc5d4d4b49a3bb4401fd3b83eed09cffcf76f2e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-21 14:17:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9875dcf81ab51de1c0902c7c164ffc02346fdc70","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-21 16:19:17.000000000","message":"Patch Set 1: Code-Review-1\n\nThis means that by default Octavia cannot run on a single compute host configuration, like devstack. I\u0027m not sure this is a good idea.","accounts_in_message":[],"_revision_number":1},{"id":"fce10d9309ecffb0569336bcc3a289cad3927b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:32:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e87dfd79d43247c1bcfa85aaa14282db : FAILURE in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ae3f528ac8d41609f4f133946e8ecfc : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2047161bfcb4e9ea8e9d6c85b772a87 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a80ec4acd2a94d4398dca3feb2b7cdd4 : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38adb1cb9734fb9a8d8cccf1fbfd5ba : FAILURE in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32fe66696bd4e8fb54db0b573d49423 : FAILURE in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/862a8059e0654a5f80ec92d6978ebc30 : SUCCESS in 5m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6b161cba63dd4ca6a5113756cabec786 : FAILURE in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b63d8bd4359a4300bda267532afebe6b : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/678dee495dfc4e1f80ecda135350137b : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70556df43f3f4a80b0c107d9807c0636 : TIMED_OUT in 2h 12m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf96e369180a409580acb320474956b5 : TIMED_OUT in 2h 33m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e5cacb8d1e84bfd91d1acb51a3b8e50 : FAILURE in 1h 21m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a919fc960be4830ab583b9e6d2d6679 : FAILURE in 1h 21m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97f945032a4e476a8ef96dc51231c0b0 : FAILURE in 48m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49e48b6a337643418ec8fca688318021 : FAILURE in 1h 40m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/143d11922f884825aedd3778438e662b : TIMED_OUT in 2h 34m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d5635000f834c7dbdb38618c267d42a : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/942bbaa8f6534760b0eeab1108c79c3f : TIMED_OUT in 2h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/57f31a6da93f4b2082c7eead67d49f6f : SUCCESS in 45m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9d8a358fe9ae4ba4ae5a64fa334389ed : TIMED_OUT in 2h 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d8ce979cf96f0ec4834b19e580f3f0f152b4c1e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-04 09:42:05.000000000","message":"Patch Set 1: Code-Review-1\n\nFor now the default topology is SINGLE, so this setting change seems strange. Maybe we add this in docs - as required configuration for ACTIVE-STANDBY case? And probably would be good to have check and warning message about it.","accounts_in_message":[],"_revision_number":1},{"id":"600730b0586e8de198bdf17ed9279153a441491f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 05:58:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"957c8dbbe58e14667c96d60d481901d3b8aecb03","revisions":{"957c8dbbe58e14667c96d60d481901d3b8aecb03":{"kind":"REWORK","_number":1,"created":"2020-07-21 14:17:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/06/742206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/742206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/742206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/742206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/742206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/742206/1"}}},"commit":{"parents":[{"commit":"f14ccb52ed004f1c30c49e87407f2ce569c045e9","subject":"Merge \"Fix listener update with SNI certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f14ccb52ed004f1c30c49e87407f2ce569c045e9"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 14:15:22.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 14:15:22.000000000","tz":120},"subject":"Change default enable_anti_affinity to true","message":"Change default enable_anti_affinity to true\n\nProduction deployments are recommended to run amphora load balancers on\nactive-standby topology and with Nova anti-affinity enabled for higher\nlevels of availability. The default value of [nova]/enable_anti_affinity\nis now set to True.\n\nChange-Id: Ie82396bacaf9c8fb011560ebd3c5ffb9ffee19c6\nTask: 40409\nStory: 2007943\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/957c8dbbe58e14667c96d60d481901d3b8aecb03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/957c8dbbe58e14667c96d60d481901d3b8aecb03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
