)]}'
{"id":"openstack%2Foctavia~742294","triplet_id":"openstack%2Foctavia~master~I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa","subject":"Added new metric - response time","status":"ABANDONED","created":"2020-07-21 21:46:23.000000000","updated":"2025-01-23 22:10:56.000000000","total_comment_count":17,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"21a45e14c6ae66c6d7ca68b6a7156da2ca355366","_number":742294,"virtual_id_number":742294,"owner":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"date":"2020-09-04 14:11:35.000000000","_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-19 00:47:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":-1,"date":"2021-04-17 16:54:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-1,"date":"2020-09-07 22:15:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-08-19 01:49:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-2,"date":"2020-09-07 22:15:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 18:32:19.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-24 22:36:23.000000000","updated_by":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"reviewer":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"state":"REVIEWER"},{"updated":"2020-07-26 18:14:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-27 07:47:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-19 00:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 01:49:47.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2020-09-07 22:15:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"93edc2a09482f8cf09a4581e23accfa0c7c550f5","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-21 21:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbd0b3415607896a38ac8a063a073974c6b011cb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-21 22:32:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24e77a93f4370e4c3af165278dd574493ebe0a45","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-21 22:35:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d4e31f38241be820327f11dbb91be264dfbe6a6","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-21 22:56:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94a24125f6e86255cc2e73ca000583c72644de43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:50:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaa697c0686146a6938b501f24826e11 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f238bd5b8deb43149e50321acc681d89 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7910d204e914c16a1bd894b574febfb : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc5be7d083b04fd2a7bd47e10f4d4906 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aadd76dc43b47a29d1046add539a605 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab1d06270514d19942b8b49a28b703c : FAILURE in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e97f20d4c0a6400f8351e4be024bcc57 : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/922ce3b6058a443ca4630b61dbe930da : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6c6e6d9d7ff440bc80cc5435b274042f : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/203d6a0184834a268ce9e450781e237c : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fafee8a929f471db0cd2ba4abe94a92 : FAILURE in 48m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24cb291e73d54157a45e01d389711200 : FAILURE in 1h 18m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a496e6a46b94274bf7c6badfda54ddc : SUCCESS in 37m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cc1f0b4ba8e74e6db54acd0277350367 : SUCCESS in 49m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7876f60af9ac445f9c75aaf171ef0c0a : FAILURE in 49m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae171a2ce78843719725cb9b11e02e48 : SUCCESS in 1h 35m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de4d290f173f4da0af07e8837ba14745 : FAILURE in 1h 56m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d074d57eedaf4581a1d0f12d30322fd8 : FAILURE in 2h 00m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d4718cd31d54b359336fcc658eef94a : FAILURE in 1h 00m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c0aa18916a64a56a28b75c78f316221 : FAILURE in 2h 30m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d5630d216c544fb7b2d37b840f92b278 : FAILURE in 2h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47edba877057123c0786ecfc1bcf6f7a90385ae8","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-23 17:45:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8d8b59185d4f80a483697627ec2ac96d684d307e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:08:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ae044ba29454e23a2e7700676b9ea8b : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaeb06d81996434f8800f0ed94225c27 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876673e919cd409cb9f51f46f6d30545 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7673334057d4ce78f771853773a9bb1 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1fb41457944e1d9a5c66e8e1c0e84e : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b539e277820643edb666dc56cbac29b6 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaa677088d234c9fb201e9f44721a36d : SUCCESS in 4m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de581247bafc4226a39175c5f93d90d2 : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f8b6f9fb497d4a5899ab5251e432a973 : SUCCESS in 10m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1b69ce27eb14ec8ae35d4f87373459b : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba48b2e9159b4777b85aee91e9270648 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38f48346e74f4ee595a191c4e8932f5f : FAILURE in 1h 28m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a569a350aae42a4826143707b3ea195 : SUCCESS in 43m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f41d75549a0043aca2e9a2d870183eaa : SUCCESS in 46m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b6a8576de7b48b796fb13506da437f1 : FAILURE in 1h 13m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f4b42fa9a6304172b495b4a796d4039f : SUCCESS in 1h 47m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6dc0b5aca51741a0814dc50271f89acf : FAILURE in 1h 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1be7da879c614da18ca31c360f9b6336 : FAILURE in 2h 06m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8693d9a4c7f34664988031a32096406d : FAILURE in 1h 32m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/caa88fa5313643a789247d12975191df : FAILURE in 2h 02m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0d3c3d5741e040418705e231c256cb22 : FAILURE in 1h 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dedac00ef861698ec358e63651e592b1e1c7a0ab","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-24 00:50:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b891509334cabb102c022abbc763d087293f0017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:00:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd801a4929c54a8682bfed40c4443547 : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13f1fe7d43124c5c97ed764ca8032202 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdae0392172d4e1bbbeb2491f13d16b9 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d513308ddb044718a8c187b09a673e9b : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/455fc486109546458f89ae9646c0cd8f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5611b0cf47f541fb9d32bb5f75bbbd8b : SUCCESS in 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00c2c76b680e43bd9d83bba67d578ca6 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5db57855a60640abaf8316ac9b9f7322 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cbb477fdac3046d9b8e8aa26a14a2312 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6141852c8a904bd391421096cb190570 : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/226e8e46261340e5b5991df22e8da513 : SUCCESS in 42m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca72c0cb0b8545df8f6bf36ac80dc3bd : FAILURE in 1h 51m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7653639c7a594fa3b23611f435cb2ccf : SUCCESS in 30m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/87175f253dee407eabc80ffa667bb02e : SUCCESS in 1h 10m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16b58d3d700a4956a98088c96de8a8d6 : FAILURE in 1h 39m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ccb99eb5bbc3421d92628f6be2be719b : SUCCESS in 1h 30m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df06f75259824c8787b0479c3afadfc6 : FAILURE in 1h 38m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c62fc1096a1b4f5fb5fb917ba53d590f : FAILURE in 2h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2e02095553f943ca9b1f36cc7fbca43a : FAILURE in 1h 50m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/42180a4f14694833b80e07d98da95cbf : FAILURE in 1h 14m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2434e4fec13e47ccb15e6e0ae86df0d2 : FAILURE in 2h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"49fa6df0bf75fa9547ac66e252fa2d137ec3f1cb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 12:07:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2a1932b6e3f220109d1eb5cec1ea46279737d008","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-24 17:58:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ff1610ceba283ea62e4cd45b98cc0975bfc699b4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-24 18:32:19.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"faf9b00ac67927ed69af071a41f2a245bcd7d29e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 20:02:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fedf8b43435f40aca1b609a6017c8ef0 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5bb5b4cfaaa4ddc9e0bee390f4fbfcb : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39ec65ccf4d40faac79ccf3f69a3a02 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7437ed58a4164fd8b0525444cf178cb9 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53a03972bff645fc9ec1e47db5ea17d4 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b252cefe2bb4475a633c44679dfafb4 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11fe3f9b9fea49378a479888e7540b6c : SUCCESS in 5m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ddd616771a945fcad0bc987175303f6 : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5c0d0069a4444e1fa037a6be8e5c248c : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c8c36aebe554ba38423950c9c5b2187 : SUCCESS in 9m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27ea55c765c34d94b7d74aa6aeb5414d : SUCCESS in 37m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dc86d7f7b9844a3b721b2ad87197270 : FAILURE in 58m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a83916dece14e7988c79dab295a0bc1 : SUCCESS in 41m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e074598488d74bf782a8fd6a651ccb9f : SUCCESS in 29m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/003b14f2240345cb91f8f6cfeaa09116 : SUCCESS in 30m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0c56171f13c4abab63329722e4db4d9 : FAILURE in 46m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d60409b476d040088798e36cf54d95c8 : FAILURE in 1h 00m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be6a2ebb24674e35b9e5df7f0689b251 : FAILURE in 1h 29m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4cfd04c364424c298c8f8b4bccf7d7b4 : FAILURE in 1h 14m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1779d60a151b4038928102f3ef0ae13f : FAILURE in 1h 32m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/429610a76c3e478590cf9b56a10b09dd : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa9d8cfdc61400e336fc3eb6d0a7ce718e593c8e","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-24 20:44:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1533a1e0753d6dfb9337145bde62d59718a793fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 22:27:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16501f7dd1904f6fae60f8abdfb2c463 : SUCCESS in 12m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b69aa00734f47ed929e6b2dde3b0cfd : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff76fb205df419e91227b70c50e0c50 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aa9c7622e3a47dcabd39c190d31b01d : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cc28ae8c1d54ed5ad550488293c2ab8 : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee904698acd4621bafcba54fd3258bd : SUCCESS in 16m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c0c680bf9a042abafce6bf3043db947 : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0046e23dc034e7fae14e9e89bbae80f : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4d78331542004035a09d0c92e6521ca9 : SUCCESS in 9m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b5cbe17ef9e4aaf88e88517aaa513b5 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/254e337e58344ada83ddd889cea19a8f : SUCCESS in 45m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98005a177aca40fa89a8c8a6c6bdeb1b : FAILURE in 42m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba5b79eb0f5d432d8dff9153e2751a44 : SUCCESS in 30m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9324f2dbdb0e490897a829ab3409655c : SUCCESS in 31m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65fec66b65da4a6493b8ad925f39e08b : SUCCESS in 30m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a060e38c8cd446018390b65efc6aa331 : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e2a2f748deb04fb184f27b7c8b5a8fd0 : SUCCESS in 56m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b36bb4aff60e49c08ab499a7755604bf : SUCCESS in 59m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb29cb81526d42a8996d82b9cb3924d1 : SUCCESS in 51m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0f8a7ccd67f743f093e05073027ee016 : SUCCESS in 1h 08m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0d1a24cc975a473aba9c91f5029b1ecc : SUCCESS in 1h 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a9f72544f6fb2e489f78669e89f0b302b03a73ec","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-07-24 22:36:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3fc58d660f841ad3340e762c49c7b3e628046fdf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-24 23:09:07.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"5ac722a0e0ed55492db86d06a7b59e986b713668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 00:45:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1b2a1d6b1a647dda07b62e52ae4c288 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d216fa10c5c148858a2ceccca033aa3c : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9219e915e3cf40ef95bbdb21bd337bd2 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a783a60efe949ce90ba121dfc607565 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b867a70c3be4e0099b3b3d6ba97fdd4 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731c11a6a7694d528f2aa349d16d0caf : SUCCESS in 14m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f2333be61314162825c3bdbbf43c80f : SUCCESS in 4m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/069898d783f148f39fd6dc8cea0baaf0 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d1df5858d9ae406ba4e8c175acd8e7d4 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5da57e8bb5c47ec912bef29803007d4 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd922eb95caa4593b5d608ce2d883b6e : SUCCESS in 44m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eabba234ab21499d8b7a72162cc556b1 : SUCCESS in 56m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/daa793e73a9a47a59b315c2c82c93542 : SUCCESS in 40m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a1aba8ca0b114440be373e31ec8a2c83 : SUCCESS in 46m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd07f09f678a45078d39e510f498ab5e : SUCCESS in 46m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a40bcbbac2644657b7ff00c71a7b7daa : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd13c5cd43614150bdcd865168205bcf : SUCCESS in 54m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ec2e403f2584bcfbc531685559e6aac : SUCCESS in 48m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d245c76c5d1c4a8282c526c496e9d009 : SUCCESS in 50m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0952a43282e846c48aa6bdc66a0832ef : SUCCESS in 40m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c379e7e9a6cf4b8397374c51ce233f2b : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"896324678c99bf68312c7974587a56b01c002cdf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-26 18:14:42.000000000","message":"Patch Set 8: Code-Review-1\n\nI have not yet reviewed this patch but I see it conflicts with https://review.opendev.org/#/c/590620/ so I am CR-1 purely for review prioritization. I\u0027d feel frustrated as reviewer, and I believe the patch author and other reviewers too, if the L7 patch had to be rebased and re-reviewed yet again if this one merges first.","accounts_in_message":[],"_revision_number":8},{"id":"580455048391703d81a1f335ee9f4ecafa8124d0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-27 07:47:38.000000000","message":"Patch Set 8: Code-Review-1\n\nWe should also update amphora/listener/load balancer API descriptions in api-ref/\n\napi-ref/source/v2/amphora.inc\napi-ref/source/v2/listener.inc\napi-ref/source/v2/loadbalancer.inc\netc...","accounts_in_message":[],"_revision_number":8},{"id":"4f4cdfc9a334c1017b620587da8d54a40e3f988b","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-27 19:29:51.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"df1179cc9bb8b431b860e3310f70b4aa7adfe669","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-17 23:22:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a3687838f01af2ac4d8115e9c1071a6ff50c5518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 00:51:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78906d5be136403b93f0f3b0c1d8da1e : FAILURE in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34bdeac8e7a74f3aae86a9be128eb6d0 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b9983846fe435e894afd363b0d83d6 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54e9622e78c54b8080f9e48271ddac64 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abd3eb4c16e248769f5135aa72483e49 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc125ac2df0451ba5f29214fa03a5a0 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6901dd3eacb4cc28c5d59927014fc20 : SUCCESS in 4m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ddbe07f2fdf64660834b64c624745ea2 : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c54df5de2394d85b7efd0ba462a0b66 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f94888f9f0154ffdb84627f99a7ed6f4 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3321afb12284be6b346502fa88b15a8 : SUCCESS in 50m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0150d54df0fb49569168914354abcc50 : SUCCESS in 56m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6acb5862c27140e896c65d627e867e99 : SUCCESS in 44m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f4af935c66a347cc8e1ef5d76dbb4c08 : SUCCESS in 41m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c620d661206244979f3c523baedcaf25 : SUCCESS in 43m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/51b8447afc84490a99c3930dd86d5bab : SUCCESS in 1h 23m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0d397b5279f84f38ac8081fdcf49ff10 : SUCCESS in 57m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b948b07eb400419da2a24547158ff29e : SUCCESS in 1h 01m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a81c2591aaec4ee6889cdf1fbcde391c : SUCCESS in 54m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bbc95250af474f799ea6e77432581a6f : SUCCESS in 1h 00m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe1584672ca247b6b3e65701a3d4d1bf : FAILURE in 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"92040f0e33a509f74d597de337e3af79e0598478","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 01:53:43.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"a4202586a820b53b5c39bab96c7100b9cfa197aa","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 05:26:22.000000000","message":"Patch Set 9:\n\nRecheck","accounts_in_message":[],"_revision_number":9},{"id":"bcf7582cf1949aee3b42001d8ce77c071ac4a5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:21:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97f8a8f383764d59a5658d243fc1419b : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0888d576df8e4a248785d37078bd678b : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14d166221f4843d89c273ce632f3741f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57ce76ffe2d6451da79104fc058d9b8f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09c018c7806c4e80a9d49d3abc9eaa83 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9afbfb744e4982a6d2da27d3734265 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3894d3af7b54914aaa2d368395647fe : SUCCESS in 6m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4a4a15703f0146a9a2f49b895b89198e : SUCCESS in 4m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8ab506b158804516b36158ce2b353202 : SUCCESS in 12m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e87b39fd2f414dea923b5401751621f2 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14311f79ce20401192b72f6ba48e02c4 : SUCCESS in 33m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a53d8e027baf4ef98dd58d47835f71f1 : SUCCESS in 1h 25m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a4efdd8b3bc44788c470edea958e33c : SUCCESS in 30m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e7e9764db15d4b8099744aa1a0221a3b : SUCCESS in 58m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7ad4b5233b7436fbb94534b30ab975c : SUCCESS in 43m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/522fc538ab194b3894669bd2bf8ea2b3 : SUCCESS in 1h 52m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ea441dda0494a059fa74d072bb49e83 : SUCCESS in 1h 24m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/97c098aeb363407d83004cecec967c65 : SUCCESS in 51m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2adffbb92bbd47f1b1d032bfb187f25d : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/92f5fa4d67a045b8a765081d507b0efd : FAILURE in 40m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/69bbe92e6ec34da1bd7956c897c68762 : FAILURE in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b71484ca0550f3c36310a1315d0c7bdd5b71a9a9","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 16:25:59.000000000","message":"Patch Set 9: -Workflow","accounts_in_message":[],"_revision_number":9},{"id":"1afff94bc01f4983678fd1da103d7dd4f111b084","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 22:05:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e27211f41c0bb50a136f182e9859297a65e18525","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 22:06:24.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"9192ecb7af9155cdd07451be247669261d41236f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 22:41:10.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"73e522838109d43e4ca35f9bff0263ca25d6ab7e","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 22:52:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"887c2c322ed447301414628bd6a9bb267834e8f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 00:47:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/58cbbf94ff2b4f5f820e49caa4730ee3 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e483c6321af5486e99c7598a494d4926 : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c80c44c4c42b4fdcbe75c01da63b744f : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64443719f59642cfbe3832711af128a8 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9f7fb3fbfee4c8f80c1b4ef5d905d33 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/275f7cb018ef4d7587ec83a5df1ccdfe : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f9cf8d5890475193f6172a7d4a76eb : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a11a9a89d7714c4aa2931b94cb682c58 : SUCCESS in 4m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ad46059bfd0342efb59f1fab2b09890d : SUCCESS in 4m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/445e4d87ef294efaade728d575886d05 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/737bf524d5c3482f825a20437df697b6 : SUCCESS in 5m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00f19194f31840739753ed9a1867bc14 : SUCCESS in 40m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12fca3dc2dc449039dec841527a42905 : SUCCESS in 40m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0eaad28b2e54bbaa74bd9c74a1a69db : SUCCESS in 1h 01m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c86dd30fcd444c42a0ea0bd4e783d1cb : SUCCESS in 38m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/220e72af51ac4dd58a3fd6994a507ade : SUCCESS in 29m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aeaea4cbcc6e4d3bae12f9c4ba603ba9 : SUCCESS in 1h 53m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c38fbf732443462b96479ee826b89db8 : SUCCESS in 52m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f73a48187bb5460e9011f1b5a7e9cd00 : SUCCESS in 1h 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bd1d7f03e3e9486e8b53951a98a1e207 : SUCCESS in 49m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a81a803c1a14b40b34afd8b49f4066e : SUCCESS in 1h 18m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/23f0a6d90dec4b9b98256de268bb83d7 : FAILURE in 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"24d7e3786114b1da74957ccf2154a70087fa49d1","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-09-04 14:11:35.000000000","message":"Patch Set 11:\n\nRebased and updated for the new pylint rules","accounts_in_message":[],"_revision_number":11},{"id":"8d4091616caa31102e05e0ea73ec7667ef04c2f7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-07 22:15:43.000000000","message":"Patch Set 11: Backport-Candidate-2 Code-Review-1\n\n(6 comments)\n\nI think this needs some more work.","accounts_in_message":[],"_revision_number":11},{"id":"11fe1af7febd3589fe7cf694de7b6d25066fd954","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-17 16:54:51.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"21a45e14c6ae66c6d7ca68b6a7156da2ca355366","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-23 22:10:56.000000000","message":"Abandoned\n\nThis patch has not be updated since 2021 and this functionality has since been replaced by the Prometheus exporter. Abandoning","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"80c8604b65ad88bc96642bcd6b8d1302ab639cad","revisions":{"eadc21b6e42a05db08157c0e3badd02999d8981f":{"kind":"REWORK","_number":1,"created":"2020-07-21 21:46:23.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/1"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:45:59.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eadc21b6e42a05db08157c0e3badd02999d8981f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eadc21b6e42a05db08157c0e3badd02999d8981f"}]},"branch":"refs/heads/master"},"afc90d55ad673e3601c431b243eb8d959799e664":{"kind":"REWORK","_number":2,"created":"2020-07-21 22:56:12.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/2"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 22:55:07.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afc90d55ad673e3601c431b243eb8d959799e664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afc90d55ad673e3601c431b243eb8d959799e664"}]},"branch":"refs/heads/master"},"ecb895da36aa94c686a1ddee2937949e8214344c":{"kind":"REWORK","_number":3,"created":"2020-07-23 17:45:45.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/3"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-23 17:45:15.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecb895da36aa94c686a1ddee2937949e8214344c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecb895da36aa94c686a1ddee2937949e8214344c"}]},"branch":"refs/heads/master"},"1451e034f86bfe8b55f818be13ba09e0c596b539":{"kind":"REWORK","_number":4,"created":"2020-07-24 00:50:31.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/4"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-24 00:50:02.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1451e034f86bfe8b55f818be13ba09e0c596b539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1451e034f86bfe8b55f818be13ba09e0c596b539"}]},"branch":"refs/heads/master"},"29a2454b7e208f16368bca0c6266ec909dc61e4c":{"kind":"REWORK","_number":5,"created":"2020-07-24 17:58:46.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/5"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-24 17:57:38.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29a2454b7e208f16368bca0c6266ec909dc61e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29a2454b7e208f16368bca0c6266ec909dc61e4c"}]},"branch":"refs/heads/master"},"d3e851440116d8cdf4f780bd29df9385b27e6c6c":{"kind":"REWORK","_number":6,"created":"2020-07-24 20:44:42.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/6"}}},"commit":{"parents":[{"commit":"1b70a0dbe5514ed79de1ed21985e9af1933129df","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b70a0dbe5514ed79de1ed21985e9af1933129df"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-24 20:44:26.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3e851440116d8cdf4f780bd29df9385b27e6c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3e851440116d8cdf4f780bd29df9385b27e6c6c"}]},"branch":"refs/heads/master"},"d2fd1300c9ab054f29b35561624237dd9886c7af":{"kind":"REWORK","_number":7,"created":"2020-07-24 22:36:23.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/94/742294/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/7"}}},"commit":{"parents":[{"commit":"f5584a0a0aa529c6345ef42166ed21508a17eeca","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5584a0a0aa529c6345ef42166ed21508a17eeca"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-07-24 22:35:32.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2fd1300c9ab054f29b35561624237dd9886c7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2fd1300c9ab054f29b35561624237dd9886c7af"}]},"branch":"refs/heads/master"},"62241822929cb6b0da8a2a03aaaec6d58cc2df22":{"kind":"REWORK","_number":8,"created":"2020-07-24 23:09:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/94/742294/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/8"}}},"commit":{"parents":[{"commit":"f5584a0a0aa529c6345ef42166ed21508a17eeca","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5584a0a0aa529c6345ef42166ed21508a17eeca"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-21 21:07:29.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-24 23:08:48.000000000","tz":0},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62241822929cb6b0da8a2a03aaaec6d58cc2df22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62241822929cb6b0da8a2a03aaaec6d58cc2df22"}]},"branch":"refs/heads/master"},"9ecd0af4493fc80800f02a374926148627dc4459":{"kind":"REWORK","_number":9,"created":"2020-08-17 23:22:12.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/94/742294/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/9"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-17 23:19:58.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-17 23:19:58.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ecd0af4493fc80800f02a374926148627dc4459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ecd0af4493fc80800f02a374926148627dc4459"}]},"branch":"refs/heads/master"},"3fb664e17a6c3a48066f9812422ac95f3e0010ee":{"kind":"REWORK","_number":10,"created":"2020-08-18 22:05:54.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/10"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-17 23:19:58.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-08-18 22:05:10.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fb664e17a6c3a48066f9812422ac95f3e0010ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fb664e17a6c3a48066f9812422ac95f3e0010ee"}]},"branch":"refs/heads/master"},"80c8604b65ad88bc96642bcd6b8d1302ab639cad":{"kind":"REWORK","_number":11,"created":"2020-08-18 22:52:52.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/94/742294/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/742294/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/742294/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/742294/11"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-17 23:19:58.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-08-18 22:43:00.000000000","tz":-420},"subject":"Added new metric - response time","message":"Added new metric - response time\n\nChange-Id: I054fd5505e5ccc2ff66ff5f0c94e8c8ee7afb2fa\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nDepends-On: https://review.opendev.org/#/c/742716/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80c8604b65ad88bc96642bcd6b8d1302ab639cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80c8604b65ad88bc96642bcd6b8d1302ab639cad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
