)]}'
{"id":"openstack%2Foctavia~742382","triplet_id":"openstack%2Foctavia~master~I757d84e4624f488382ec4ff9af3f489cd9a85ed3","project":"openstack/octavia","branch":"master","topic":"story/2007943","hashtags":[],"change_id":"I757d84e4624f488382ec4ff9af3f489cd9a85ed3","subject":"Fix missing dependency in amphora create flow","status":"MERGED","created":"2020-07-22 10:01:24.000000000","updated":"2020-07-24 11:58:51.000000000","submitted":"2020-07-24 11:56:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742382-1595591784315-1d32c05a","meta_rev_id":"582b607a60126d72284011d3d49e4b480bcd8d0e","_number":742382,"virtual_id_number":742382,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 11:56:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-23 20:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-07-23 15:05:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-07-23 12:24:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-07-23 16:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-07-23 20:04:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-07-23 07:01:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-07-22 10:55:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2020-07-22 22:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 10:55:42.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-07-22 22:35:47.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-07-23 12:24:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-23 15:05:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-23 16:38:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-23 20:04:31.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-24 11:56:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9714f159e96da6c7a5907ef8f6d2f1ac84b2e1f3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 10:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa0fe9a25a25e100303370a1eb65ae1fed13e094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 11:38:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14c4fcf1e1004d16a55b549b19bec20d : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4bbd70e05ba4563904c6c78ce55164e : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524686bf64af43118dc4062935bdc51c : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2a51fce1ba3480a85500d2f5b4649c6 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e063a3024bf474da59f149a3384d297 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed758c3fe0044c8abb38fd71e5c2e78 : FAILURE in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/975a3bd0b4e84393b8220e102e2c79f3 : SUCCESS in 4m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/92472745692843789f6500e4e9b3c511 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/70980f01e2b14c8380a1f2a34b0813ea : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/601e107717a641aba1b1cdb479318280 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac6279182d2543b884eb594b974efef0 : SUCCESS in 42m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bed7c3984a274103af02b191618fa286 : SUCCESS in 54m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3460121867b487fa945d4654b120bbc : SUCCESS in 32m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/daedfcd7d2ca4563801a927525a0ebd4 : SUCCESS in 37m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0352a4213224bae98676c9ed445dfd3 : SUCCESS in 51m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8d75846210204cb5bb379388f80f302a : SUCCESS in 1h 30m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dca0bec8229345e0a8e28aa220c7aa77 : SUCCESS in 1h 14m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59510f5ba8ba4d59bd2a65e7ea2868ea : SUCCESS in 58m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c07d14c7b3f84628a6c4b9fe3b1bdccd : SUCCESS in 55m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc5c13b4f6e0468eac27be94f97e59c3 : SUCCESS in 44m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/14ba135a2db149fbb0166ecf38e3ef69 : FAILURE in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35ae6102dd985165677e487b418693b2244276c2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 13:18:25.000000000","message":"Patch Set 1:\n\nrecheck\n\noctavia-lib 2.1.1 released","accounts_in_message":[],"_revision_number":1},{"id":"e3ea65bcc1274ded0f0c256f65d5e23110ca5871","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 14:45:36.000000000","message":"Patch Set 1:\n\nWaiting on https://review.opendev.org/#/c/742422/","accounts_in_message":[],"_revision_number":1},{"id":"ea533aebcc2657820fda4ec111cbad416ceda992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 15:34:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20ebde0089f64b1dbb48639380c9f160 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2f1107ef96e46bcb1e10ee31de0f2c6 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf2c0b1f32c40f18ff9e38e694317d6 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfa63085774f4805b7281029c757cec0 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c2097a640404b61a7f63cca356d195d : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533f644e16284be4b0bd2b48074127e0 : FAILURE in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abe1ea2b4eb144a9a5f975de6a5b96df : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1b20ca6218b443c0a85a88a4d0c8991d : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/746d2776705443269eff39c04f7c1099 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66fad011894b40ba96bf939e7c376829 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/adec251f1c134b9285038c23926ae4c4 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30bff9ef6c2c4313b01283b410fbf482 : SUCCESS in 55m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30c73664e09d4ecc9d214026869fa3bc : SUCCESS in 48m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23d61e3c752046a8bdf7e8c3f7c1ec0b : SUCCESS in 44m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/086be9d8a2a248b8a6517ebecbd07d45 : SUCCESS in 40m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/232da7b32788474d9ea15f4c381e6209 : SUCCESS in 1h 26m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd95c7bf9e12400c8387d562ede6ec85 : FAILURE in 1h 16m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71702023af954dfda963fed83960ac51 : SUCCESS in 1h 16m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a2a8344edfc4b09a16791259957f052 : SUCCESS in 1h 01m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e571b924a8f642ceb0b0eedf88f0c770 : SUCCESS in 45m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fd29ca575c0f46f0b2522958eaea18fb : SUCCESS in 1h 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd8634f9669593800ba4e7b1b63f92741472c871","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 20:07:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13b4de1b5d5f239473423295ceb7f5f30473f587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:08:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d77201f68d434afba0021994af8893dd : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d70ec3002c64df1b1b5be5f794659de : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953d9caa12d04352b658895bdac39825 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f0b3263b1442ea9d1d6a6ade520cfe : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62416cbdcd594455bd8fdf255c74300b : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10db115e988543bd9743369a146039d1 : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901ed6cf7dae4513b17e794ddc21da11 : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9cf25325428d45d4ae57bfcfe9946570 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b79c242fe5d6434cbbd45720c53ac77d : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3030700de7a460dbb1264acc3f34f3f : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f21deab8e7214087a2eb0b381b56e0c9 : SUCCESS in 48m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb65a98d33de4fc19fff322d66507064 : SUCCESS in 1h 16m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a07cf5f2396c430f9de2640f3e5e3165 : SUCCESS in 59m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cfc128120d314bce895b0c4a8a021982 : SUCCESS in 57m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cd1ebe5f35a48a4b7aad5de4327408f : SUCCESS in 51m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38b50dc68b5f47a38ab3529343395434 : SUCCESS in 1h 53m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de407db9259c4da3afe6ad698f6c9dfd : SUCCESS in 1h 03m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62944a5602104e97a9e5d7487f97b3fb : SUCCESS in 1h 27m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55c96c914b8643a58504d5450838874c : SUCCESS in 1h 10m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1a0db61315a440f9bc957a59af0d567a : SUCCESS in 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/669cd15624fd49cf8db9730f0ebb9300 : FAILURE in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d31901b34780ea5b1d1b0188e33e899bac64034c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 22:35:47.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"76810f49b122903d9a9f465128e73eb2b8973d2c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-23 07:01:37.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nPassed RDO testing (original bug reporter)","accounts_in_message":[],"_revision_number":1},{"id":"52633f320de34e48ca6fac4d25cf3b09857a7ba8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-23 12:24:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"505d19042efb045ec37e12d326dbbd719e13c361","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-23 15:05:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03dd6e77807505011cc35ada0e79a70d0d63102a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-23 16:38:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"5372a4a80dd2fb39df5f758cbf8362f7e199c23f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-23 20:04:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4e258074d7b8a7a51bf8370e44aba65c47f61d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:04:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22458b7515337e5b16945a2d44e5c70747e6b66c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:47:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ef1956a33024a3fafe8ca227dab0fa2 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d660f7a64064a358d2325ff313b489b : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2bb7b9d54b44fd9a0731cdece1faf03 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51fc9d2b28734a5fb99ec3a373879fee : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30fc95d532d4bdca06ea0dd5a0489e8 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6224b3a4c26c4244a62a5638472a5098 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1dac4746f47942e8abed8513ebc5a6a2 : SUCCESS in 10m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d2ada219a414221832262b6a434d4a4 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/244457728ff944f5b7a414d20330f27d : FAILURE in 41m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c120d3dbda90439eb7f5adcfd0a85503 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3ee3b776e29e4c0b8206b7ce4a3dbbea : SUCCESS in 29m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09eaadb2198d4df18b74f5efd65ab9f7 : SUCCESS in 31m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a8407083276044d2824a242fc6aae07a : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"9df786339ad2924ec0973ee4edefdd19a29b8374","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-23 20:51:34.000000000","message":"Patch Set 1:\n\nrecheck\nUDP issue","accounts_in_message":[],"_revision_number":1},{"id":"e153410a486ab1033d8ff2109dd1ce0a44c93c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 22:33:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b281ee44994438b506ab3371f9cf4b : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7845aead884a4f57bde95e400070933a : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d1fd880c6847c283bb9438228fbe0f : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93687e72c3114b9b8c2eb72c7f539752 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a8c918ccc0e4996a963a9ca19110b24 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e497ebfd9d4819a5e121bba558b13f : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/129abcb5905d4a5180c959bde9ab9430 : SUCCESS in 7m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/709af64a07a946ddbe93bc0492261645 : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eec298a6460648d3a711af4cf0b09ea5 : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08d64c0133404c108f3a3a826e244319 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16a64527e7bf4158af7abbeea6eb8d84 : SUCCESS in 45m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/012da28195c341c795b8e98e4d182451 : SUCCESS in 51m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04f1254256564d82bcba28bcb24d63d1 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a1e655c3ed9f477ba8785388339a0d03 : SUCCESS in 1h 04m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d354e3c750ad4e749a1e49ef1e8e8db6 : SUCCESS in 40m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d3f6f2b21bf48ef84ef8fd3459a0e1e : SUCCESS in 1h 27m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/67b4c879064046dfb9887f066270d480 : SUCCESS in 1h 39m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bcefb86b7340414fb5f1a255d2579f7c : SUCCESS in 1h 01m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0de1d6d02d69439a86e6b9f9cef87c99 : SUCCESS in 54m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04f47aee53354534bc88242406f83931 : SUCCESS in 57m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d578f5ee509549a4bf5d519248ade604 : SUCCESS in 1h 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1102d3e08e86e0d17346a7990647056b88e9c8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 22:33:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86510a2257630b46a3bbae9858f2dd96b5ca9514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:06:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c3045d90a574a878299d36f93f54ecf : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/503d77bc62c5453f9450e89df93fd14b : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc81b6c656d4481698b44cafa0ea44c2 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/089f017b53be4a139f96d89bfa733ce3 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304fb2dc1f6445efaae0b03d7c1d6670 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b756714eb5f24f46b34f04c3aa4a136a : SUCCESS in 4m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21afce40e2bf49b48819d72293ec8e7d : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0411cf798eb14b6ca18dba1bf48f2be1 : SUCCESS in 48m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fda9fe9e48c4dd59ad68390226495a7 : SUCCESS in 42m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/094d48f6d8114c07bcefe2b09cd4da30 : SUCCESS in 32m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9ca533dcbb2a40ed8af459dc6a699171 : SUCCESS in 30m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83ccb690b77a4817901bfc1f20c971ca : TIMED_OUT in 3h 31m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f97173c2a4b04fc9bd4c0e94b043bbf9 : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"80d049a259c7f0042d8db86dcabf0a4739dab896","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 05:15:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab0681d86d5e2b832bb427022f8042e6d346d376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:25:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6c9c3d4ae44a789f21947a08a68803 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d763bb012ef04591933381a8fa63ea47 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a403602a199444a8bba0d4dd8d8d0dec : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c057acba62b0419bb48fa41076e20774 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d94fec0724541d884dad57742f2a6d1 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc9557a746c45ce9c9d04a6cf501c12 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c06eb61ee0d4a77a61269b536f717d0 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/961e114ebc434155b57f62ff06ddd8b2 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8044e21fad3f4b79a997f2d732ff6697 : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/860715f465dc4fc0974246a02e1a5775 : SUCCESS in 5m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d34d17be2f51417bbc2011f294002736 : SUCCESS in 46m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94ef88fd157f424eb82f2282de71e92a : SUCCESS in 1h 05m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e13dce849ee4d0cad1decf87b8de7a8 : SUCCESS in 33m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1b43643da7cd41e691b19c0e3c9fd23a : SUCCESS in 29m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e70580adc74946f58104afd244e55c8d : TIMED_OUT in 3h 08m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4eaf5cc8440c4292939eec81c571d33c : SUCCESS in 59m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/41c6378d630f42bb83c42bbb46ed54d7 : SUCCESS in 42m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d8c1b299b4140b1a460d79d7632bfc6 : SUCCESS in 52m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65895d3da0d34d068bdd4d0c32cca00c : SUCCESS in 42m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7cfb41fad30a478e96890a596032afd6 : SUCCESS in 44m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3cb1c943d3ec4a84ab1b394b68e825c2 : FAILURE in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee9f288d37fae040ddd9b48d95e3f26667b09fcd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 08:31:12.000000000","message":"Patch Set 1:\n\nrecheck\n\nI don\u0027t know what is going on for the active-standby job to be timing out of a sudden. The tempest test times out as well as task \"Export legacy stack screen log files\" so there are no logs to check why tempest timed out. I ran this exact patch locally on active-standby and Ubuntu Bionic -- it passed 10/10 runs.","accounts_in_message":[],"_revision_number":1},{"id":"e160587bacfa869a9eb1ff936d44ed515aadc05c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:25:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e1792a094c34a928681f0cdf612b10b : SUCCESS in 11m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8765ae92d66d4ac9a6e0d42007f2918b : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2ac9189732443b92fd51242cc02b6d : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b33fe247f4344c3aac38fe3a151ccdc7 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1577d10619e4a92a2044a459f355677 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23d043740c94f18935f49a3acd7ed64 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52379f1f0a444e17b2429b7212c90a68 : SUCCESS in 6m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d46fc581da3940509bbbf9d5ffb42f02 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/67a9f60ca0064c6ab1e5b1d43e325a92 : SUCCESS in 11m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33f0ccaca7774c94b6488d95c0f89ea9 : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a79ed9e380294109bdae7fa694fea9ca : SUCCESS in 40m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92e754c49b164f1db80588502b295ec6 : SUCCESS in 1h 09m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8cfab46132a458cad3cf08d4b87351c : SUCCESS in 32m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1d861541d76a4615a14425ff477b7332 : SUCCESS in 31m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a8720e471ab41c8b1b6ec4021f5b9af : SUCCESS in 40m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3bd3edd926a42828ebf9ad146080336 : SUCCESS in 1h 46m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/621ffa614533490ba5be885bfe5c1a18 : SUCCESS in 58m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5992f140c164af780e3323517b71846 : SUCCESS in 58m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0c28439915124a5eb98f833cbd4938a0 : SUCCESS in 53m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/95f989b0b49d4c158e1eaa63fbedf1c2 : SUCCESS in 50m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7d7d85ca2c654015bf1c72b5c69e2b8f : SUCCESS in 1h 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddd2eaa38e36fc9a610eed561f767948679ff5c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47ec5b57fbd4e303f26ff36c0886f35a573ed969","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:56:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0c9989e42869e24e5d07da1e998fb48cdfbfdae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:56:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80dac75d7ff04c35a9bdc093e8078ce2 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b11b0583414da096ba6b25eac26e9a : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/642e56e3d3bc4574a9087c5ed892ed62 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44ff6c44d96c44779c28654eba76071f : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50df3abb4f0547159b8f95d52914f7b6 : SUCCESS in 14m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/203849c132964b82a370edbddf58d45e : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80c95a60f1814e79b32462e3e0438282 : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e32ebb078e04debacd07756425cfeac : SUCCESS in 43m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9de01856ff6b4bc9bffb0fdb681e5465 : SUCCESS in 1h 02m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d22ea70162f24f238ea384885a49d669 : SUCCESS in 35m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cec037fb3dce46f099ccf73889df0357 : SUCCESS in 43m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3889c718d12d4827857ac40e620afc75 : SUCCESS in 43m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24d25816a11f4f52affda8524bd3b76c : SUCCESS in 1h 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"27f9d9963cc424750b9e10e2a99cbbb1ea788227","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:58:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c198e67505b4d289c6ba6c434b8a953 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3af678528c6444018d39321a1ff1e51f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae6f3443594390fd543a007d67b78132e697fb29","revisions":{"ae6f3443594390fd543a007d67b78132e697fb29":{"kind":"REWORK","_number":1,"created":"2020-07-22 10:01:24.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/82/742382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/742382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/742382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/742382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/742382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/742382/1"}}},"commit":{"parents":[{"commit":"f14ccb52ed004f1c30c49e87407f2ce569c045e9","subject":"Merge \"Fix listener update with SNI certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f14ccb52ed004f1c30c49e87407f2ce569c045e9"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 16:32:52.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-22 09:58:23.000000000","tz":120},"subject":"Fix missing dependency in amphora create flow","message":"Fix missing dependency in amphora create flow\n\nTask CertComputeCreate requires SERVER_GROUP_ID but this dependency is\nmissing when nova anti-affinity is enabled and load balancer topology is\nSINGLE. This patch fixes this issue.\n\nTask: 40408\nStory: 2007943\nCloses-Bug: #1888397\n\nChange-Id: I757d84e4624f488382ec4ff9af3f489cd9a85ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae6f3443594390fd543a007d67b78132e697fb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae6f3443594390fd543a007d67b78132e697fb29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
