)]}'
{"id":"openstack%2Foctavia~742401","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I757d84e4624f488382ec4ff9af3f489cd9a85ed3","project":"openstack/octavia","branch":"stable/ussuri","topic":"story/2007943","hashtags":[],"change_id":"I757d84e4624f488382ec4ff9af3f489cd9a85ed3","subject":"Fix missing dependency in amphora create flow","status":"MERGED","created":"2020-07-22 11:57:06.000000000","updated":"2020-09-09 11:12:04.000000000","submitted":"2020-09-09 11:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742401-1599649665943-a980bdbc","meta_rev_id":"c18bdb496ba2dfb6641f758258b08144506d1726","_number":742401,"virtual_id_number":742401,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 11:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-09 06:43:02.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2020-09-09 06:43:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-09-09 06:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-09 06:43:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-09 06:43:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 22:35:19.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-07-23 16:38:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-24 12:08:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-27 08:02:21.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-07-31 11:30:29.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-09-09 06:43:08.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-09 11:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"262bb4f9cdf4b71c3cd7547445d14abef6ac4b50","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 11:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24818997c366144fff2a854e0de6a53ded0a0a10","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 11:57:45.000000000","message":"Patch Set 1: Workflow-1\n\nWait for master patch to merge first.","accounts_in_message":[],"_revision_number":1},{"id":"830b1011e616a80023c11ffc061fb06e5650c057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:20:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ff1f3468ed1429d9e4a85a308a84075 : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83bdc8f9130e4ff0bc583d3bd78fafbc : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c78aa53dcb5450ebfc8f3f41b56743a : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384099a4f9c7450590bba8ee106198d3 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c52ca38783f34ccba98ff72e065500ff : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0122001c8c141da83e1dd0795a5b1e6 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60aeeed8ffa7426ebd144596fed56984 : SUCCESS in 5m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b18b0e7ec6b94cf7934331db9ed83323 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c753fdf9fc1243a09522a998616e8796 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85720179a9de4872b415bf2879b106fd : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4021cd2e7ab4fa389991e26cc553c83 : SUCCESS in 48m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02505ea3081f42d28305eb70cb01acc8 : SUCCESS in 52m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa20c652f7ee488f83651a2b38818904 : SUCCESS in 48m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e77f56a1e0d849bcaedbb96629b2ccd0 : SUCCESS in 38m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9b96972fa8e49ddbdab993646091ed4 : SUCCESS in 35m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b820ae97924e411b851618c0d80c3c48 : SUCCESS in 45m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adf5c85080c749628a376c7505e171a1 : FAILURE in 55m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d12ae2a3d1074153a79e9f59616b0af4 : FAILURE in 1h 19m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/982a51d0eead4b3e995d54486ed422d5 : FAILURE in 53m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e579f59986d6c55b8509a407ef1c9c78f4136bb9","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-22 22:35:19.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"2e02c911e686080c6a015141ce7ddd7b18f16ed8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-23 16:38:47.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"16d29af4f7b06e48a74c6ccd635817412013d417","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:58:30.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"1bf7869d2e9388ed5b407f5dbf621b30a3760728","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 12:08:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b131e1d993c40b4c6359ca55f7498091a7cf04d6","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-07-27 08:02:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7398182d35abc8716bb6d1e6ea4501d07bf8399","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-07-31 11:30:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a393e0368b45a8cf35faf3c99cb1e5f65c73dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:21:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b44a1d350625432f8de83985ec5763fc : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce7a28e9c064ad894de436edc595854 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba010b39bce6419e81c9dab44c8bec45 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d04f64d2e38c41b585f0f5e8e94f9772 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33aca94d7fa44132b4a65fa589b39f29 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbdd002538674768ac2108d603b7f92a : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c65c7381732d42d7b8023009edd8ce6b : SUCCESS in 5m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/32b01d8a83814065b4507c35711786f6 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0a775cebd073427d90cb2ec82333566f : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f998c3fdfea74804b6a0c26ca4b86137 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f19d3df127a34e04b7d721653f94ba1c : SUCCESS in 45m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1af197e7d1e64b8d82e2398648890dbd : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f9b3317f020474fa9fc653c6c41339b : SUCCESS in 1h 28m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ec2c42e13bc3487a8509bbc9788e2d86 : SUCCESS in 40m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fb3966190c047c2a547c5c3c29d87bc : SUCCESS in 46m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba12b0bc032f4c418f5682b9359b1d9a : SUCCESS in 41m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/673267bea15742db904e42853e2a9782 : FAILURE in 43m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d50229830fb341ecbf8db9e6a70cb058 : FAILURE in 1h 07m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9b9c4885688d4cdc9b7367a9eeeb0e35 : SUCCESS in 1h 41m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dceafdd9ccf2de3136f9b6314872acb30900580","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-22 14:23:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"09b90f17c945fc87505a698c0d715de9df82b413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 16:11:01.000000000","message":"Patch Set 2: Verified-1\n\n(29 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e06022e19974b68b4e3e06eb6e7baf7 : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e813b3efc3e044709ab6ff8b32a6e37f : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946ca73e68074d32a1edde331dab66c6 : FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba6dc57a2ab40fb991b3643df759cf6 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ec1ad64319d4318a783c63562848efc : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b633e972aaa4b0b8aa39d22d746bb80 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1736062b0e6c4a738f0bd0ae7df26c4a : SUCCESS in 4m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d53b25d8dd547d5b335dcf295b2ae7b : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a3b323bacfe543d68c95e01827b614e3 : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8239a0c4195142bfbed34f3350ab81b9 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e3c2da626f849f7a55585dca7b5dab6 : SUCCESS in 44m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d61e3d1abc2a4fee9f96ff0a9efbe980 : SUCCESS in 43m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f7f7a132a5546be861c055d5f37a70b : SUCCESS in 33m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b05c25c97147443b85218d5af7a3ba42 : SUCCESS in 30m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d353e05e698d49be8196a5c1bdbe043d : SUCCESS in 29m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/138b61ccb7bb4a29aec24321533e69fc : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/92532940774147faab755ad7d0e14423 : SUCCESS in 43m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2c11fe13868428a818e8a339aa53139 : FAILURE in 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2197a726db884b108fdf9ff8bac1cd95 : SUCCESS in 38m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/drivers/keepalived/vrrp_rest_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/exceptions.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/drivers/keepalived/jinja/jinja_cfg.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_utils.py\n  Comments left for invalid file octavia/amphorae/driver_exceptions/exceptions.py\n  Comments left for invalid file octavia/distributor/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/controller/queue/v1/consumer.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/cert_task.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/retry_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/healthmanager/health_drivers/update_db.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/cert_task.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/common/keystone.py\n  Comments left for invalid file octavia/common/stats.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/common/exceptions.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/common/base_taskflow.py\n  Comments left for invalid file octavia/compute/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_updater.py\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/common/types.py\n  Comments left for invalid file octavia/api/v2/types/quotas.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/availability_zone_profiles.py\n  Comments left for invalid file octavia/api/v2/controllers/availability_zones.py\n  Comments left for invalid file octavia/api/v2/controllers/quotas.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/provider.py\n  Comments left for invalid file octavia/api/v2/controllers/flavors.py\n  Comments left for invalid file octavia/api/v2/controllers/l7rule.py\n  Comments left for invalid file octavia/api/v2/controllers/amphora.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/api/v2/controllers/flavor_profiles.py\n  Comments left for invalid file octavia/api/v2/controllers/__init__.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/volume/drivers/cinder_driver.py\n  Comments left for invalid file octavia/volume/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/l7policy.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py","accounts_in_message":[],"_revision_number":2},{"id":"a1ae57772754369166a38d760cb66c42c5f28aa5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 20:27:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1503d789a79e3ecfd92424461e257ae60a8347c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:04:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90bd0b446bf44391bc8dec9a8983f5ca : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a75c04db8c844fc5aaf112232d78477e : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90dcaa4cdbc74a0eb0a27c279fe8678e : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/146780f1882144dbac61f2739771544a : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b590c28188094b759828b7a1184eeee9 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2efddfb17b224b21a483c65b3ebd41b2 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d48d17dc226542ecba57cc837d007fb4 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/89a3cf6f76084f7490aca9d5515441d6 : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4ef68705ee64db789c057a3e43d82f1 : SUCCESS in 10m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49c4fa65ecb54afb8cc7675c0a7e6489 : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4fc34111f134ad0a0c8b84978ce23e1 : SUCCESS in 54m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e25c5c8fc60f4b3797e32fc56dc17f68 : SUCCESS in 53m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bbb6472c8b54c7192a47552751d7d6f : SUCCESS in 56m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1ee15283f5b342ab86bcac51868c8e0f : SUCCESS in 40m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b07b4c2295341cf9b8f62ae66e68ddc : SUCCESS in 41m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3cce9dd2d1c448c941136d6eaa310b3 : SUCCESS in 1h 34m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d87c93a76ca5428dbb111b45b1050c57 : SUCCESS in 1h 02m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17f03353fed148bcafafab3cc8086613 : FAILURE in 45m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b1ca11ae163f49188f25a74e24631add : SUCCESS in 1h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"234595657d484ed3b71ee07667cdcd6427615b44","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-07 07:26:10.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"532dd42702ef0bc47b8f6af81ee7a9edea356d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:26:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a5f5b84d434ee03f3eb212f1a879098bfa1e41f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:23:12.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"b06aa4637e09380582747d38babde5140dc883bb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-09 06:43:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"657c19092d5b022bd70f8bf36f65d94bf94504a9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-09 06:43:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"13884befaa516d67e5fa6c1ca981f1913cff2902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 08:57:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a441f778a8a24d6b836fb29c3ed230e1 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38a127beb6ad4b098ea8bfd9a5e75720 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939570015d1b45e9beb0dad68696fd47 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47003a4d47fb4a2eb03912b9fe3e5d2f : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f96f8e9e5b5e45bd861c624f350995f6 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35bc975ce4f49c397612458c803c451 : SUCCESS in 14m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af16aa7651ef401490504fd27d1c4b88 : SUCCESS in 6m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0dcfede7a4cc47389fe33d41532a915d : SUCCESS in 7m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/987120fe0c4547a584b64ae2f37915d2 : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/730ab4fa30564ea69b2d70b0ede70a17 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92ad1aafbd86469e91174fda9ef40853 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f641f9b364344cb1867c8014b2670c50 : SUCCESS in 52m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a43b163c7784b30a99989b969f37904 : SUCCESS in 38m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1ad314a8b51143558cb294ac1823a6b9 : SUCCESS in 47m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b634afc8e8bd493993db4c2e588ed89f : SUCCESS in 45m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/537a92edf11e4eedbe1faed2c11fe4b7 : SUCCESS in 1h 45m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df0a967d91d74d91b72205ee5a71153e : SUCCESS in 1h 02m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a564da1450404309b98c82be2656669b : FAILURE in 1h 06m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c01f2a6f3ec45d19a48fb6edbd8889c : SUCCESS in 1h 07m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"60259cc9e9ed1bd1d954a9a900cdd2c238713af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 08:57:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1ab3ad44a6b9538dddec9ff4317e6ec229958dd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:07:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d7eed8d66c94a23b3c1db488618d1a2 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a120e9515a7f4797a46e90ebe4f4f872 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e54f0a5e074d4cfd9d0336a31be140b0 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/866f8bc7ea314d2ba2413de2c99499f8 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8df4c283534d84a6b80490f3979cd5 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cee0a84ee827441abf9c1ea9e5c64dc2 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7de02ca8e8d84251953505a3fb7d5598 : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09fc33cb9a5748bb87d9a28b0e09771c : SUCCESS in 1h 13m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02baea76fc5a460088e4b139c497f620 : SUCCESS in 56m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/16d2d6d510e84db39a218e248256f5e1 : SUCCESS in 41m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/949fa56ba68648829fd781d53ea3746f : SUCCESS in 40m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/adb0543b002148a39a5a1099c720cebd : SUCCESS in 35m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e73b63eaff374f75bb920c2a1968ac75 : SUCCESS in 1h 54m 21s","accounts_in_message":[],"_revision_number":4},{"id":"15cb0c265c09972abce22416cdd6b7a026c466d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:07:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f50b98ff54ad8940940ddaa0e592e1e2209e7ce3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:12:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2ed409a56d41a28fe93e15336e6e0f : SUCCESS in 1m 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/821c31b650764b29a505a261e6401ed3 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e5cc2493f69ad911cf9bb4d0adcad693aecb06d2","revisions":{"2da18070a642323ad48c981961284a9534ece4be":{"kind":"REWORK","_number":1,"created":"2020-07-22 11:57:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/01/742401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/742401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/742401/1"}}},"commit":{"parents":[{"commit":"176791e675141b37ddf139a9d9f2ae6bb814fa20","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/176791e675141b37ddf139a9d9f2ae6bb814fa20"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 16:32:52.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-22 11:56:40.000000000","tz":120},"subject":"Fix missing dependency in amphora create flow","message":"Fix missing dependency in amphora create flow\n\nTask CertComputeCreate requires SERVER_GROUP_ID but this dependency is\nmissing when nova anti-affinity is enabled and load balancer topology is\nSINGLE. This patch fixes this issue.\n\nTask: 40408\nStory: 2007943\nCloses-Bug: #1888397\n\nChange-Id: I757d84e4624f488382ec4ff9af3f489cd9a85ed3\n(cherry picked from commit ae6f3443594390fd543a007d67b78132e697fb29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2da18070a642323ad48c981961284a9534ece4be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2da18070a642323ad48c981961284a9534ece4be"}]},"branch":"refs/heads/stable/ussuri"},"1623786d4269664c2c0078068bfda1775d184ccc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-22 14:23:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/01/742401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/742401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/742401/2"}}},"commit":{"parents":[{"commit":"44feab3c2e6da78dd224e8fc0aa0087f5bab624b","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44feab3c2e6da78dd224e8fc0aa0087f5bab624b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 16:32:52.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-22 14:23:27.000000000","tz":0},"subject":"Fix missing dependency in amphora create flow","message":"Fix missing dependency in amphora create flow\n\nTask CertComputeCreate requires SERVER_GROUP_ID but this dependency is\nmissing when nova anti-affinity is enabled and load balancer topology is\nSINGLE. This patch fixes this issue.\n\nTask: 40408\nStory: 2007943\nCloses-Bug: #1888397\n\nChange-Id: I757d84e4624f488382ec4ff9af3f489cd9a85ed3\n(cherry picked from commit ae6f3443594390fd543a007d67b78132e697fb29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1623786d4269664c2c0078068bfda1775d184ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1623786d4269664c2c0078068bfda1775d184ccc"}]},"branch":"refs/heads/stable/ussuri"},"0df91b6ca968e8b4e793fbc78404f2b240cb0def":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-25 20:27:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/01/742401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/742401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/742401/3"}}},"commit":{"parents":[{"commit":"b0f6b24863e1305b12022848ae7a4f62c0ab1d47","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0f6b24863e1305b12022848ae7a4f62c0ab1d47"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 16:32:52.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-25 20:27:37.000000000","tz":0},"subject":"Fix missing dependency in amphora create flow","message":"Fix missing dependency in amphora create flow\n\nTask CertComputeCreate requires SERVER_GROUP_ID but this dependency is\nmissing when nova anti-affinity is enabled and load balancer topology is\nSINGLE. This patch fixes this issue.\n\nTask: 40408\nStory: 2007943\nCloses-Bug: #1888397\n\nChange-Id: I757d84e4624f488382ec4ff9af3f489cd9a85ed3\n(cherry picked from commit ae6f3443594390fd543a007d67b78132e697fb29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0df91b6ca968e8b4e793fbc78404f2b240cb0def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0df91b6ca968e8b4e793fbc78404f2b240cb0def"}]},"branch":"refs/heads/stable/ussuri"},"e5cc2493f69ad911cf9bb4d0adcad693aecb06d2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-09 06:43:02.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/01/742401/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/742401/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/742401/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/742401/4"}}},"commit":{"parents":[{"commit":"9b826fca5ced76374c678e18361cffb5c36bf810","subject":"Fix missing params in amphora base and noop driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b826fca5ced76374c678e18361cffb5c36bf810"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 16:32:52.000000000","tz":120},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-09 06:43:02.000000000","tz":0},"subject":"Fix missing dependency in amphora create flow","message":"Fix missing dependency in amphora create flow\n\nTask CertComputeCreate requires SERVER_GROUP_ID but this dependency is\nmissing when nova anti-affinity is enabled and load balancer topology is\nSINGLE. This patch fixes this issue.\n\nTask: 40408\nStory: 2007943\nCloses-Bug: #1888397\n\nChange-Id: I757d84e4624f488382ec4ff9af3f489cd9a85ed3\n(cherry picked from commit ae6f3443594390fd543a007d67b78132e697fb29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5cc2493f69ad911cf9bb4d0adcad693aecb06d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5cc2493f69ad911cf9bb4d0adcad693aecb06d2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
