)]}'
{"id":"openstack%2Foctavia~743059","triplet_id":"openstack%2Foctavia~master~Ifdc24033db288ec615391e576c1c77e6b35aba1f","project":"openstack/octavia","branch":"master","topic":"story/2007958","hashtags":[],"change_id":"Ifdc24033db288ec615391e576c1c77e6b35aba1f","subject":"Fix haproxy user flow log format substitution","status":"MERGED","created":"2020-07-26 15:00:54.000000000","updated":"2020-08-06 20:11:10.000000000","submitted":"2020-08-06 20:09:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743059-1596744582473-ea02c699","meta_rev_id":"24109f3c581feea5a4e432b2b38a71692831ac82","_number":743059,"virtual_id_number":743059,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-08-06 15:15:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 20:09:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-29 09:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-07-27 07:36:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-07-27 21:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-07-29 09:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-07-29 09:48:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 07:36:54.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-27 21:42:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-29 09:48:55.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-08-06 20:09:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3db287de7a13a4809884221095ad1cf96b75e897","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-26 15:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6170f8c153baf8a97ef91740d5f1e4edc2f43023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 17:00:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/474a717ba64c47d4acd8f070a49d413d : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d63eabc3a5c4f869983cdd51029da9b : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2754bd5dd4af43c2b6e862896a3b10d8 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e191f442a33a456abfb594e5fdbf08af : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f9240e1252142cd833c5d6004db3014 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933cf3acc5344624b3624cff63f15ed0 : SUCCESS in 12m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a0607147c9b04435bf5e714cf0c93326 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8e83f92b32c34d56accadb2ab8881b81 : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7fee3d4ecb841a6a97084e5f1d842f2 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e7d1ae880c04ecd921f0a9b0055cf24 : SUCCESS in 58m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19d700f294994f59a621694edb3c0800 : FAILURE in 42m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/964effde6dc2412ba7d91efd85389aa9 : SUCCESS in 1h 06m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/15de77e8f44f465bbe20e9aeb655ea1c : SUCCESS in 1h 03m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5a3f59f9a05458383353c6e50228df5 : SUCCESS in 57m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82bbbad1ef1c4bf7a09a22a3fa12863d : SUCCESS in 1h 31m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91f8b359e74e4952b974dbf3fbeff1be : SUCCESS in 1h 10m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40240a8b6fbc457c8f91098034a3e25d : SUCCESS in 1h 57m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0d38873bfb946f3bf6af6aede8c1512 : SUCCESS in 1h 36m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64f93b670f5840cb9915052c83d2df41 : SUCCESS in 1h 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/02efc7b1ec4a49db9caa04ff8cd70607 : FAILURE in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"751d0899a84917207da5db2685c057b435a5e593","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-26 17:04:18.000000000","message":"Patch Set 1:\n\nrecheck\n\nknown random udp issue","accounts_in_message":[],"_revision_number":1},{"id":"820e515262497976a508785bbfce82b672ff882b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 18:34:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92d85df507be4776a20d2d9b380b4747 : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b42e4a4827c4a07b1ab180c8e8881dc : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c0afdd94724bf29d7f14d95e11d613 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89c542a11c694168a30e8997024ae3de : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14465ec0eafc4c5cb14a49778b4d614b : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9717a31c196b493fa706f2872c64f3c7 : SUCCESS in 13m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/77d1d1f3458d4949822f2b0df1c956de : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/42702c72bc6f4703b0ed6cf0a697965c : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/574142f57589406a9fd813f05d0bac76 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0a26574f6214f2e8c0c300a25f07d10 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ad40fdf26464192aaabea0d2b75f7dc : SUCCESS in 42m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6039c68320e44bcb8d92041b8aa3b2ee : SUCCESS in 39m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/48032b9dc307442e87a7b9c596205723 : SUCCESS in 45m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02b0353291f24adda3cc550fc7b98bd9 : SUCCESS in 30m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6302f4e258514103967c3d5f30386933 : SUCCESS in 1h 29m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9facb62c9e7242399a68709b5f12ec78 : SUCCESS in 1h 12m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ae8bd748e1e45fd8116340be33c87e3 : SUCCESS in 55m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fc7f28ebb40646fab06e8680d4a0e6cb : SUCCESS in 51m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/34a945f3f2db43398a4205589e8b0bc9 : SUCCESS in 1h 05m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7aa47a15ed2b4ee580ef82c5d5ae90c5 : SUCCESS in 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5e8eb10239c994340b715057a89a10f0ff19c0a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-27 07:36:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9765ab1b2b021f689fdc03f2f05687657424fa74","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-27 21:42:39.000000000","message":"Patch Set 1: Code-Review+2\n\nYeah, I think this changed close to the end of this work. There was discussion of if it should be different from a jinja/ansible style substitution so it isn\u0027t confusing/conflicting.\nI am fine going this way with it.","accounts_in_message":[],"_revision_number":1},{"id":"6c661baa4d353e0649b85c89e05f56213033d2de","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-29 09:48:55.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6c49a2535323bb460f4e4249ccaf94342e6988e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 09:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6c6ca0391222db7aaea61bb8785488bc741d6fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 10:46:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e57948ce91aa4043a6a2c523fddb191f : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d46cecc0e94be8b53c996b98240fa2 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9a8b9feb8254964a645d40a4bcbb5ca : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eac9c620a784824ac6f5b86ceee027f : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838262e6528d4e85932610d8ce37e9cf : SUCCESS in 12m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3bc70e8b58d467798b2d8bd5fdf9c1d : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62999cf91fc244e49135bc7a3809233b : SUCCESS in 42m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72a44fb0cb834107bc7318603f0bf65e : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4fb4ba16e782467b8751f266b424a314 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c231022a6c28452ba205541f15e48320 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66048c544de14418bf8176e501b76ecc : FAILURE in 52m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/095879e4cc07471a93b9764b90fab591 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"fb724161296b2d051db7636c86d5135b8bc8f798","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-06 15:15:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"372182fc83e91f308ef2c3cb9a764745b24574d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:07:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c13a603dfd24a558169b43b55930289 : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45502b465ad440e68fef11c29ad648a7 : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa2809f400d4c5cb96357275046d3c8 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eba4acb2611497e83aac8c9ea95ec9a : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e9a634fc1be454b883746fbae3ea63c : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf623c741567420f9eeb8da608949167 : SUCCESS in 10m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4ce2cb3032444b2a81290e4db64dd06c : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0bffde5cfb9a4c4089152e1caecae90e : SUCCESS in 9m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cacb14bc8eaf43708b8824b16849cd6a : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07b31fb5610a492b969770634296370c : SUCCESS in 49m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afddea50e509423997bffcd42105e6c0 : SUCCESS in 48m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8bc87abee9e45e59f091460a63dbfa0 : SUCCESS in 50m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ca0af11ba63743009ddbe5af9bb1cdab : SUCCESS in 53m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd6e7ae813d148babbc8c5ed0b48ee74 : SUCCESS in 56m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e23fcf4af9e142e2ae10c081cf64f363 : SUCCESS in 1h 45m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f109d1e1951c40c2b90e75428f9ff21b : SUCCESS in 1h 24m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2337c39bad3c40619befd1407e61023f : FAILURE in 1h 41m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d48ac41fc744479aca96d230327a981 : SUCCESS in 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d0d271dda4564f8bb9447054d0015afa : TIMED_OUT in 2h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fea92c93d0e24772916ee910e6360bf1 : SUCCESS in 1h 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59a9832a67ed66f1e813f839df0c779213ed2992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:07:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1fad49a5b27959dbb34b3f842f443a596d4b861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:09:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d798747a0bb74ceffef42ec45e20567213084aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:09:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6316afb5804449f69cf700a5119cb29b : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4ca52d72704be191873e8ebba8d08d : SUCCESS in 11m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ac3427e1be24b5a96aea2c603d60916 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1ca30ce0ff94d3084651ea66e84faaf : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db50bb85b504855aa582c5df9c6aefc : SUCCESS in 12m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1235a01530c74a84afa0dfc6c6b12f92 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4ea6188baa04a91ad981568b34fd6d0 : SUCCESS in 44m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb6e338022104928b46e712209ccec3d : SUCCESS in 50m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f59425e4814d4069958724fda0e503d8 : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ba950d8772114246ad7c2408697c7ad9 : SUCCESS in 38m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/416209f50b3a4e50913b0d7f22211315 : SUCCESS in 1h 06m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5be620cf9cca4493b271fdee3006ab1c : SUCCESS in 1h 55m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6dd48ff5499aee9ab74e27a6e0dbd8f5bef75989","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:11:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0c94ff29bb446da02b9a0061ca6923 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9f002770431ea89bc0edb38ecef18d285bef846","revisions":{"f9f002770431ea89bc0edb38ecef18d285bef846":{"kind":"REWORK","_number":1,"created":"2020-07-26 15:00:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/743059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/743059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/743059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/743059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/743059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/743059/1"}}},"commit":{"parents":[{"commit":"ae6f3443594390fd543a007d67b78132e697fb29","subject":"Fix missing dependency in amphora create flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae6f3443594390fd543a007d67b78132e697fb29"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-26 14:40:28.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-26 15:00:33.000000000","tz":120},"subject":"Fix haproxy user flow log format substitution","message":"Fix haproxy user flow log format substitution\n\nThe user log format string has two special parameters: {{ project_id }}\nand {{ lb_id }}. These parameters are documented as being wrapped around\ndouble curly brackets and a whitespace, and are being used as so by\noperators/installers. However, the actual substitution is single curly\nbracket with no whitespace, i.e. {project_id} and {lb_id}. Fix this\nissue by aligning the code with the documentation.\n\nTask: 40444\nStory: 2007958\n\nChange-Id: Ifdc24033db288ec615391e576c1c77e6b35aba1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9f002770431ea89bc0edb38ecef18d285bef846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9f002770431ea89bc0edb38ecef18d285bef846"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
