)]}'
{"id":"openstack%2Foctavia~743476","triplet_id":"openstack%2Foctavia~master~I89519d59ab0f2cf265e41e3def540d725c6e6e89","project":"openstack/octavia","branch":"master","topic":"missing_ws","hashtags":[],"change_id":"I89519d59ab0f2cf265e41e3def540d725c6e6e89","subject":"[Trivial]Add missing space between words","status":"MERGED","created":"2020-07-28 07:35:36.000000000","updated":"2020-08-10 08:35:48.000000000","submitted":"2020-08-10 08:32:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743476-1597048372832-39e11680","meta_rev_id":"c25e546569f70d1b0d216c98785ce430db9f7aa1","_number":743476,"virtual_id_number":743476,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-07-28 15:34:17.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 08:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-10 02:40:53.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-06 11:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-07-28 07:37:07.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-08-10 07:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 07:37:07.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-28 15:34:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-10 07:01:05.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-10 08:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68794aac2f4f6368573f4060430e51c1e339f066","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-07-28 07:35:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e05badac1f0700d8039e89c613c194cec5922c6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 07:37:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f9bf50d2a41e1e83ce53ac793e5b1cf35e53dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 13:25:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c14dbec560ed413a8003bb228bdb94ae : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959591857b764059b16dd83844d7d71b : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc234ced60644a2b5052575e26f25de : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df9657b6a7fe457eb66f6d221690b3f4 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ff95ae503f45859f3bb4cc2f95f2fb : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a8bf2e91a641b6b0cb54b9132a694c : SUCCESS in 12m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/18dc38128fed4f48a0e43266ae2c192a : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7ece3933b3214e2caa8302662bb183be : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1288f27b3f2747e4a1c5e2215692e395 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed1d88aee41749e18c434727a9efe082 : SUCCESS in 43m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6669cb962fd744c49389b570c8a98f4e : TIMED_OUT in 2h 33m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abb1e02297ca4c0d9211f8cc73a5f69f : FAILURE in 1h 15m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/30201e0e49124324973225a3832ff934 : FAILURE in 55m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e8a8863998d4629a65a088f7480bb00 : FAILURE in 55m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8671dd08f13749418838e8191f789b94 : SUCCESS in 1h 50m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/72dc52979c6b45c3b1fb0d68a54be6b2 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b48e87dadcf340dda2a10fcd9436ecba : TIMED_OUT in 2h 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/63689e020d584b8194b56ec8385cb8c6 : TIMED_OUT in 2h 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b089c85a233342bc84d062458b0311b8 : TIMED_OUT in 2h 34m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/db1c028bfe694ba4b68d7b88b1dfa3b4 : TIMED_OUT in 2h 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a23bce632ae576a0d5dd7600507d072847f7d6ec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-28 15:34:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4172a595ce6aabc99cf22f5c19d954a5c26b4f46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 18:18:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9d0d52a64bc449da47c485421569c8f : SUCCESS in 12m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80b37106472649f19dba8ee6b39de786 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c148af26c444c28b3fd78ac806e15da : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/212a3827f3a44a4ba142710e9ae44bdd : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0592432aa2e74e6884353239b9f8cbd5 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbce997acf04cf69ab352624dcd368a : SUCCESS in 12m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d39fa6cf4e74413db6612c23a973bc0d : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/35646db0ea764cec8ebf14e428818ca9 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/863ee99af14c40b98dae7444046e01c1 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb743b10b0804a7bb678e4ef09772cb5 : SUCCESS in 43m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9565a64e4ae04140ad2d608b161852fc : TIMED_OUT in 2h 33m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9561c061650b49d7838b5bb999538e29 : FAILURE in 1h 11m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8995dd3c63334c7e8fbf09343fffad2e : FAILURE in 1h 06m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f4c616781694b3e8f9a2c86db2131e4 : FAILURE in 57m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4be5609a1c94b6eb414ab9bf55eb83b : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc0988e6b43e44978f180d82a9197335 : TIMED_OUT in 2h 35m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc5669b1b97b4779bc98bc8e5dcec72e : TIMED_OUT in 2h 32m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/882250331ee24795b7bdeb097f82f563 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/306f37d677114c2981d35dd18f05e5dc : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a5a9ac45fa5c4a1faf900305a115bb4b : FAILURE in 37m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7d6dbcc495d1b6674a381064788834bda3e89d9","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-07-29 00:56:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b03da27a0d4320f2d9cdc2762fbcff32c46e0c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 04:09:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/090f4274b62746b8b3eec7dec5fd160c : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f411038e3c2e4d469e696c8cef67a252 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63aee434daf04e6e8d6958145ad9d645 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c8fad8464e64da48f8955a878f9c90e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a2ffe2597a440d0840756b6be4607e9 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e84fb7cd4ed4d4b8795d8400785361c : SUCCESS in 11m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d9f2e256af0d40e5bc77709cf37e0aa7 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1ddfe8d39ba24363a0e8494a3c65e03b : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/217ac978c4ee4814a10a57cc78e73028 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ecfa762c948c450bb7df224c1c8af547 : SUCCESS in 37m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be44b7e04fa6489e81fc96aa30fbad95 : TIMED_OUT in 2h 33m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c829d5358e2e4f40a2d11382347ae251 : FAILURE in 48m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0abcb15fb53c4089990deeec851e3b00 : FAILURE in 55m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac92f8b591cb40bd982005916b740950 : FAILURE in 59m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/308733b3a77c477eaeaf93487bf34b77 : SUCCESS in 1h 35m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/490841273ce6464686008d6a4881d862 : TIMED_OUT in 2h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2c5242fcf9041b7a26642119b289083 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a6f0b4d2f0f4861aac8b8bf66a2d0bc : TIMED_OUT in 2h 34m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6873894e4f634c9f9020cdbbb57f77ed : TIMED_OUT in 2h 34m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/87188a7f90444c028ca0f46ed81c6618 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bb48c5517a863db3cac1f0cdfd6f027493d662c","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-07-30 01:27:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c10a8f69c7911a81eadb962a342ee47a70ca432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 04:05:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be6069cae54418d9b860aa06b684561 : SUCCESS in 9m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c5c2d6db30747649e7e1027d362ce0a : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea9c499167b14d6b91ee5cb95015a1a6 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e84de3b24794d5f8cbf25fcd79d53ff : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e1d18e16a15458c97363ffbcda947ac : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839c2053e8644562a699b7237f1a8550 : SUCCESS in 12m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/138d0f69123946838ff434b1dfd1a7d4 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/adcbe87d78394c42bdb6c5f1f096ec29 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7c5699de0a1472fb1c597eed29c1faa : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c3d59fe1ee14487baa1a8f504195bdf : SUCCESS in 48m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/751d507bbec44476bfa106c1c43ada90 : TIMED_OUT in 2h 33m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f583af4896734ff9a4a839b76971e506 : FAILURE in 50m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/faebf1c8944448ea93329d83d5252e33 : FAILURE in 1h 15m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8235420589b7447b81e48cc10661807f : FAILURE in 1h 04m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f71f4805c1174d20849261f26dfe5258 : SUCCESS in 1h 00m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f896c4ed27344aba22fc1f74594fbf5 : FAILURE in 2h 31m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af8420b13dc34b40ab51cedca6b5000b : TIMED_OUT in 2h 32m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4db8046f56f140f7a20942538707ccd2 : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5cb7802bb73546b0b79cdc672461400f : TIMED_OUT in 2h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5499fffff3044ac185f0342a8237816f : TIMED_OUT in 2h 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6cd07d31e97d8912657da4893479c854003d8f8","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-07-30 05:25:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94cfe8812e1dc45501b87234e034431af2b92000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 08:02:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a7d5cd14ebf4319a140dd9ba2753c3f : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e0ac60eaefd4c46a41eb11a6b7614ab : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284879fbc21e48a2bb5cfc6e0a84c0fb : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7b524915e24c78a33d9d8b281108fe : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e3fb0f7872c480bae6c3868ccdd9699 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5418b6b585e64c2bb0006e266d17e227 : SUCCESS in 12m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db87f34958f445e08962eb5840bde3e6 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cffd5729fa17446cb6b6f45a362f99cc : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb501fb68d5444ef9ce3a38dc635290f : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/264b2c3f7bb64e1ca2d09ea6debbb0c9 : SUCCESS in 43m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02b902dc0d2c4318a380500bf996d8db : TIMED_OUT in 2h 33m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45112cd5d5814b729c869c7ffdfe79fe : FAILURE in 1h 18m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/46258d32b8e44a86b813b18355dfa5af : FAILURE in 48m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d73c55e8597c4cd286a84f4f7b5cd62c : FAILURE in 55m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/834be895d9aa4542b6d4d580f9209e44 : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/077ec9535f934eae9ff9f3ea05762a4c : TIMED_OUT in 2h 34m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35e26587c94449babe0d1f01d2d3e074 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3fd2b257c21f47e4bb0e15b79696db25 : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/141a3796f0454a4b9f293c2ae8916938 : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a94d5f7858c949abb4aeeb2dff1b2b1c : TIMED_OUT in 2h 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8d585b78913774f3c2714abb6cdc2aabc98f2f8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-06 11:49:32.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2baeed0052e08ef0d2bf2fedceaf585726660d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:25:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7048ca3f981f465ca58f7d60f3298d19 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/918b15fcb8a3489d9efb1ecd3c2c46b1 : SUCCESS in 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d934abaf85264ffda120e5dfb7f6327a : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/238fe9b3f82d474a9c42468f19974d5c : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6bb0149738144d386075dad229bcd2f : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c7529831254888b8837f07f53c9669 : SUCCESS in 31m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/526fd45baf8641dda98a2ce63792c0d5 : SUCCESS in 14m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a0585002b41c48f0931c7a0ac907585f : FAILURE in 20m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d66d5053ac9845d48fc15b843919c97d : SUCCESS in 9m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1074bd0ef0c148e8bf9d77386912876d : SUCCESS in 1h 19m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04c76591350945c9a0b4800dcf869b90 : SUCCESS in 1h 32m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fdee889941649e6a83db8477226efd8 : SUCCESS in 1h 28m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7eec0f2481914fda9186f362ea1686c9 : SUCCESS in 1h 25m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6439f22653c426a9c7e8bd24ac19387 : SUCCESS in 1h 23m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6bd117e27d3241fe93090dd8ad2ff98c : SUCCESS in 2h 08m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e02c201e53934f739f958eab957667c6 : SUCCESS in 1h 45m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba7ac13ad2344d90b7c4df3fb6e9180d : SUCCESS in 1h 47m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a52dc5eaae6443fba11da8e25791bde8 : SUCCESS in 1h 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a018080e3d144f299e0dfba88dfe765d : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4ed642944e434400935cc24cfbaa2c90 : FAILURE in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96e59b40275b97d06c684e81d21769148b7c0d1b","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-10 02:40:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f97e4cc8195e2778183654077a31a49003ce499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 05:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3be698aa788c499d8e3e3a768c1602a5 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4446297b16dd4ac9810c6c5d3832391f : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cea79811b104b808794b6f642391f12 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2cee378d4e4baa8a22be4f4b598589 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eab6bcf2c617441988902eb1ddf1dbab : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83245bdaa94a4fafb611c86300da1215 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/982b7f769454472d868638cd6612e7d1 : SUCCESS in 6m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/13438a35526f4ea78c66b03b2a2278d6 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bdebb7385a346debbae8562f146d0eb : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4bd0560c3e8242d5bd6e5a07f2bc93d4 : SUCCESS in 57m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fc704565dff4cf6b781389da1378aba : SUCCESS in 1h 27m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7aea07bb81a4a3fa1e0ebaf2a843f76 : SUCCESS in 1h 16m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/35184550603a4933820522353bd232ea : SUCCESS in 1h 12m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97f2548343864004b50bc9678884b1cb : SUCCESS in 1h 16m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46ff4f378b534e3090245baf2d843e76 : SUCCESS in 2h 02m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0961dcfc3d94a51944b7121538850ff : SUCCESS in 1h 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5680f247732a432eb65a11c8b8ce8395 : SUCCESS in 1h 40m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3043c342e1c844089b13e3f8d4399b77 : SUCCESS in 1h 28m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/25fb24c360804d1b96adce2299f84f83 : FAILURE in 2h 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1da9616adcea4dc5b21deef12aa585bb : FAILURE in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"045730656f2d888a955c0c6fce9c603443f34a89","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-10 07:01:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c9119995078d6a8364e23847336963a4c5e9ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 07:01:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65ae24948b2994523e9b69c0a4b113a52720a90d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 08:32:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d863be562bc483238edff1662f41c24aaf75109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 08:32:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf4726969cbb4a31a4c1499d519e5c2d : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6faa374e182c4b6487ff33c24ddb883f : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b76a2787acc5459dbeab6a553ac2eae9 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b860fa4ed244c0184ad8bc5bdea2f15 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7464d21f9974b5dbdc5d881edf17e2d : SUCCESS in 13m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cad041b367ab4344a1353ef37d0ab0ba : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da8e49f842564b4c8b6ba802e6f10b2c : SUCCESS in 58m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d08f0ff4a1547b8a8b181c6ac508276 : SUCCESS in 1h 12m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52eaab1444f3451c83d47422e6a374a0 : SUCCESS in 1h 05m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cf03259f468b4a17ae6393c35c95317d : SUCCESS in 38m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b525ede6858a4472b6603eb0669316d7 : SUCCESS in 31m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/80bf940007dc4c93a83311ae63d8c9c5 : SUCCESS in 1h 27m 07s","accounts_in_message":[],"_revision_number":1},{"id":"91d5e009868eb5056c32639afb7d70b18ea7b796","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 08:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84366e8e90df45dfad278b33a81697ca : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01541ade06a74d7965744be6de49b322380ccdd8","revisions":{"01541ade06a74d7965744be6de49b322380ccdd8":{"kind":"REWORK","_number":1,"created":"2020-07-28 07:35:36.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/76/743476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/743476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/743476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/743476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/743476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/743476/1"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-07-28 07:16:21.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-07-28 07:32:37.000000000","tz":480},"subject":"[Trivial]Add missing space between words","message":"[Trivial]Add missing space between words\n\nThis is to add missing space between words in log messages.\n\nChange-Id: I89519d59ab0f2cf265e41e3def540d725c6e6e89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01541ade06a74d7965744be6de49b322380ccdd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01541ade06a74d7965744be6de49b322380ccdd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
