)]}'
{"id":"openstack%2Foctavia~743520","triplet_id":"openstack%2Foctavia~master~I3efa89992cc4a8e2645803dd867d7d2f6e39b966","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3efa89992cc4a8e2645803dd867d7d2f6e39b966","subject":"Fix amphora image build jobs","status":"MERGED","created":"2020-07-28 12:27:28.000000000","updated":"2020-08-11 17:22:47.000000000","submitted":"2020-08-11 17:20:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743520-1597166408539-60d190a1","meta_rev_id":"f9a178a029abb3a920dd795f14c321e6c3cc2076","_number":743520,"virtual_id_number":743520,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 17:20:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-06 08:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-08-11 15:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-08-11 15:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 08:48:17.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-11 15:43:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-11 17:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca033011aef260d8f72e7d019cf8a8ca091d13cc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 12:27:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e38fdd7d96dfba9e576c457db5ea92460eef931a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 12:27:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5b52e1732c8d2a675dfd364cd9b1623ab552ff60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:33:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/f3bd825e76ea4198ae07a03f46aea8b8 : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"097ac7644f39e4b090c02aad66b161abb089b9f7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 12:42:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9b1b41c6213b6acbda78ba725c164d488adb541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:55:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/226e4c04daef43bd8c00e35532001f3b : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"66099e29211025c0fdd72b0f4b0f5cdae2dcd3cf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 12:58:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19df71727ed6823443ca5f92a3d19c4eeb9fade0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 13:11:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/293cb7837d794539a7268aa515056863 : FAILURE in 2m 39s","accounts_in_message":[],"_revision_number":3},{"id":"481371cb6aa12f29b3e33ac18299f34244f3c470","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 15:14:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ccdb7baf458d348fe9fe66892c81932c6b5ec7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 15:18:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/72c70c00ac1848da85c67a4f16379268 : FAILURE in 2m 26s","accounts_in_message":[],"_revision_number":4},{"id":"bf9b6711a275c832a28adb6dd44afd1cdd5a2136","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 15:21:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"824adef5664fc127df65eb9bef116f1fe6051782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 15:37:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/914b09e3f259498a9d591772ef87fb1d : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":5},{"id":"e2657d0faf4e4d447ba287fe92ca1263bf63393a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-28 15:42:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7fe72f7d19f186e61c3a067c76bb68763a408d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 18:24:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58241874cfcc430285e010c66615a50b : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb038ebb22fc41d0af2de48c87fdff10 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9bdfa99862846c2bddef2d48f28529f : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cd38d940ef74edebf52362db017c362 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18b664b5f4ce454fb3edb3e896f97417 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb9d32849bd42a3b0fc50c85e33fb50 : SUCCESS in 14m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/86e60b70742046bfb2b4698e92baaf31 : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/89e62c7f323f412c93ed56f3ea4eabd3 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7dfbae37c4a74f69a01eed3871c6fb45 : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/647d3871fcc048c9a8009614a1789881 : SUCCESS in 49m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4b5da78f66c43bd9d5e20233f14b94b : FAILURE in 2h 28m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d1b85140fbc45afa864b8f7a4b37ec7 : FAILURE in 1h 02m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2bea4b2ebd8248d9904de242ef500c0a : FAILURE in 1h 07m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95d210ee7319456c8a4885dd6f9f15a8 : FAILURE in 57m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3e24d29dd299459b834269da708e153d : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7dfd8d8c117b4376bf1a27a4cd24ac0e : TIMED_OUT in 2h 35m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3cb0a6174b6441fac97081ea97385bb : TIMED_OUT in 2h 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b7b5eb88952b4e3197a1e2bf5f169331 : FAILURE in 2h 17m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/daee78961e2e4ec4ad60e1fdee0e55c2 : TIMED_OUT in 2h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d62d10b35829479e812bb31d18d8f788 : TIMED_OUT in 2h 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"97152c85290f7d0e79e13632827bfbff8b776940","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-31 06:42:30.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"65a5db7a91ecdccde00f8c330d41dc148525207d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:38:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8768660f8ed7436ab5eed41056a676d3 : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b47711bf741f403ba566cb83028ccbe8 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e46ba857c934d5fac0f7bc94daf32f3 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95be858def843ab917253d87b1b1560 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5767c300b52446fc855c314f8f26ede9 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18aa1c98201143439ce420ebab7c18e4 : SUCCESS in 9m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/94e87879e4f24244a7c6f13793fb837c : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8fc9e5867a1c4171ad85f822e9491b61 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d0fdbd75ef041c2963d2b37b590b2dd : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7013ee051174449a3aea32fb963cea0 : SUCCESS in 41m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6b2b1a572d64ebca9964684a369451c : SUCCESS in 1h 08m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4f9f97debd84d5e87b0d3ee474bb8dc : SUCCESS in 33m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/561179ba235c4733b2a3f138c226bc4f : SUCCESS in 40m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b080d98af43b4edf94d264ba2ba37f02 : SUCCESS in 1h 01m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3dc988c0cd641aaacbda37f0d29e45b : SUCCESS in 1h 32m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fa3585857d3d48f7956a4e6106ac715b : SUCCESS in 1h 31m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/391a5f0e0b914982a59e55bf043bcf5c : SUCCESS in 1h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/86b1a4ad081c46aca07ba02612717934 : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4585935a907a4950b2711b69d45fff2e : TIMED_OUT in 2h 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/35d13a52d50c43979853c23785b453f5 : SUCCESS in 1h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d2aaefd6be0b605bc4369f241b1f89ae6a705b1f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-06 08:48:17.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c438cc29d467fe57052f81d92bf67c6d5ee2f2e6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-11 15:43:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":7},{"id":"1a6352f7873ac1d3c6811b3a68386769589a0bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:44:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9763c5e738c3ee885799991ea578a45ca74b07eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 17:20:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11934de3068c4fcaa65888e92e863842 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e612b39f165649a7b58594e64da4dcde : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e116d143be44b51a63e4b87cb75324e : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29ce7a34efb74b0687968ff274c2ea65 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb2a41687cb4d45acc50f5801d68869 : SUCCESS in 16m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12e4329b98284e0983f4e75f5c824d0c : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/490c6a78832e40f1b929c562822d536d : SUCCESS in 34m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d33dc9f52b374f80b0e8e21743322671 : SUCCESS in 1h 08m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dca963c7a38840258fd9192b43bab3be : SUCCESS in 46m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12c76b6bef954074a55fd71914260128 : SUCCESS in 39m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18894d3253554181b582eedbc06ae22a : SUCCESS in 44m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d75c265829e4eb6b1c95aa91c35ad2c : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":7},{"id":"c83401758be6ea101e4a5bbc4fd859695f2f6d6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 17:20:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f9a178a029abb3a920dd795f14c321e6c3cc2076","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 17:22:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36df8b1fa3894c3083733eacf9466f04 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9df5f75d4913f966582290791c92a64b5e25acd4","revisions":{"78f0798fd4ab90311fa3556ca2f08a4fd91632aa":{"kind":"REWORK","_number":1,"created":"2020-07-28 12:27:28.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/1"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around June 15\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78f0798fd4ab90311fa3556ca2f08a4fd91632aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78f0798fd4ab90311fa3556ca2f08a4fd91632aa"}]},"branch":"refs/heads/master"},"a50cdfc91612f097345f1c62278dd076b2cd987b":{"kind":"REWORK","_number":2,"created":"2020-07-28 12:42:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/2"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:39:28.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around June 15\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install required setuptools package.\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a50cdfc91612f097345f1c62278dd076b2cd987b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a50cdfc91612f097345f1c62278dd076b2cd987b"}]},"branch":"refs/heads/master"},"1ced81aa660979eb01b4e197e0ac130265e0fb49":{"kind":"REWORK","_number":3,"created":"2020-07-28 12:58:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/3"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:58:26.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around June 15\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install required setuptools and virtualenv packages.\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ced81aa660979eb01b4e197e0ac130265e0fb49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ced81aa660979eb01b4e197e0ac130265e0fb49"}]},"branch":"refs/heads/master"},"8f294b9e6c2fb1bfe013b1923e737ed95455c902":{"kind":"REWORK","_number":4,"created":"2020-07-28 15:14:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/4"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 15:13:00.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around June 15\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install required python3-venv and set virtualenv_command\n(defaults to Python 2 \"virtualenv\").\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f294b9e6c2fb1bfe013b1923e737ed95455c902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f294b9e6c2fb1bfe013b1923e737ed95455c902"}]},"branch":"refs/heads/master"},"61cf3e57da4e93a3ea9b8e548715eabec170fbaa":{"kind":"REWORK","_number":5,"created":"2020-07-28 15:21:44.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/5"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 15:21:32.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around June 15\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install required python3-venv and set virtualenv_command\n(defaults to Python 2 \"virtualenv\").\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61cf3e57da4e93a3ea9b8e548715eabec170fbaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61cf3e57da4e93a3ea9b8e548715eabec170fbaa"}]},"branch":"refs/heads/master"},"833d262a383c58c3af98e519bfeefdca257a2772":{"kind":"REWORK","_number":6,"created":"2020-07-28 15:42:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/6"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 15:40:56.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around mid June\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install python3-venv and set virtualenv_command (defaults\nto Python 2 \"virtualenv\") as it seems to be required now too.\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/833d262a383c58c3af98e519bfeefdca257a2772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/833d262a383c58c3af98e519bfeefdca257a2772"}]},"branch":"refs/heads/master"},"9df5f75d4913f966582290791c92a64b5e25acd4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-31 06:42:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/743520/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/743520/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/743520/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/743520/7"}}},"commit":{"parents":[{"commit":"d15cccff2fd382475f971a646f5579bd43ef190b","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d15cccff2fd382475f971a646f5579bd43ef190b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-31 06:42:30.000000000","tz":0},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around mid June\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install python3-venv and set virtualenv_command (defaults\nto Python 2 \"virtualenv\") as it seems to be required now too.\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9df5f75d4913f966582290791c92a64b5e25acd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9df5f75d4913f966582290791c92a64b5e25acd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
