)]}'
{"id":"openstack%2Foctavia~743653","triplet_id":"openstack%2Foctavia~master~I98e16d466d31f7ccdb00f859f68afe489d9bdc14","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I98e16d466d31f7ccdb00f859f68afe489d9bdc14","subject":"[DNM] Create yamas driver for statistics","status":"ABANDONED","created":"2020-07-29 01:05:27.000000000","updated":"2020-09-11 19:11:07.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f556e5379fdccac49e767869ecee9f6908af0fea","_number":743653,"virtual_id_number":743653,"owner":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 02:48:30.000000000","updated_by":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"reviewer":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"state":"REVIEWER"},{"updated":"2020-08-04 00:26:25.000000000","updated_by":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"reviewer":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"state":"REVIEWER"},{"updated":"2020-08-04 23:13:24.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-08-20 23:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"798d6ba2b93bb8c4c62f5c39e968837854a4cb57","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-29 01:05:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"419901673ed9f84885f121a1684341d373bd5868","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-29 01:06:09.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"413a6e9fbc6a202cb17bbd4d09ead904217777b0","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2020-07-29 02:48:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db2f2a34485a15a902b6db19f277e35c14c2c83a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 03:49:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4711027624a445c59b74c4a657cfdf94 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef6114645c614736a217467be9da23df : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92e584c025844c928eefbe48af22ef2c : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d044f8bb554942bd0ec2cb0c2eb67a : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fefe8d0cdd144018528f9dff29837b3 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c1396abb42f45c1b3d5b550d038438b : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2c39333e44487ba04cec6b1bedbe34 : SUCCESS in 12m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce01a1db09d74b498f50e886b984c753 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d6ea5f0a58314dca8a2759764b365d8a : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b2a77b510744fe09aedea4f21b427b7 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e5a873468fc4c20a8c6fdfb8a97099e : SUCCESS in 34m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f793f7db2060446187409547c80a6904 : TIMED_OUT in 2h 33m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05f32d14014b4126a185e5acda89f950 : FAILURE in 1h 07m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/075d94cfea4b49c9b4f236c2c06d840b : FAILURE in 1h 21m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3771f68bc0d94ee4b88acd4bfa830856 : FAILURE in 55m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3856724ddabd434cac6e716ccbe60f40 : SUCCESS in 1h 29m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aa3b5cccaeda4513b410fc7593c660fc : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0d1e8c8e0304e20b1ab575ce9e33e55 : TIMED_OUT in 2h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c6f9301eaafb4f668e2ac75ab6607cc2 : TIMED_OUT in 2h 33m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8e2cf0e59764106b99b26619219e058 : TIMED_OUT in 2h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d42024927f544ab4be367dac4f3cbfd9 : TIMED_OUT in 2h 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d3ab918a9a8d7989567d152c190ebcc99da1ab7","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-29 19:58:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac714f6a0909f6ed9a3b72abacaa7caeba40077","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-03 23:34:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fb3d673a22299709252103ece8ec466912b8f504","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-04 00:26:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e72212389f79e97631b1b11b32dc9de198afb7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 02:59:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85a7f783e9ab48a5bbec5ed92efa7c52 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/066c7695777e484a920260a89efd056e : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d45f0e86e744c6aa39e85470ec1ccfb : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397909ad649e447abaacd897f192d2f1 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7120b2a18af04b64986cc57158ebcd5f : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91514566753f4b3d906b954eafc70c8d : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb45013f18f4798a90a827ecfb67d15 : SUCCESS in 11m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/54d5047d9d5f42e5b0dc40a60426e71d : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/63f16c698545465b81fe167036df6807 : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55297732138c44b89ad284248d69f3bd : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ed14672f0f243dd82c32eede727e535 : SUCCESS in 39m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bd52b82aa2343d591676917242fddce : SUCCESS in 57m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/913641dd6f0e420da3db0ea91019484a : SUCCESS in 39m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f07b9021a2bc47d6bdecf3a9f4584ccc : SUCCESS in 39m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e89152acca24e9da7feebf5621b2b24 : SUCCESS in 30m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db2266f4004a4d6da0f77f3dd9e7d934 : FAILURE in 36m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/63673ef51c5b4b4aa21beda498272eb9 : SUCCESS in 41m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80206a039e8d4f0a891bdf563ad15896 : SUCCESS in 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dff159fa5d014b179ddcfb7e1f4d77be : SUCCESS in 40m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c0ba4a8813949fd8c927a55e1498dfd : FAILURE in 2h 26m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c72c0a42535540f5a56de6b9a81fef34 : FAILURE in 35m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6141fb31664f9e1e0cfdaf20e81a2bb381b4ab41","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-04 22:36:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"180024cc6be29c845c75ec6ba5f3a756970d558e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-04 23:13:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"42ade6f5f3bedd498b414bdbd6c2cb62169e8a75","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-04 23:13:24.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"b9f023ae5c6cfbccaf5c0e5a5e1e28855846a515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 01:35:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/95249ebf9a6a45cca7ae0d0323048302 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3019ca16048b413c972fd35898b830e6 : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b288f4c34f342c58f929b8eef3d00ba : FAILURE in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f203a935e14313984a2f266963f306 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f2951909d444846a725083fb44fd0d8 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7fda6cf88af41b2b3976935b6c80d34 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b421ff9472cc4a1f91ce049bf4e7c4a9 : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e9bfcc36d92e43b9acf7b060998fccad : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e68c0ae19c1948e195c4f4ad0af3c320 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e93c68025e04451881cb005233ad289c : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d700acd4b2f94d7880df2548f4f2c13e : SUCCESS in 49m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2b94303687845dc824d77d9d0663f78 : SUCCESS in 1h 02m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9613f178c74a4f23bbf2d624a287c27d : SUCCESS in 50m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1513e7b1453946db9c383548cb7a3352 : SUCCESS in 42m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8c49400d71b49659d0005e3075bb1d2 : SUCCESS in 30m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a86333ff733648b8a3cda76b9c4fce07 : SUCCESS in 1h 36m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f7756ce00e644f6d99f90bd199563725 : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/864df6ec2b78439d817cb653994046d3 : SUCCESS in 1h 07m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ba5cd9f94a304d82a7bae7444a569fff : SUCCESS in 50m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/007db91326754d59adc16c15ee395f3e : FAILURE in 2h 15m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/93ce1abde32340c9955fffdd80f962e7 : SUCCESS in 1h 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bb2be36e7a5820e2428cbf7463a2c8611168d03b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-08-11 00:45:22.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7a2951c6175c169819ea4cf41886854afaa25b76","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-11 00:47:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f1d0af1e044fe0b3db3c57503ee8c831eceee089","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-11 00:48:14.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"3c6fc4054c15d395569c7fe13e91af67f98638bf","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-11 02:39:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f95ac4fea8af551a0b7187f56b8ff270df5386ff","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-11 02:48:16.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"96a5920d8013b25e51cc8fb4fcfe8b48d24865ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 05:08:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/576a5ab238a545309e366ca1c8eb852d : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f678f19b1b4b45c080c49656f08f3fec : SUCCESS in 7m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ed7d3a604544647b2e1ee0fdcff4aeb : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce2cec1b0ce49d48ef1d5b2ce9120ae : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0012b0630537410e83ebeb3f01a404ae : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e60d7881952041a49b8fe7bca17a4d2a : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa986f7a00b4f508eb93ccf697d219b : SUCCESS in 10m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2940f3151bf84b78b4be785847a00c19 : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7cbf4132c44846b2a738d2351bb4fba0 : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6fbbab21bfee46ea856286f325b632cb : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82036429897243279f6d351f122b2830 : SUCCESS in 42m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ea86fd2adcc4144af80b5c37074c984 : SUCCESS in 41m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0306e122c9c64819a792dfd43f2aef96 : SUCCESS in 31m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0b81bc69087348cb9849dc807cf256df : SUCCESS in 32m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cd8564c40744c45aa7098817843c6f7 : SUCCESS in 31m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad5f1bc7eb044c32ae265fe3337eabdb : SUCCESS in 57m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c1840a09452d4efb809eaf25ea6f3456 : SUCCESS in 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5237245a38a74106a242190414e5c652 : SUCCESS in 46m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/91bceb28d00845208310abc4ac25cbcb : SUCCESS in 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4dacbc4a360e48578d1104a0b8583796 : FAILURE in 2h 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/38821668ca2a4a7dace2939f20511fb9 : FAILURE in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fa0d691aa192b99751a5e6dc6f0ab07205b53789","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-13 21:23:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"495b85435b6fe25f4f3c6c0837056830a0e72496","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-13 21:31:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6dff69ade1beaafba9a385db15c011f86bf4b982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 23:41:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/394dec217e6c4f03ae9b9c2278abdf42 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/208a2c8632d04312a57872d96dcfa50a : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7665c4bd9ab64430b33149d0c752069e : FAILURE in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826bde508d98463798735b5fcf4d4b41 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd1ae8495264520a78c3ff1f3e84d99 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/771992c94eec4fd88b15917dd3327d26 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93f111c125746faa171708d0c42bf8e : SUCCESS in 14m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/89af33ef3ee24a609e161c8027805159 : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/22b7ce7ccc7d4f7297f2bdb0363de76b : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3771828b35e41cd9709c6ddeaf04641 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d0f955f649242038ba4be78fd4f248e : SUCCESS in 49m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd3ed678ad424970a14bedee479aaa88 : SUCCESS in 1h 03m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2eaafbe264be4bb88abe234abe08ccec : SUCCESS in 50m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b1f67393917b4750b2ea90da077abf34 : SUCCESS in 49m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9d39eb68eb84254a4b57aa3ffc649ac : SUCCESS in 58m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/698c4755eab4456ea89cd3b1aad3616d : SUCCESS in 1h 46m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ec1d50c652e40f68059c0d8d37922a8 : SUCCESS in 1h 04m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88575001bdbc428cb784457d5c2ee373 : SUCCESS in 59m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6cba33ef72841b99219f219f3bb42e0 : SUCCESS in 1h 12m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86cb49df683242468aefd4af17670b6c : FAILURE in 2h 05m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d7b5917bf8c14d218b30d8ca0196ca17 : FAILURE in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4b789b26ef5d9a569a63533f2daa7442ebc6d1b8","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-20 21:52:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6f08b77de7c72b09bfe4036d2876ed7b4b03ee04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:41:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dacd373b7cd94248af99edb996621623 : SUCCESS in 10m 26s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6864b646c2d4308af61e8afeebcd2bb : SUCCESS in 3m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/49be77c9ff7143af917da367f5aa4705 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a958b967a825480da017d2c7324d01f5 : SUCCESS in 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82203541f4a7409caac751938d333cd1 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d29c7a60b1c457da75aab48bd13da18 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa67b7162d4e4ff692a8451847b3efb0 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfd64a44c5ee46d8a99a0ea62b3d245c : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de067559487144bfa4df1238e4cf9d6c : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/174f42a617014a24bcc09d80339f2bc4 : SUCCESS in 5m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c8ffe6d603cc4c15938ba28f4a2e489e : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7bd3348f4cd9452592edeaf4321293d5 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/917b69b3a595456ca9ae8830a7bcb165 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ebeac8a02abf41e3ae659e536ac25124 : SUCCESS in 45m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8a13166faf844829b3ea8f509170a75 : SUCCESS in 41m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3fca4ed13ac54a25b1f33b0099b57f8d : SUCCESS in 33m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b8c90d8166fa4ef482952e785818324c : SUCCESS in 30m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a21efeaea044eddb0cf006fd656c278 : SUCCESS in 40m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/675e3f4bf2264ed6a34066cae66e06bb : SUCCESS in 1h 02m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d8cf194a53e49469f46c34398383bb6 : SUCCESS in 58m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b15f280bf24d4225a48dd800dce1c899 : SUCCESS in 50m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f766af331b134ee2a53b17de53935caa : SUCCESS in 52m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b40f745fd5b34a3b8f30416804bda4a4 : SUCCESS in 59m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/34e059646f284ddfa5fcc981b9897083 : SUCCESS in 1h 25m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6bd451b44b28d22981c3e2f729f9e87c30dbdbc3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-11 19:08:08.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"f556e5379fdccac49e767869ecee9f6908af0fea","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-11 19:11:07.000000000","message":"Abandoned\n\nNot needed upstream, will be maintained downstream now that another suitable example driver is available: https://review.opendev.org/746822","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"e9bb8ddde921c90457987e42c9b6169c743e28cb","revisions":{"cd60b97a3058c504b9e1b01ab77038ba5c180fc4":{"kind":"REWORK","_number":1,"created":"2020-07-29 01:05:27.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/53/743653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/1"}}},"commit":{"parents":[{"commit":"42ca88b7306c1aa19f0e5bec4de4e98655dcca2a","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42ca88b7306c1aa19f0e5bec4de4e98655dcca2a"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd60b97a3058c504b9e1b01ab77038ba5c180fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd60b97a3058c504b9e1b01ab77038ba5c180fc4"}]},"branch":"refs/heads/master"},"a650df65fe924e211ebf4e951e558c60798d16c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-03 23:34:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/743653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/2"}}},"commit":{"parents":[{"commit":"e11f859d1a22bf0917b5c98c607606b8ef1c2d37","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11f859d1a22bf0917b5c98c607606b8ef1c2d37"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-03 23:34:26.000000000","tz":0},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a650df65fe924e211ebf4e951e558c60798d16c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a650df65fe924e211ebf4e951e558c60798d16c0"}]},"branch":"refs/heads/master"},"9ae402eff4e16b2cdbd0da95d0487146624d591f":{"kind":"REWORK","_number":3,"created":"2020-08-04 00:26:25.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/53/743653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/3"}}},"commit":{"parents":[{"commit":"e11f859d1a22bf0917b5c98c607606b8ef1c2d37","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11f859d1a22bf0917b5c98c607606b8ef1c2d37"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-04 00:25:59.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae402eff4e16b2cdbd0da95d0487146624d591f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae402eff4e16b2cdbd0da95d0487146624d591f"}]},"branch":"refs/heads/master"},"9eaa1650fb15967ba37e3589fd7a61ed6b05d7c5":{"kind":"REWORK","_number":4,"created":"2020-08-04 22:36:09.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/53/743653/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/4"}}},"commit":{"parents":[{"commit":"e11f859d1a22bf0917b5c98c607606b8ef1c2d37","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e11f859d1a22bf0917b5c98c607606b8ef1c2d37"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-08-04 22:28:17.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9eaa1650fb15967ba37e3589fd7a61ed6b05d7c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9eaa1650fb15967ba37e3589fd7a61ed6b05d7c5"}]},"branch":"refs/heads/master"},"1fbf2cdcc94877a59a5744d083a72e30c253f817":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-04 23:13:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/743653/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/5"}}},"commit":{"parents":[{"commit":"ea85b4b22e8e8db6274fd89b2e0c25649974c0d9","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea85b4b22e8e8db6274fd89b2e0c25649974c0d9"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-04 23:13:21.000000000","tz":0},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1fbf2cdcc94877a59a5744d083a72e30c253f817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1fbf2cdcc94877a59a5744d083a72e30c253f817"}]},"branch":"refs/heads/master"},"268b239a697414b5dd978df156f0d6180d614aaf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-11 00:45:22.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/743653/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/6"}}},"commit":{"parents":[{"commit":"4f6234229aae6a56062c2afa4e9213e9a5ba53fa","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6234229aae6a56062c2afa4e9213e9a5ba53fa"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-11 00:45:22.000000000","tz":0},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/268b239a697414b5dd978df156f0d6180d614aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/268b239a697414b5dd978df156f0d6180d614aaf"}]},"branch":"refs/heads/master"},"51d8396fda7fc43b9c3c3b703d3424cb8d8b81c3":{"kind":"REWORK","_number":7,"created":"2020-08-11 00:47:14.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/53/743653/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/7"}}},"commit":{"parents":[{"commit":"4f6234229aae6a56062c2afa4e9213e9a5ba53fa","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6234229aae6a56062c2afa4e9213e9a5ba53fa"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-08-11 00:46:32.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51d8396fda7fc43b9c3c3b703d3424cb8d8b81c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51d8396fda7fc43b9c3c3b703d3424cb8d8b81c3"}]},"branch":"refs/heads/master"},"b14353d9f3faf6b546be279eb4c335bd39133537":{"kind":"REWORK","_number":8,"created":"2020-08-11 02:39:15.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/53/743653/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/8"}}},"commit":{"parents":[{"commit":"4f6234229aae6a56062c2afa4e9213e9a5ba53fa","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6234229aae6a56062c2afa4e9213e9a5ba53fa"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-08-11 02:39:01.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b14353d9f3faf6b546be279eb4c335bd39133537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b14353d9f3faf6b546be279eb4c335bd39133537"}]},"branch":"refs/heads/master"},"d55841308a840725930b6defd4db09f9ff9ccd60":{"kind":"REWORK","_number":9,"created":"2020-08-13 21:23:21.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/53/743653/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/9"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-13 21:19:57.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d55841308a840725930b6defd4db09f9ff9ccd60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d55841308a840725930b6defd4db09f9ff9ccd60"}]},"branch":"refs/heads/master"},"31b7fcfc0bfddec4878185044e582e587e3e2b62":{"kind":"REWORK","_number":10,"created":"2020-08-13 21:31:50.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/53/743653/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/10"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-13 21:31:06.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31b7fcfc0bfddec4878185044e582e587e3e2b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31b7fcfc0bfddec4878185044e582e587e3e2b62"}]},"branch":"refs/heads/master"},"502ccd75dd1138621db557a16fa768d9c0c9bb7d":{"kind":"REWORK","_number":11,"created":"2020-08-20 21:52:09.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/53/743653/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/11"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-20 21:51:57.000000000","tz":-420},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/502ccd75dd1138621db557a16fa768d9c0c9bb7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/502ccd75dd1138621db557a16fa768d9c0c9bb7d"}]},"branch":"refs/heads/master"},"e9bb8ddde921c90457987e42c9b6169c743e28cb":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-09-11 19:08:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/743653/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/743653/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/743653/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/743653/12"}}},"commit":{"parents":[{"commit":"a446c04f779faffa8262d2f1d2c9ce7e506d6525","subject":"Revert pyroute2 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a446c04f779faffa8262d2f1d2c9ce7e506d6525"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-29 01:01:28.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-11 19:08:08.000000000","tz":0},"subject":"[DNM] Create yamas driver for statistics","message":"[DNM] Create yamas driver for statistics\n\nChange-Id: I98e16d466d31f7ccdb00f859f68afe489d9bdc14\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9bb8ddde921c90457987e42c9b6169c743e28cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9bb8ddde921c90457987e42c9b6169c743e28cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
