)]}'
{"id":"openstack%2Foctavia~743723","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ifdc24033db288ec615391e576c1c77e6b35aba1f","project":"openstack/octavia","branch":"stable/ussuri","topic":"story/2007958","hashtags":[],"change_id":"Ifdc24033db288ec615391e576c1c77e6b35aba1f","subject":"Fix haproxy user flow log format substitution","status":"MERGED","created":"2020-07-29 10:57:40.000000000","updated":"2020-08-14 14:23:10.000000000","submitted":"2020-08-14 14:21:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743723-1597414916108-88ec3396","meta_rev_id":"b1f1ff6af1af4eef9124b632be656efc60934243","_number":743723,"virtual_id_number":743723,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 14:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-14 12:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-13 14:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-08-14 12:41:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 14:41:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-14 12:41:12.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-14 14:21:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee82b2fe9eeab66e85f2ee2ecb30587ebb9ac476","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-29 10:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c132e96812806c89b45ababf72d3e5e5b38bd22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 12:03:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/583b50a5c31b47c390307cdc89fbb57e : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b0b04bf11d5498d8e4f0178f67917e7 : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5fca591cffb4a1c86a2f668d0457669 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae1296a24cb4d84bfdc9a492cfc452d : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3667b2ba78f4440a9d39c62b6aea3f50 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2982d87e3a40ccb1c4c16143745086 : SUCCESS in 11m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c131f0a1fa3b49989bca0a1c01ea771b : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ae8d02b0244455885b142e01b5c6df2 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d806616fc5e248dc8e746a349d121e9c : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfd6f10ecfff4fb3ab3a2264f9560dac : SUCCESS in 37m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fad8e1bb7bd945da8a9b5e8fc8ee5150 : SUCCESS in 55m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b2d4ddc6f87046b19b386732c2400c58 : SUCCESS in 52m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b85c23339ebc496bb577f19c16827477 : SUCCESS in 43m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00b8e9f2ad544d0bbb880dfae7c0b940 : SUCCESS in 41m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3487acc9e26245798216889ba38d5784 : SUCCESS in 44m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71b46ac3f1474f208914a4e001c0d66c : FAILURE in 45m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89d71b3263664791bd5b4079d6e8b50b : FAILURE in 1h 01m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/23387a49cdf947388aea592661591591 : SUCCESS in 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1ff2a7d00b8bbaf5d0e17e73618e773504a25b9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-05 10:24:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f4394aa1dfab313a776f90546704965f76b52949","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-13 14:41:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4484305ca623e8ab8afe64203fcee14bdb2e42a9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-14 12:41:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa5f68945ebfdfbeae7e3c2b693564eb23bfdb01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 12:41:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfdbc47b8a63381a9f2c20f9e2ef2bad6cc6e2d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:21:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849274260511405cabb91ee13656926d : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5776d92dc3a402088a1fe36f82b0cdd : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b843f5f186c4b18a16892fc07c62edc : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/550e5601688a4620a719a421de25bba0 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be46f6f4b4ed4eb8b901c7a5bda13baa : SUCCESS in 11m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8076a708690d449f91e7f58bf1104d8a : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95f86424897d4198aae665448dbb9ea2 : SUCCESS in 44m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3c12b26f10f4cfd8a71ec8b2c57ee10 : SUCCESS in 41m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a69778c9c634ccfa6c8ad5591220f83 : SUCCESS in 43m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6f164d8e1fbc4b26871c9d2fd6931d9d : SUCCESS in 37m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8396b71e2c6d4f0fa089c4dfcf96516b : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7d0252fc823513ffaa5e72cd9450c6b9be1a212d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:21:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cdd70b803ef7a40729498ad22a3f7c385afa8e90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:23:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd05307c828e479aaa0901a8fde61968 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3f61aecf11354e26edaebf5bdfeef9466830401","revisions":{"c3f61aecf11354e26edaebf5bdfeef9466830401":{"kind":"REWORK","_number":1,"created":"2020-07-29 10:57:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/23/743723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/743723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/743723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/743723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/743723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/743723/1"}}},"commit":{"parents":[{"commit":"274393fdd4f157d7c6121b0e2265831f9ff4ca2e","subject":"Prioritize policy validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274393fdd4f157d7c6121b0e2265831f9ff4ca2e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-26 14:40:28.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-29 10:57:18.000000000","tz":120},"subject":"Fix haproxy user flow log format substitution","message":"Fix haproxy user flow log format substitution\n\nThe user log format string has two special parameters: {{ project_id }}\nand {{ lb_id }}. These parameters are documented as being wrapped around\ndouble curly brackets and a whitespace, and are being used as so by\noperators/installers. However, the actual substitution is single curly\nbracket with no whitespace, i.e. {project_id} and {lb_id}. Fix this\nissue by aligning the code with the documentation.\n\nTask: 40444\nStory: 2007958\n\nChange-Id: Ifdc24033db288ec615391e576c1c77e6b35aba1f\n(cherry picked from commit f9f002770431ea89bc0edb38ecef18d285bef846)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3f61aecf11354e26edaebf5bdfeef9466830401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3f61aecf11354e26edaebf5bdfeef9466830401"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
