)]}'
{"id":"openstack%2Foctavia~744047","triplet_id":"openstack%2Foctavia~master~I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","status":"MERGED","created":"2020-07-30 20:34:10.000000000","updated":"2020-07-31 20:30:58.000000000","submitted":"2020-07-31 20:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744047-1596227327270-8e31e638","meta_rev_id":"298ab7b685c8f0f46e19eca4c74936a56cd00fa0","_number":744047,"virtual_id_number":744047,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 20:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-07-31 16:53:54.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-07-31 17:45:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2020-07-31 17:45:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-07-31 16:53:54.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-07-31 17:45:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 16:53:54.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-07-31 17:45:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-31 20:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c20518b50f306535e35229849785ce5691095162","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-30 20:34:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b929be91ecf91bd806ccc7d82670895becf8302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 23:21:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12115dfd2a784a3b802c444ae7478418 : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23afe757082f4ce2bf2fec982a74cb65 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace0e95707c7449b9f005296ee502df6 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfa4e6368ec64edf9a38ea14475018e3 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e0294743ce45ff9b0f2d878fa98901 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9efe243b9e0a471a8ce7a99ea8dabb78 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30b7ba6dc8ab49c78f347feef50ccd37 : SUCCESS in 7m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e12e15722d94d2cb27520855b1b88db : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/55d984fda81944cb8fa7eddb1c1ef1a4 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56f27b15c2dd4812bdd47f52b6f5fd26 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f88b0d5a743c453f8b9a43a4aa6a5918 : SUCCESS in 42m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ceeb43cfbc547e591d59307b981c0a6 : TIMED_OUT in 2h 33m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe044ab2870d4ce2860273f49ead032a : FAILURE in 50m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2184c74db16249a19802fa186ba5f51c : FAILURE in 58m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55e083c6d2364748b0fd50e867cd6f68 : FAILURE in 1h 18m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e84c2739cb547eea82762fa8ba6c70f : SUCCESS in 1h 29m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ac0e89e3083941c88bc36b94f29ab065 : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b807c6111bbe45bbacaf3c4c0e9f326a : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3665ff589eb74b2b92b6a0a98074add4 : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a2fbc21afff4b93b84c9c237ca25634 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/400bc005f0c444a8ac33c47fbe9b3721 : TIMED_OUT in 2h 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7b46ad820c7fbbf2361ac5df7e1d0518c5d0feb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-31 15:51:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e428af406691c998ebcec624ea2a00f59c137cfa","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-07-31 16:53:54.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a4498ca5029f02344d7c78b1713b91cd6bb3fd48","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-31 17:45:16.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"f11a19670fbdf29098f00fef17f98beb049c04fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 18:33:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/896be88b1b8b41629996441a80bd2ff5 : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b2ff2d7b8854ee982fae5978c736791 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf25e7675ba4933bc07d3603cbcd6a4 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb5d859cbdba4a239ddc696859b916cd : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/245b6ec057cf44f8bb3ddfd5bf08cef0 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55205d73abff400fa0f744ac8c585b0a : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95b8f40b36947b98d1b13407d180d27 : SUCCESS in 5m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/98075e2328614ce5896e50483db50ed3 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6f14ac7ab4de41548bbcfb79205d3390 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc1ef85dce92477dac9ce5a2f759d712 : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92b4432ebd6a480fafa99df411a369f5 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9bf7a9cda93d461b83164298f9c63875 : SUCCESS in 51m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e26a5240f15741feb126443d30e8c38d : SUCCESS in 32m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d9cce2e28076432686fe9fcb78ac71ea : SUCCESS in 38m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d514a315fc194bbb98e574ddf7b2d727 : SUCCESS in 1h 13m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/687492b0d9bb46feac011ddaacd3b618 : SUCCESS in 1h 36m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/378978c805d14ef7841be65d7fec7197 : SUCCESS in 1h 00m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07b044c7c87240f7b390b08dfd8cdf83 : SUCCESS in 1h 24m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10c33f48c87749d38b7aebed6579372f : SUCCESS in 40m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a0c685b7a4d74bed9447de3e5918ee81 : TIMED_OUT in 2h 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f416d76921cb417bb80eee4847249410 : SUCCESS in 1h 05m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1426249314d67fc8df93111c4ce03eb6f0629227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 18:33:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3afa1b55057e92f79b5cce0ce26ee7180ef6c86a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 20:28:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73f0728b49b54da98e0ff5922343258c : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93823aca29044f91809f747382939b80 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97a5f1b57957442e9ff874c23f28b0fe : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c242e0f5f9c4954ac3cbcd7043ec599 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edc4c25534a447bbbc941809fc78515 : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b55163256924607aed801973f4765fd : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6168d7eef1f42cb9233d0516d96ca7f : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8ab7eacb490645b5b836befc82b1f4e6 : SUCCESS in 42m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb68857f1e134c7a95cdee7e9ce4ea39 : SUCCESS in 1h 13m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2c64cc88884472c9ff1ef772e128aef : SUCCESS in 34m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f2327a538e1c4e21b9a63e4830433c32 : SUCCESS in 43m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65e3c5d959024f329493e1f2da7b618c : SUCCESS in 58m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74257e03de5542ad94a29869dc4444d6 : SUCCESS in 1h 46m 09s","accounts_in_message":[],"_revision_number":2},{"id":"9b09bb4088a5356112b775183c77012d69f40637","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 20:28:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"298ab7b685c8f0f46e19eca4c74936a56cd00fa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 20:30:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdcd715c072451c999b2e2725006e58 : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6db3332bd8e542039dc90c71324a8695 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08fad7496612962418c81c2d981069bde275157b","revisions":{"7f7b583ae25456b2adde779f1dec87c86295a5d0":{"kind":"REWORK","_number":1,"created":"2020-07-30 20:34:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/744047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/744047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/744047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/744047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/744047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/744047/1"}}},"commit":{"parents":[{"commit":"c78a1b2c299bc5a7bde58e98262ece6e4700b6e8","subject":"Merge \"Add quota support to octavia\u0027s l7policy and l7rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78a1b2c299bc5a7bde58e98262ece6e4700b6e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-30 20:32:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-30 20:34:09.000000000","tz":-420},"subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","message":"Fix accepting \u0027insert_headers\u0027 when unsupported\n\nThis patch fixes the Octavia listener API from accepting the\n\u0027insert_headers\u0027 parameter for protocols that do not support header\ninsertion.\n\nChange-Id: I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb\nStory: 2007967\nTask: 40464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f7b583ae25456b2adde779f1dec87c86295a5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f7b583ae25456b2adde779f1dec87c86295a5d0"}]},"branch":"refs/heads/master"},"08fad7496612962418c81c2d981069bde275157b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-31 15:51:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/744047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/744047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/744047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/744047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/744047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/744047/2"}}},"commit":{"parents":[{"commit":"d15cccff2fd382475f971a646f5579bd43ef190b","subject":"Change amphora statistics to use deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d15cccff2fd382475f971a646f5579bd43ef190b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-30 20:32:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-31 15:51:53.000000000","tz":0},"subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","message":"Fix accepting \u0027insert_headers\u0027 when unsupported\n\nThis patch fixes the Octavia listener API from accepting the\n\u0027insert_headers\u0027 parameter for protocols that do not support header\ninsertion.\n\nChange-Id: I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb\nStory: 2007967\nTask: 40464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08fad7496612962418c81c2d981069bde275157b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08fad7496612962418c81c2d981069bde275157b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
