)]}'
{"id":"openstack%2Foctavia~744515","triplet_id":"openstack%2Foctavia~master~Iedd22e75a9124a41b776305f2fcb0caaabd07d6a","project":"openstack/octavia","branch":"master","topic":"http2","hashtags":[],"change_id":"Iedd22e75a9124a41b776305f2fcb0caaabd07d6a","subject":"Add line terminator to certificate list file","status":"MERGED","created":"2020-08-03 15:48:05.000000000","updated":"2020-08-11 19:50:49.000000000","submitted":"2020-08-11 19:48:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744515-1597175325752-60d190a1","meta_rev_id":"672190c77c96b60843657486b4f4eb466266b416","_number":744515,"virtual_id_number":744515,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 19:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-11 15:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-11 15:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-11 15:33:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 10:23:45.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-11 15:33:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-11 19:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4ac8e849c3792ba08fe595fd9f332c933ed7756","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-03 15:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"348c58ed73fa07c0faba424eeb2a98f92796107c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:31:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82723ffbbcf247c5b34a274b25d7ec56 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5d10ba8327f459e8732a512857dca53 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b3b72ea80a4d4b889a75d202db89db : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f2c419bfad46a383268066c27d7796 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/566c0e3211ec4060bce31b79fa5b965d : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b423bf6b2ec4f359e38523db273ceea : SUCCESS in 12m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7fbe0afc60f64d06a92bd038a318d1b1 : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/55bbffa66bb249c4a9e57e7f34ab7245 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0b4658a26aa4f7282d3e8a56c324645 : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fbdc73263a2a4a3987e0f65c3c4b0a09 : SUCCESS in 44m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/289fbe38c3f344f3aa5371aef055216a : SUCCESS in 57m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc56ab6d894f45579a09f6c018e6dcc9 : SUCCESS in 32m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/08f674e0d6984f1199f9f98286ac8b6c : SUCCESS in 46m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c335fc495a294fb2b82319e6ebacbdb9 : SUCCESS in 39m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/861aaf81d4d34f348472580b3dfd7643 : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aa7e7cd2392f488286f51f83ef80b8b5 : FAILURE in 1h 27m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abb2455e2f984e9c9613c2767fadb072 : SUCCESS in 50m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2fb0ee931ec249e8856647720ae8e947 : SUCCESS in 50m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2547ce17e95948cdb2985bdc5a5c1181 : TIMED_OUT in 2h 34m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/511a0a72ceb742179829c52d2d584d31 : SUCCESS in 1h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1811b4ba961b358a8303c05bcde92cd61d61bcb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-05 10:23:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f111d9ed18afe1d0b5f0ec125430f4a57166f010","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-05 13:39:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"158f01de82b51afe942c7fe4a93b4811993b7ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:25:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1ac2233fdaa4103ac9eb07a14bbb3ec : SUCCESS in 10m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a92e30807b44fe386c1bbd6e886b5ba : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54762f60da1b4c89a5dc55410505486a : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/089a8b2d3c3d4eb0b48f5df6d12b9a8e : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbdebe88df8143e994eedfe9be3987c4 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38467dfa805142839757778eddc1745d : SUCCESS in 13m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/58c73c9f338e4d14be593fb090c916ad : SUCCESS in 7m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e0924bbc7bff4445b7c540f8bbcf179f : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c740f1d642314b04a20f6dc2e4c6588c : SUCCESS in 10m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad2f466d324742bc97f3a1bc7abe5f1b : SUCCESS in 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbb44c69d6af45a88093ba1a2230359a : SUCCESS in 1h 03m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e660950ffe84ce8b01cbeb81f701d2a : SUCCESS in 42m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f2c7e25427a9488992b8c97821b08891 : SUCCESS in 33m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1529eed4a33d4884aae23504b821e54c : SUCCESS in 48m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/882e177b879b467983c2fc6ec7c01e22 : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/970c5497b3424609be6f81325ebb84ff : SUCCESS in 1h 06m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d109d237671b4a30b2819152a678fc0f : SUCCESS in 1h 07m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b43e1f874342cb971f2dc70bc706d5 : FAILURE in 43m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2fb048513a2f4e54b3adfee49fea97bb : FAILURE in 2h 32m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8da02c6eaf7442c69c28d46248367063 : SUCCESS in 1h 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd56b3c83afeafddce8fc678978de2981bebb51d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-11 15:21:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a6e84ac98cabe2e60289553a4105dc6f50108ff6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-11 15:33:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM\nPassed the barbican/TLS gate so doesn\u0027t break 1.8","accounts_in_message":[],"_revision_number":3},{"id":"d2a4a17dce6959319033179c87ea88e49da1756b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:15:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9933f890e918471fa0929e822d12fc55 : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4fef3b74502455e9a1dd8aba600a0bc : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd3ba4a2d1e4f4f98de500d437d3a27 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/041e6377c476449ca973cd453a9e288c : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6d8970200c14089b14a19ad9bc0c816 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2cc3922d7a5422d9b929b7664e0afc0 : SUCCESS in 11m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf939fdb0d364283bd36334503046833 : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/02d962cd433544138b288c0b528c05b8 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5d204920ab643cb90c3c7cdc31571dd : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8b880e1aa8642248c6c4ffb7cce3296 : SUCCESS in 39m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/610edd2df7cd449ca87db83c78ffe1f4 : SUCCESS in 49m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28ad2bf1c60f4d889febc2e305737989 : SUCCESS in 44m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ce70676c4fb2415c9f5fc2b3efb25449 : SUCCESS in 38m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e1bcf6f1dd24f8b9de926d9f1645f65 : SUCCESS in 33m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e88bcabb005041699c76fe3bb8c5a3d1 : SUCCESS in 1h 25m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/848866a28e1f4855809413048a26948a : SUCCESS in 51m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c030aa5f44c94a26aa4261863e2ec781 : SUCCESS in 56m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6aca231b3de34b7198851ae3225a147c : SUCCESS in 1h 11m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82311a11f8344550894dc16c6bcf81e4 : FAILURE in 2h 25m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7832db908527416b97cac42c2b731aea : FAILURE in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ebe144a1b51c29b2b7d94915a5b89ce9e8608c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:15:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"adadee7c5af551fc3006b7fdc91ec295f4bafdb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 19:48:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0a20253032e4084bc9454de7dd3f334 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96bdb7f9ad9473b8868b700d11e48e6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3b3e422ded4690920d47cd2adacadd : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cda2f357db4f4f1daa9b398c65bc77a8 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2b0e1d21804877ad633f3caffbe2d1 : SUCCESS in 11m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cec808d979424b72bc4baaac3e4c0198 : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4d337d19c76490eab4c90f052287afa : SUCCESS in 44m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72ae92d3a1da44e0bd25caa71c3e23b4 : SUCCESS in 57m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/906943d2abd1495a9015116d2f3879bd : SUCCESS in 39m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e87d236fbf3b4149b1f81c7274f1e248 : SUCCESS in 40m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27bb5a3467fa4ecb9b36438bd416a38d : SUCCESS in 40m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6545fc24fa7c4796901c4a535e04b481 : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":3},{"id":"2aadad532acf69d7603833802b6908eeaad17c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 19:48:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1cc531bf55ab63488d49149e8de13b3ad9e721d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 19:50:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b7a494daf04b3e9bbdf48a91f3fef6 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"13fe928c8ad979568f66ba738b86cc2d91e936fa","revisions":{"2f220368c5e7237dd1884b2bb7eea09454e38fa3":{"kind":"REWORK","_number":1,"created":"2020-08-03 15:48:05.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/15/744515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/744515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/744515/1"}}},"commit":{"parents":[{"commit":"01a07afa1f069873f1f3eacec0691a458a96c5f4","subject":"Add Copr cgoncalves/haproxy repo for CentOS amps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01a07afa1f069873f1f3eacec0691a458a96c5f4"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:18:22.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:46:58.000000000","tz":120},"subject":"Add line terminator to certificate list file","message":"Add line terminator to certificate list file\n\nCommit c9e1551 changed file mode to binary from text. When on binary,\nPython does not apply newlines so the resulting file has no line\nterminator. This has not been an issue up to HAProxy version 1.8, but\ntesting on HAProxy 2.2.2 fails configuration check.\n\nWorkaround this HAProxy issue by adding a line terminator to the\ncertificate list file.\n\nTask: 40593\nStory: 2007974\n\nChange-Id: Iedd22e75a9124a41b776305f2fcb0caaabd07d6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f220368c5e7237dd1884b2bb7eea09454e38fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f220368c5e7237dd1884b2bb7eea09454e38fa3"}]},"branch":"refs/heads/master"},"67751ec1e1b52020dc8db6414f276d791b57791e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-05 13:39:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/15/744515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/744515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/744515/2"}}},"commit":{"parents":[{"commit":"08fad7496612962418c81c2d981069bde275157b","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08fad7496612962418c81c2d981069bde275157b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:18:22.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:00:54.000000000","tz":120},"subject":"Add line terminator to certificate list file","message":"Add line terminator to certificate list file\n\nCommit c9e1551 changed file mode to binary from text. When on binary,\nPython does not apply newlines so the resulting file has no line\nterminator. This has not been an issue up to HAProxy version 1.8, but\ntesting on HAProxy 2.2.2 fails configuration check.\n\nWorkaround this HAProxy issue by adding a line terminator to the\ncertificate list file.\n\nTask: 40593\nStory: 2007974\n\nChange-Id: Iedd22e75a9124a41b776305f2fcb0caaabd07d6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67751ec1e1b52020dc8db6414f276d791b57791e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67751ec1e1b52020dc8db6414f276d791b57791e"}]},"branch":"refs/heads/master"},"13fe928c8ad979568f66ba738b86cc2d91e936fa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-11 15:21:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/15/744515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/744515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/744515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/744515/3"}}},"commit":{"parents":[{"commit":"6a762f8a072ab3b3eaede32c6d471d874f2ad945","subject":"Merge \"[Trivial]Add missing space between words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a762f8a072ab3b3eaede32c6d471d874f2ad945"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:18:22.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-11 15:21:27.000000000","tz":120},"subject":"Add line terminator to certificate list file","message":"Add line terminator to certificate list file\n\nCommit c9e1551 changed file mode to binary from text. When on binary,\nPython does not apply newlines so the resulting file has no line\nterminator. This has not been an issue up to HAProxy version 1.8, but\ntesting on HAProxy 2.2.2 fails configuration check.\n\nWorkaround this HAProxy issue by adding a line terminator to the\ncertificate list file.\n\nTask: 40593\nStory: 2007974\n\nChange-Id: Iedd22e75a9124a41b776305f2fcb0caaabd07d6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13fe928c8ad979568f66ba738b86cc2d91e936fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13fe928c8ad979568f66ba738b86cc2d91e936fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
