)]}'
{"id":"openstack%2Foctavia~744520","triplet_id":"openstack%2Foctavia~master~If08a8169498cdfaa75440e8971ba0caff45ac4c4","project":"openstack/octavia","branch":"master","topic":"http2","hashtags":[],"change_id":"If08a8169498cdfaa75440e8971ba0caff45ac4c4","subject":"Add ALPN support for TLS-terminated HTTPS LBs","status":"MERGED","created":"2020-08-03 16:13:46.000000000","updated":"2020-09-01 11:43:54.000000000","submitted":"2020-09-01 11:42:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744520-1598960536625-9d651cf1","meta_rev_id":"0593a036dc2bb20b66db7d97088abf530f1f6251","_number":744520,"virtual_id_number":744520,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 11:42:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-31 07:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-28 19:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-28 16:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-08-31 07:35:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-2,"date":"2020-08-31 08:37:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 16:21:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-28 19:30:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-31 07:35:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-01 11:42:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dfe3d66a4e5e78452985083410609b386832267","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-03 16:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8544de24dd88e5f58ed72c37b77008cbd085a10f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-03 16:14:40.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"25945f8eb4db74ff0b507993e36d9243fb389bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:59:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d33c474a456c4a1bba0456ec98bf05d2 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/085c7eb853da4a339a53aa81f1caf979 : FAILURE in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47730167ce6441a8b6aeed89da707950 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db0cd323a554c73b129f66e5011be6d : FAILURE in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fefaeb40c0f4494f825598219fcf8cb1 : FAILURE in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff128850da6542e78f50c15a5b41819c : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/494d5bfd480043bb904f4349c9a8b01c : FAILURE in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ecd91e540387432daefbcbefa1cef2d8 : FAILURE in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9a590f77c013458f82615022f9dac3bd : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fddc55292546435eb4eff7f1aa22a6ec : FAILURE in 3m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c1be435805a4bd5b94bfcd58e1fede5 : SUCCESS in 39m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce6a4dafc2964d5da84eb956227c202c : FAILURE in 2h 30m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e901fccb186b471f90927839fb082d43 : FAILURE in 59m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f4c08614873f48b9992d87f3a1a2cf8c : FAILURE in 57m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01aa33aca55b4262bc69d796595b9b74 : FAILURE in 53m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a1baf61f47d54cdeb87c49f6f777cdb6 : SUCCESS in 1h 36m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a91bd966437a43bf9ec7c3b1f3092534 : TIMED_OUT in 2h 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1afbb65e1754f59bf9584d305e894ba : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d700172b325641c385780e44a69893b2 : TIMED_OUT in 2h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d992a509b462405ca0e7c4424aebdfdd : TIMED_OUT in 2h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/160afa89584b4ac9ba4e179948a8bb9b : TIMED_OUT in 2h 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3d7f43eb81e97b0541f7a7c4e902a25dd4cf1f4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-04 16:35:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b1993cd90860310c922b7b8c4db2751c94ad75f6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-04 16:37:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9b64d23adccaa708633f71599405faafcde047ac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-05 01:21:50.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d424aa74bbbe6c181e92293da7668fcf3bfb6c20","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-05 13:39:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e804573b0b112d3f30246bacc9af3c94d3e09721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:44:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e3a6459541340d4b0ff3bc382c58908 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31ba3fa16dd34646913871e803cd0552 : FAILURE in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27f62a7df09143708cd42a2413f8557b : FAILURE in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e7e42abba94f6cb9c59ca9d5a5fdac : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d30ddfa0df74fca9eccf48e371a8c00 : FAILURE in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f492f8be5f6a4653aaeb3baa9a2c21fe : FAILURE in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ca09f0f5bf4281876ccbce2d7277f4 : FAILURE in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66445dcc0a81466f90933807fc631f97 : SUCCESS in 6m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f6ccf08e61ff48deae45e52579778bdf : SUCCESS in 8m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/58a60eb7a58d4e5abb4b0595560c38de : SUCCESS in 11m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5adcd328c18485f83132b5ba9696fdb : FAILURE in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3cd0f806ba794a3582961fd104e727da : FAILURE in 48m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4ee6c6e0a3b4060b76d0b915cb066c4 : TIMED_OUT in 2h 32m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30359ff0643c4615ae95c173f37fe8ce : FAILURE in 1h 15m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/258510856ddc484da05ef441f6b42517 : FAILURE in 58m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2632dba130614eb3b839a45931481a3d : FAILURE in 1h 00m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8afa58d6ad3143309dcae04c84f0561c : FAILURE in 1h 27m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5dbae6c20b6b4cfdb594ac330ad32e4e : FAILURE in 2h 15m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58461525b20b4f79b0f2f16e451daf3d : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aa1eb98ec3c747deb7837603acc7a12d : TIMED_OUT in 2h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c15b4882fdd94220b770e25d5140378c : TIMED_OUT in 2h 34m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a07661b13c384c53bbc40a616e9089da : TIMED_OUT in 2h 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44698587b561d1abbe79bb340182786b2e6f24ba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-06 13:00:02.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8540c78eff43525e8d109677974d3d4f50d8e4b2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-06 13:59:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d545ab1cedad6e2f11966dc7477fbcf87610879a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:42:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c62689160954d06a5f1de58fdf10b5e : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c25d9309bb54bdb81a8cf10b5ff6eb7 : FAILURE in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f86b9350329430ea724be135715db04 : FAILURE in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7994f5d0c2b4296976ee3badd87e298 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22537e28da7047aa80f9c09abaf1e148 : FAILURE in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb45fb3e538d4bc48eee7c749bbb8c1a : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ed93ff2ba14123a00d6abae13d2ab6 : FAILURE in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50eb01fd44640b284a94e207938beef : SUCCESS in 7m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/695de1bd964447e1926a858ce3ffb2cd : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f174b92cf4284cae9a13a8d4a4447cd4 : SUCCESS in 9m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8150061590f44e0b47f347a13ac690b : FAILURE in 5m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7dabc0276c9e41888827315c2afdb891 : FAILURE in 43m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ab09447d46445bbb8bf3d652f1f7360 : TIMED_OUT in 2h 34m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd93f74c833149f688a7aaca2558c6dd : FAILURE in 1h 10m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/164a54a5e3014af78256e73c46433b0c : FAILURE in 53m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32336d0f9cc7410b90bfaf529a7426d7 : FAILURE in 48m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34e62088c0a44d6ca5ac706b786106c5 : FAILURE in 1h 11m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27ea8983c07a4c99a9d72080fd15e900 : TIMED_OUT in 2h 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/421b144d86b24f70adb3d8446bd6e56e : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/19e1e0d06bdd4307a4ab076ed3b68de9 : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d3b47fe98714a6ea80e4b69eb861add : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f26f678c23d64765b245d5e4e90ea475 : TIMED_OUT in 2h 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0833117243e769fb4f95edc06f7c0edaa6b3abef","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-10 09:17:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8836074fa6c195aa8c5509d30f954d0f9d232481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 10:42:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c51b0f6f86124652831645ae6199fc97 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f00c10c545294b09b1dc4d9715d5df11 : FAILURE in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97d43560ce7d47de894c7484cbf2d713 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9267e918516e4419b86f846dd871d5ce : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c8c6f728ce344379af164e273aca050 : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/557d449c29c642a9bfff824e1915aee9 : FAILURE in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90003d9c38644fcfa58841d3f8b39184 : FAILURE in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6236808db35e47ffa0fdd92c9a68cdec : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1cd80cba90a74528b910de767c8a2e32 : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d6f8994728dc4473b946d9da6f90ae4b : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d232a195e5a549b78ade38e28ac0a855 : FAILURE in 10m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd141807c56a4582b5f2b6893abb589f : FAILURE in 26m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e279b7cd47974f079ffc9484e806d29d : FAILURE in 38m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/651c1842b82b4d5d8a567be2683f7c86 : FAILURE in 42m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2312aaadf4be4d6fbbf9cdaae6f6e5da : FAILURE in 38m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e28513427c2643b6aa3797f2b7b692bc : FAILURE in 39m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a9846a09a2f4876b429c4403d14f3bc : FAILURE in 1h 20m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a74f01550deb45a1ac5dd908850e5f74 : FAILURE in 41m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c78da0f1a134baa877a118ad31049f9 : FAILURE in 27m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a4238d39d524d699fabf75b76ee7706 : FAILURE in 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/49bb9ce30d9e426a8ac1cb240a428dbf : FAILURE in 20m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/202a52c33f184500b17fb8dfa71caa8b : FAILURE in 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cdc47dea1d37e4e95c4e12642a9e2daf1f1ce79c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-10 17:13:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3b8a06fe9e82de029b54b4a60ad8c77c4bb3d1d2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-10 17:34:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"661fc41b87db568351ae7db48598aff04caf1dec","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-10 19:46:26.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"62957bd366976d0c17941e7cf4a02966d66b0f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 22:26:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6bf913b32d3447d3b5ceef1f2634b829 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa1ef9246459494c989377c71db6556f : FAILURE in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aae3ecff3704b238b0b158f2534b65b : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf11d2c3715545c1989d70c66c205916 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cabeef105ff4f17b69502871dee0aed : FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20f3834847b64b8e8a25023e5d74f24d : FAILURE in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0755ed9147fe4ed8b91c09d855fe452c : FAILURE in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52294280e46e4c8187da20e3c06836bf : SUCCESS in 6m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/439bebdf3d9547419a35664211aa3d40 : FAILURE in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99a3407e8dc14a91bff23b05f9149b7d : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b4755c412f142018ea3b5a49df636f7 : FAILURE in 4m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e91e1db59cee41dca4c20a98780c8876 : SUCCESS in 44m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39155b407dc246858bf3cac2b3e1756b : SUCCESS in 59m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7eeb86b4c47471a90bbb191b4182918 : SUCCESS in 40m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/26cf4f9597d648ca8e1847b3fb2ece9c : SUCCESS in 43m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e2dacb5358b437e9e2ff552cce14a3e : SUCCESS in 1h 24m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/22cd8de315ca4cc39228415a0650a3c2 : SUCCESS in 2h 01m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2477a95100fe4e4588f8ecd7f4401e2c : SUCCESS in 1h 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7827cb8be3dc44b7a7f33cd19cadb670 : SUCCESS in 1h 13m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1deb9aa6715f4ffb968e37194397b3d8 : SUCCESS in 50m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d7aec5b76e04e9c99d1a687273b18db : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1c2dbae88fa749fca60fb41a8e7875aa : FAILURE in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e468b31013d0168b04c61145a7c192c70715a131","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-11 15:21:42.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"60495c2c4564cf72c8462f1179d0893d38d36500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:39:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ab1e140cf85b4308860a6e13d2aa0f1b : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf1737c74aad4ced90ef7ef4f9d37536 : FAILURE in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cce0db00fcce43f9bab1cf2ca5574655 : FAILURE in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e1cbe40533465cbc845951f31afe7f : SUCCESS in 10m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef0205af7be43f89a53b02d290cedf0 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9abf4e7053549c19b218ebe0481c846 : FAILURE in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442f9c04e64246b9a1f13382ce9515a8 : FAILURE in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f735a304c844490ba5acada721b9316 : SUCCESS in 5m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5bf3dffcef754b83a400e0d10617d51d : FAILURE in 10m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bfb068edbd3f45aeb19fc801051773f8 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df9019bf92664ef29bbacfa610f6179c : FAILURE in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fb7468998af4f0980d516052cded672 : SUCCESS in 47m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05906163d3ec4acf8aa05da7c845744f : SUCCESS in 43m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de3a60091c4946469c5d522c9daf93b1 : SUCCESS in 35m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f7a2da6b5507481e9888dfea170d88e5 : SUCCESS in 31m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a738cbbf15284ea4ae39b78489ddd2d1 : SUCCESS in 31m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/579570d621854921910bf34ca3a0ad48 : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/413e5f3f8e7141df8119ad32e3894d89 : SUCCESS in 44m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af601441679e4798a658f87e9a20bb11 : SUCCESS in 48m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71f21cf6c7b84a7290e0c3c1705df1bc : SUCCESS in 57m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1e2ce1ea86b469e8b2e5f27e1086ed7 : FAILURE in 2h 25m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9500fe62c49f48d18a4fea1e39ab251f : SUCCESS in 1h 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bcd679d578733332ec776e37d9d86226286e27b1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-18 19:05:40.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"6000dd8ccec30698218e2fa8c94ec54af8e27af4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 20:49:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9bf5ae7c0ca94d339f96d12bf1dde2e8 : SUCCESS in 9m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edffa185faa74608900470ab8568d89c : FAILURE in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43e5574eeb7545bd9f7a844395f37027 : FAILURE in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82f9eb752716467c9ef75c6842cddd37 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a115fa0d1dc24911af104040aff9e825 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3409dcf0b384858b9537efbde6056a0 : FAILURE in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ecb5334efbe4f6c9a02b62b11968340 : FAILURE in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0c8185525744448937e0de4e0736367 : SUCCESS in 5m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/980221ed71b34d90addcb648d239c79a : FAILURE in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f99a01d002ec433d9f2d936e03c69e93 : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce15d04d24c44fa38a45061291bebbf5 : FAILURE in 4m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34a6459041164780b6fa676b4460d566 : SUCCESS in 49m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5692f25fd205441192ee38a602e6f58a : SUCCESS in 52m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f62d0a13e584daeb04be9e6bfa965af : SUCCESS in 1h 14m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12d4d68c77e3405398f4bcecbac60e03 : SUCCESS in 38m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e3e86446ec14cb68369f84cc0a998d3 : SUCCESS in 1h 11m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b680a37a0aac425a969c5e72d6e33ca6 : SUCCESS in 1h 02m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c40e18b1f37146adb92547573e076b74 : SUCCESS in 1h 41m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/85a8fadc709b4f09964b7fbddad70c8d : SUCCESS in 1h 03m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d95330ac8ec4e53ba1b6fff4bee9f70 : SUCCESS in 53m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3bd5c24a572f4bd4ad5196d7675873c1 : SUCCESS in 1h 06m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b4ae34217eff4bbca3e11bc66c13a2ef : FAILURE in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cb3bf9e338ccdd335b8e97559cb51a8f484f00e7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 09:31:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ec399573a95f8a93b0120114760e97b56303ff45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 11:24:12.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cff8b75605f1464899176a2c5bb7b4ac : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/135009b701fd4ad788faf13a4f1592f3 : FAILURE in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f0025e90a9447b1b94db7d6e0eb6c1f : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71aea24d71f64220849c1eff3cbe29fb : FAILURE in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7638b6391d24d399f64bd5f1e2f8c8f : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a709895028463193e965443c55a314 : FAILURE in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8086cbbaeb4c9880946776e17e5bd0 : FAILURE in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5de61312c704f9ba597215338ad56e2 : SUCCESS in 5m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b8600f97f43c4727917fa2f12b6d7476 : FAILURE in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0147ec94ce114bafb6776f9ef5dc870c : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72239835bf1148c28f645d6e73e0aa83 : FAILURE in 4m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc4fd308dc344cb8a6bcb3f2203dee06 : SUCCESS in 41m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6adf8ecd24ae4fab9f8d6e5c441f2055 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/838fc5ecb48d4da79b8ddea326a58edd : SUCCESS in 32m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ceafdb5cdb6649b0baf858cdaf2a1c09 : SUCCESS in 50m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8a3b452aadd479dafaba11c47ef07e1 : SUCCESS in 34m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5dbbcb74f0cf48159c782e7f5f6d2bb1 : SUCCESS in 1h 00m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b9de08985b6b428aa0b216a7f42037fe : SUCCESS in 56m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25339dd418024f00866bdfd5401bb037 : SUCCESS in 1h 26m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ecec4ea09f44c5fa3849b8b3fc21251 : FAILURE in 40m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b8bf1c48c50f409c8c7eb175d6d50600 : SUCCESS in 1h 11m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/015e1de7c82e4a0dbaed3ef126f4a5bf : SUCCESS in 1h 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9b711fb74b74c5d62f2a56fde713e3a1adfc60a8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-27 11:49:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8d0a14f5981ac3677a1ee76f92f4c06914a17bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:32:32.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c79aae9e78f14b74810e9310f9405cfb : SUCCESS in 5m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff8ad7d064b041bab4311bea97e03c22 : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aeced8d97c34011aac6fe609dbd96fb : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/152692fb60e54621bcdadefded469fe7 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7c165b005d4966b46ebeec82a748c9 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0142c2c8d7431780c6d002d5e2af4c : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d063bd089e5f45a78b74a52f83c67e0e : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e5d6580e59453bb748a5b4508ca212 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01e5444dfedd4e2e9a241c64c0096fcc : SUCCESS in 6m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/133f0077f2cd4f5eb4b557185e03dd14 : SUCCESS in 7m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d47ddc3618b645cbb79aefef678ec108 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/efb24b6a36a04a75bca212932a6a4183 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60b9734cf09940fe9ddc3a66247899c1 : SUCCESS in 49m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/924b7bb2eb4140c3be035abfd821cc53 : TIMED_OUT in 2h 34m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b800f3ee5334a0087406b20789ab2fa : FAILURE in 12m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3de2d627d7824f869a71e8833602d58b : FAILURE in 1h 08m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e91ac02206e248fc8072ea29b41faab3 : FAILURE in 1h 12m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c0e4c90c6a246baa952ad13d57f1d41 : SUCCESS in 1h 48m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3721378ac5474ef9854a016faa76d64b : TIMED_OUT in 2h 35m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3fd190d8896e49c9923ceae390cfd7d0 : FAILURE in 14m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1e436d118e54ea4a4595f17d67c8286 : FAILURE in 11m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3bcec953069f4dd28d96f23996c621b9 : TIMED_OUT in 2h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/97a51e332a89434dae287447e5f64ee8 : FAILURE in 11m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/849900cdea5d4c1cbe1f8a31ffc3a5a5 : TIMED_OUT in 2h 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"feb11ea6c9aff630418266bd376a02e8511ea061","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-28 10:13:18.000000000","message":"Patch Set 11:\n\nrecheck\n\noctavia-lib 2.2.0 released and in u-c.txt","accounts_in_message":[],"_revision_number":11},{"id":"e9024a462e0b8b1c4d8daedc62ce7b9656cb320a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:16:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7563cb33c43a4d428f1ee033629801d0 : SUCCESS in 7m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6c190f9070443fe8c773f1a6b434586 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f1dfa44602246108f6c2e335b7f03cf : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cdd755f9b484e4ebb2d7841d9579df4 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2926e9ca7fbf41fcb30fff9f64aa9f45 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edbdb7deffc044cf818c5e07ec8410bf : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c337aa6a92f4fa89e0e06c3985c6866 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f4c68e0bea4742916be0c3f6e60926 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/701d6f9701914d6a978b59f3d64b90a8 : SUCCESS in 7m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a93fff61a6d34b10b83f57367fefbdf8 : SUCCESS in 6m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/474968aa0b0544059d76c11e97908e11 : SUCCESS in 8m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bde0c3c7f294df5aed4c62cd4d770ab : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae27a8db5b4a4cd3bddabf600734f8cd : SUCCESS in 1h 06m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24c982fa48df454ca641d73819bdeb9f : SUCCESS in 1h 11m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bbeaaac3708a47a9b7240aaa150853b7 : SUCCESS in 34m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/86897bc34b01421dac9906d51531299b : SUCCESS in 57m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83ef2cfe30ab412284e37d0883e47606 : SUCCESS in 44m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/590f8a2c20b24cbd9791ff59d05f51fd : SUCCESS in 1h 57m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/409f1c83c2ce44a6ab951e119bfa4f62 : SUCCESS in 1h 03m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be177b04453b4210b65fcd8bc3fdfd0f : SUCCESS in 1h 02m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/486c111770cb494ea52089c74b3e019b : SUCCESS in 1h 08m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ca1b69e23c4949c59cf5b1584174c30d : SUCCESS in 57m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9778239b75c348a4bfde4af653765924 : SUCCESS in 1h 38m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/002b110466d94bd5875129e8a262d421 : SUCCESS in 1h 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4e1512b195b22fe031fe3c0a5e6de19dc73e2e31","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-28 16:21:28.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"0e48b8b7cb07d6fecfdeb4c937d3fb5f1ba54cd2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-28 19:30:45.000000000","message":"Patch Set 11: Code-Review+1\n\nGuess this also fixes the cover job failures?","accounts_in_message":[],"_revision_number":11},{"id":"795742e7bdb242ce2790dea0831a0220d57d7fc7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-31 07:35:51.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"cb0ec2f54f992cbf1d72b3cda3ff40bc7632f04b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 07:36:03.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"e235d86061cf0e7442946071f2031c5565817fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 07:51:53.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c2f92dcaafb24c7aae4739e661cde6dc : SUCCESS in 6m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/649441b81f044e04912c843cce2cb5e2 : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c5bfcc8290242f7a36640841abb08fb : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2371457e834defb46642e78440e5c2 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43f48710928b4dea830c8055d0495861 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab5f898935974613abc5b80cd83ced3a : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a3da03c2ff5423d8afa72e3c17d8b92 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f9ac136042f4bb688ae4d508d7a5999 : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8691137b7d9647a1928a596e863fd167 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1d908df07644607a74f63d7f6bcb1c8 : FAILURE in 14m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e18692d7679349e78a20464ef1dfbb8a : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c71e0c614554193b42578c4efaa9c59 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6ccf576cfbb34b7192276acaf285338d : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5927e4f77fe44bb987b6bb62ba9f7a6c : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/561a871c8581432ab2ac8e0eb52a7ac0 : CANCELED","accounts_in_message":[],"_revision_number":11},{"id":"885d0eeeb29e3bed3aee2009eb8d9cc48ff99528","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-31 08:35:48.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c7a31fe5da6b6d1c8c993186c2f0f38e0eddd2d8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-31 08:37:38.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"4c82c87271b3def67096f3794b964fe7d0f7d648","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-31 08:37:48.000000000","message":"Patch Set 11: Backport-Candidate-2","accounts_in_message":[],"_revision_number":11},{"id":"7aa9a100705c67edc0a023caf467c709f572cee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 09:53:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e54b9ad1d724538952ef00f29cc0657 : SUCCESS in 9m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61ca9f6a944b4bd7a67b390e972fdb0d : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7ab3381fc0544c18032272cf856d2a1 : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1eb1961a62e4ca5b19d6a1d5a6daff3 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072ee1fd1e1947eabb5f43f5603b5491 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c3b47f4694f48ea80484079c26d478b : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbebfa4be4ab424ab43ac8c706c6965c : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343e8eea4a70418aa3d946856c624b9a : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad938edea3ad4f4495de93a5445be02c : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/211acf43de904b7da5f939a9e0768535 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9d4888c844714206a49b1812658edf31 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/921b4d5c5c8149c08dc4fa026b3358c4 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef5ec8d6714548b9b247008bfeb8d2b8 : FAILURE in 15m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29b2133dd0c14540a7422285d67a16b1 : FAILURE in 39m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0042bb695cb8414aaccdf3d3a6f622dc : FAILURE in 42m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64c9450637e9449f9147d94296d27843 : FAILURE in 38m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc8d1403b0854f6f8048a14739cf9752 : FAILURE in 16m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4117de5ab424cd78142718d55492806 : FAILURE in 18m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea942a29d4174b7686ca9b3b59c6de00 : FAILURE in 16m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de9d055f314446b1bf4bfd600bf61171 : FAILURE in 17m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a4c7a6e58b943d198a49443eb6b1770 : FAILURE in 15m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fec6efaa5c5d4ce695924e853e351039 : FAILURE in 16m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/863be5c93909414d9ac433dd8d053ff3 : FAILURE in 15m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/df3ebac61ea84e0f9ee45d15afb6d1e5 : SUCCESS in 1h 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a64b1146386af21ddb0f45c13c66a8ac61870c89","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 05:11:43.000000000","message":"Patch Set 11:\n\nrecheck\ndevstack should be fixed now","accounts_in_message":[],"_revision_number":11},{"id":"4b87d1ea0152ed313a0ec03a53085cea8dace01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 07:52:16.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4864be247bfa4e0b805e72bcb616e83c : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5183048e2f864aaeba2c98ef82868a43 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/321bbed448504033834fe9c6b32c64b6 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e2a24c023b24d838d1743fe4c466af3 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81b2738962a42448ff8fa39c1411f17 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1d95ff0eb1490d9be78c708ed7d082 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36f0d6e6654147c79d5fb1a056220563 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7599079bf54a2887c4e9e958884b9a : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b1c8a2e494e4f1c90053744b3035a4f : SUCCESS in 16m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a86e452e75db4002a535e75b97854905 : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2accfb7852b74712a07376dbc62bf92e : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fadf65c7a9c84408a0f2401aed191f38 : SUCCESS in 12m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16bc97509f544a1b80b1a454c62325e7 : SUCCESS in 43m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1055527d956480b8bd861b77d37afc8 : TIMED_OUT in 2h 35m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b9105e2591d426698d9698707c0b364 : FAILURE in 2h 13m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f65e04b9b90d48ada90bd0271b50a2c9 : SUCCESS in 2h 13m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e24615063f5541928f5dd3666221da09 : FAILURE in 2h 09m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98b1eb5cd4b1416e80745c2b008bb747 : FAILURE in 1h 22m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bf88233b79a94f3d8e42b3256943e42d : TIMED_OUT in 2h 36m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d84bc8c80284823b329879a2dfdc7d8 : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9ff8b80f9efc4e6bb5aa8af97ca9ebbb : TIMED_OUT in 2h 34m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6940659aa5cb4596a077e403bd1eb5ae : TIMED_OUT in 2h 34m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e4eaa4beb57e493ebf44619ea9f89cd6 : TIMED_OUT in 2h 34m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a5a2a5ed744a41ab9dceec1913983507 : TIMED_OUT in 2h 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"50f9f2034d10804227686699aace2a1541062552","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-01 08:01:42.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"61a8dfa977e434baf7df60b625c710cef5a57466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:44:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f0295b31685e43e99f06fc4477283a8f : SUCCESS in 6m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b9a519dc35a3446ba9480912f7b007a9 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16399b051cfb48f9badd837be06d453e : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c76df3f7f1c4850950cb66283eb81bf : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8384559b334da3b85bf54d4f85aca1 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dc7986c0d584a12b93b6242b5c72c60 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c005fa5113a2493aa21937798dd387af : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3047e1e3eb1e433f89bd48e0f88ba7e7 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02c8b7cc47ee4ca3bb5e6e231a48452d : SUCCESS in 9m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0a42a13d608c4ca4af158964da61d30f : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/085996ce356546d3a100b324136838b6 : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/606f67a92e5d45eba963387ec7a279ca : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f63ff62d0e643439c2929c7b90aa756 : SUCCESS in 56m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a5a1a994e814586a98de4df3e0ee6b2 : SUCCESS in 45m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/beb55129d8e544d3b4d7b83ae527cf92 : SUCCESS in 34m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/de7065214d4843ae83682c449d257ff7 : SUCCESS in 54m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/885e7427f06d4c05ae1055f4e2c0256f : SUCCESS in 56m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1052e0d2e42248398803ef883b49e0ee : SUCCESS in 1h 32m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1cfa6c63349042c4ac74d0841d55ce5b : SUCCESS in 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e1965643b694974a9ce80c2baf61461 : SUCCESS in 1h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/90c95c8bf0a9435fbc3ee8a233774334 : SUCCESS in 46m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e304b43c85524c5294664b1162c6bb1c : SUCCESS in 51m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f9bf97e62e534a4fba4c251ab1960a5c : SUCCESS in 1h 15m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7ba1a4cd1ad54fc2aacc82bbab71666e : SUCCESS in 1h 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5c793a2c4cb54bad7fdb0d9112ab6d5fc2eb27c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:44:20.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"6a6d2e1ef5e5658b4b719024e75e9c9ff43f3993","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 11:42:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"cd672f6ff0f9ff05bc0cf1172895d0da0e205565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 11:42:16.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23cd7db14b2a4894b81e820f3ac4deda : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/575e1e5a93344284bcc61dec6e0bbcc7 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c158b818da2342faa266c6a2d95819a8 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354e081d49de419785c2b34aa66603cd : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5929758d2df3457d92980575cad88761 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e906dbba4716475c849f85ebb73d7e0c : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401a001561d248f0aff0ee635757e5b7 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/414221a38f9a4e6baaabc9464993327c : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/624a66fb62f645dc946e965dcf329e79 : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ad3fad983804cbbaac08d9e259df263 : SUCCESS in 41m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22b52171ab2e41d69d96a8eee430dae5 : SUCCESS in 1h 00m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd6963b13c0541d5996ad086977cd2bb : SUCCESS in 41m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/862d84a7f0cb457394f5f3a46a75a6d3 : SUCCESS in 40m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26cba58b17474bc29520aa432ba485f1 : SUCCESS in 32m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eac5138439d44412859979c7e95a7fe7 : SUCCESS in 1h 54m 00s","accounts_in_message":[],"_revision_number":11},{"id":"3ff79bcfb70484acb67438d9cd73cda919445c3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 11:43:54.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d09358779002417da8260d93aa2578e4 : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf45ae0051b4e0cb7d40aec6aa3728d : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f56e292f32d4c9caa20fed623671508 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"a5f0524fd00b1fdb9d47bafba81155eeb769f3ad","revisions":{"11d7b84ff23312e384d41ff8bbec8974246e6773":{"kind":"REWORK","_number":1,"created":"2020-08-03 16:13:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/1"}}},"commit":{"parents":[{"commit":"2f220368c5e7237dd1884b2bb7eea09454e38fa3","subject":"Add line terminator to certificate list file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f220368c5e7237dd1884b2bb7eea09454e38fa3"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 16:13:13.000000000","tz":120},"subject":"Add HTTP/2 support for TLS-terminated listeners","message":"Add HTTP/2 support for TLS-terminated listeners\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11d7b84ff23312e384d41ff8bbec8974246e6773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11d7b84ff23312e384d41ff8bbec8974246e6773"}]},"branch":"refs/heads/master"},"fc973595e54d37ce95141ec98b8c7d1faf57cd45":{"kind":"REWORK","_number":2,"created":"2020-08-05 13:39:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/2"}}},"commit":{"parents":[{"commit":"67751ec1e1b52020dc8db6414f276d791b57791e","subject":"Add line terminator to certificate list file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67751ec1e1b52020dc8db6414f276d791b57791e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 13:25:30.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc973595e54d37ce95141ec98b8c7d1faf57cd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc973595e54d37ce95141ec98b8c7d1faf57cd45"}]},"branch":"refs/heads/master"},"ef19c2a4027c0ee07dfbc5fec2fc855272b04a68":{"kind":"REWORK","_number":3,"created":"2020-08-06 13:59:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/3"}}},"commit":{"parents":[{"commit":"67751ec1e1b52020dc8db6414f276d791b57791e","subject":"Add line terminator to certificate list file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67751ec1e1b52020dc8db6414f276d791b57791e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-06 13:58:25.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef19c2a4027c0ee07dfbc5fec2fc855272b04a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef19c2a4027c0ee07dfbc5fec2fc855272b04a68"}]},"branch":"refs/heads/master"},"24725ec03fcb3bb57340556e0a5bff218a01cdc1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-10 09:17:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/4"}}},"commit":{"parents":[{"commit":"8c292e7219f67f23edcfd631102f4155d0d5c5cb","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c292e7219f67f23edcfd631102f4155d0d5c5cb"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-10 09:16:02.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24725ec03fcb3bb57340556e0a5bff218a01cdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24725ec03fcb3bb57340556e0a5bff218a01cdc1"}]},"branch":"refs/heads/master"},"26685b66a9690c254a2871764dd3a875b4551d9b":{"kind":"REWORK","_number":5,"created":"2020-08-10 17:13:37.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/5"}}},"commit":{"parents":[{"commit":"470940db7f8449cf06695c6372b671898fffc38b","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/470940db7f8449cf06695c6372b671898fffc38b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-10 16:59:32.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26685b66a9690c254a2871764dd3a875b4551d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26685b66a9690c254a2871764dd3a875b4551d9b"}]},"branch":"refs/heads/master"},"deef866ace85f5906ddfe4aa85198cf906e9bb16":{"kind":"REWORK","_number":6,"created":"2020-08-10 17:34:49.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/6"}}},"commit":{"parents":[{"commit":"470940db7f8449cf06695c6372b671898fffc38b","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/470940db7f8449cf06695c6372b671898fffc38b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-10 17:34:28.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deef866ace85f5906ddfe4aa85198cf906e9bb16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deef866ace85f5906ddfe4aa85198cf906e9bb16"}]},"branch":"refs/heads/master"},"d5e720bd41a37a36016b730dff9858d5c13c2916":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-10 19:46:26.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/7"}}},"commit":{"parents":[{"commit":"7ba67b3ec6fd1a22a92fe52a67813134018bd25d","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ba67b3ec6fd1a22a92fe52a67813134018bd25d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-10 19:45:46.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5e720bd41a37a36016b730dff9858d5c13c2916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5e720bd41a37a36016b730dff9858d5c13c2916"}]},"branch":"refs/heads/master"},"321472348cb05f6166641766835580f555ca7f4a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-08-11 15:21:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/8"}}},"commit":{"parents":[{"commit":"326ddb5f2db59162aee53c516703191e7650cb78","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/326ddb5f2db59162aee53c516703191e7650cb78"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-11 15:21:27.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/321472348cb05f6166641766835580f555ca7f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/321472348cb05f6166641766835580f555ca7f4a"}]},"branch":"refs/heads/master"},"8016651affc087fc90c03b30fc0087a791b8ba67":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-18 19:05:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/9"}}},"commit":{"parents":[{"commit":"bdf133bf5c2eff03280ad6d3c674e52b5cdbc9c4","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdf133bf5c2eff03280ad6d3c674e52b5cdbc9c4"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 19:04:37.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8016651affc087fc90c03b30fc0087a791b8ba67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8016651affc087fc90c03b30fc0087a791b8ba67"}]},"branch":"refs/heads/master"},"dba52766e9c3e52a51cef4005bcac1136766227c":{"kind":"REWORK","_number":10,"created":"2020-08-24 09:31:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/10"}}},"commit":{"parents":[{"commit":"dcda721320986627cfee8b6a0f759585e62ad799","subject":"Add DIB element octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcda721320986627cfee8b6a0f759585e62ad799"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 09:27:42.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nDepends-On: https://review.opendev.org/#/c/744519/\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dba52766e9c3e52a51cef4005bcac1136766227c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dba52766e9c3e52a51cef4005bcac1136766227c"}]},"branch":"refs/heads/master"},"a5f0524fd00b1fdb9d47bafba81155eeb769f3ad":{"kind":"REWORK","_number":11,"created":"2020-08-27 11:49:17.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/744520/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/744520/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/744520/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/744520/11"}}},"commit":{"parents":[{"commit":"0b1d8dd5e76edc4b3058a36b47f6c059f970e516","subject":"Merge \"Update devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1d8dd5e76edc4b3058a36b47f6c059f970e516"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-03 15:48:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-27 11:19:52.000000000","tz":120},"subject":"Add ALPN support for TLS-terminated HTTPS LBs","message":"Add ALPN support for TLS-terminated HTTPS LBs\n\nALPN is a TLS extension for application-layer protocol negotiation\nwithin the TLS handshake [1].\n\nThis patch extends the Listener API to include a new \u0027alpn_protocols\u0027\nparameter. With this parameter, users can set an ALPN preference list\n(descending order of preference).\n\nPresently, the amphora provider driver is limited to http/1.0 and\nhttp/1.1 ALPN protocol IDs. Support for \"h2\" (HTTP/2 over TLS) depends\non HAProxy 2.0 or newer.\n\n[1] https://tools.ietf.org/html/rfc7301\n\nChange-Id: If08a8169498cdfaa75440e8971ba0caff45ac4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f0524fd00b1fdb9d47bafba81155eeb769f3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f0524fd00b1fdb9d47bafba81155eeb769f3ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
