)]}'
{"id":"openstack%2Foctavia~744905","triplet_id":"openstack%2Foctavia~master~Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62","project":"openstack/octavia","branch":"master","topic":"http2","hashtags":[],"change_id":"Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62","subject":"Add HTTP/2 over TLS support to amphora provider","status":"MERGED","created":"2020-08-05 13:39:38.000000000","updated":"2020-09-07 15:02:54.000000000","submitted":"2020-09-07 15:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744905-1599490807935-1545f216","meta_rev_id":"2ffb6290bccdb76b4cf8bdf62e12444cbf1682dd","_number":744905,"virtual_id_number":744905,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 15:00:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-04 08:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-03 16:05:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-07 13:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-07 13:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-2,"date":"2020-09-07 13:57:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 16:05:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-04 08:34:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-07 13:31:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-07 15:00:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a58c9603181479a0e77df2e7d5e242f090cfa02c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-05 13:39:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b13cd170d796c320ac99597c5268e9cb5861b732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:43:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e44855a8d1954a69adcbdb67e2343ed2 : FAILURE in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a6859a104414a34ba99db21c92c1c89 : FAILURE in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963b72475ce24da2bd7a36ec852149c6 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02291ae538134512b2d4ab4bafb3728c : FAILURE in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d646ff81a6f249eb86747178f417cadb : FAILURE in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93450bcf9cd4c4289a2aecb35e665af : FAILURE in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46d735cd8e84456c8378f2c2a012e060 : SUCCESS in 6m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aff081e0447a4df8a82aefa0dffde9d8 : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/78a9d577c2ea471d8353de9bd72134c0 : SUCCESS in 10m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21194064c55040cf957e6603acfad7d0 : FAILURE in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/986aafab0bc94e86924fc9d6a4255ee3 : FAILURE in 43m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1b9983579554e9f8d63a1270f06984e : FAILURE in 2h 31m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b54250d0f8394ae9a98f5c07dad4ca63 : FAILURE in 1h 05m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e3fb0ae380fb4cd49a9131e3409a3620 : FAILURE in 55m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df8f879a91b040e48cf2cdfd4000705d : FAILURE in 1h 00m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49f7fdb166e94a75a93703ddf4f17655 : FAILURE in 1h 07m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fc6108c98a0e4450b602f68621d964df : FAILURE in 2h 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0018bcf1c34d456e8adac7845fe4113e : TIMED_OUT in 2h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/126d5ecb60294a3fa30ef559c7296cc3 : TIMED_OUT in 2h 36m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/edb7fdbdd78944fa8e39df5eff62c36a : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8c2e138653744b4990d32cc30c80a6b4 : TIMED_OUT in 2h 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d02b4b97da788e3e0066f0400e8d181ce4152c3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-10 09:17:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ba1438ac5b7739945a6b279e90b3456f7fe096a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 10:25:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eec4ca9c109e407985989020b63ced62 : FAILURE in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d9ddc4854384b27b02950eae8eb042f : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c4ddba8b10486e9a97f30fa4d9d283 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e90bbd790e64fbf905729364f2822d9 : FAILURE in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a634e47bd0347998ea5cfc967e0de16 : FAILURE in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b8232686bd43fbb967ffe9be6150d5 : FAILURE in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/417b51b4c078410181c86ea6dd34264b : SUCCESS in 4m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9fc75839b9254ccb9a27f7484fd1345e : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c4f5944e9a864badbe4b23644efd50c9 : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e391c3a996fd42ddb0623082b26a9976 : FAILURE in 4m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/262fc376301049f0bd66c0a6152c7c4f : FAILURE in 41m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/562551189e0440f0a2c99ccf124b9cfa : FAILURE in 29m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c989d4e6ad2b4f8aad468b74773dfaa8 : FAILURE in 48m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/89abc183eefe4aebb9c792219906f00d : FAILURE in 43m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ab1d3affed64449a0849e7311f5dc76 : FAILURE in 44m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35a28d06ff46480f908f3acea7a4a57b : FAILURE in 1h 01m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/15499cecf94342cd90a52496342c0c08 : FAILURE in 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/55d4a117531d4693a237f56f7306966d : FAILURE in 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f26dd2dcea3a4e09b697916f765fefd6 : FAILURE in 36m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5cd75df9dbfb4074955da959d02c3638 : FAILURE in 30m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/913e9297e4db44698752bcfa4f5d446d : FAILURE in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"382698ed4fcb5b920d189da76b1a4a790b905147","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-18 19:05:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"810a1b2347e9f294557f38df1f7e28f79e046662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 20:04:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d997ef49e5d41f19ff970446cb1bc2a : FAILURE in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/141d5ecb2d3644edb0292447b6677899 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed21b8b0b3914da791323b25a03a4dc6 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61502ade63d5465a9e17d9a6c24a40a3 : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cd11cb7aaed4a989ed99c2020a0924b : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badbdb01b3b14d298a16795a176c31e4 : FAILURE in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ccce785d81e408c8cfa4a30cd7fe37d : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3797cea58c35444790f63ebb7d2513d7 : FAILURE in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/290092d22f364be08b8bab9627e3c1c3 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17ea14f74da94415a53299c7736af5e6 : FAILURE in 4m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c57adbb400d41549de6c2f4ce68db97 : SUCCESS in 50m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d509d71b06774e3a8b64019e1acec81d : SUCCESS in 39m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/426d90e5d8d34dc5bd3daec86c69b207 : SUCCESS in 31m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f69c8814afc04d199c41e5f7bebc7192 : SUCCESS in 28m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/152c8017a89542f4bb96954c182b15b5 : SUCCESS in 29m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97ac6c9d0e0242bc9b8a034172db37fe : SUCCESS in 56m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ab812abf7e0413b98d3b8050769f414 : SUCCESS in 44m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aeb1cd9668084553a92d63c0b1091678 : SUCCESS in 48m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a0c4412075414b37bc123238b9360d6d : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8b4f269244c3408da2f3ebe872f914e3 : SUCCESS in 47m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/212dcdc4589246caa672a52820b2f936 : FAILURE in 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa02379da217d1d5753c95c988ac0f7f4c259e82","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 14:01:26.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c1b37f35a6946a7c71b696eee941d71054414abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:52:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ee624d4b91e441c90af56f9fdecd50c : FAILURE in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb23860b310c4f4a9e33fc3963a2a4b2 : FAILURE in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd98135f45394883a609c797a6b71337 : FAILURE in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be909a433dc4792831240ce356905ae : FAILURE in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5728a5ed95ef423da662af41f7110929 : FAILURE in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec6703cb07f4076a7fcdafcac3bbf84 : FAILURE in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcb64e4f33e44afaac175e4f5940e3a7 : SUCCESS in 9m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b2d44f0133041149c81cc3aa0f6184a : FAILURE in 9m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d1017c7b3e864380999e116c1d5e6d48 : SUCCESS in 9m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/152d3e1876824bd8ba50f19261fc1579 : FAILURE in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ce57899adccd42799bb240014ec60277 : SUCCESS in 1h 02m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4595a94b7e4c44b1a2fb060d07ed9976 : SUCCESS in 1h 21m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0ab5772b9304fc7a4114ab6120fe496 : SUCCESS in 1h 09m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cd76ecd1efd44e698cd99f3087c172ca : SUCCESS in 1h 03m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f69ac2f98114eefa83e5a184d7a8b8d : SUCCESS in 49m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab472a1d680f40b7b02681fc2f45cb5f : SUCCESS in 1h 41m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0dca5cc7a4d149a28676b722bd4a5d3d : SUCCESS in 1h 35m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/18ac6fdfe7da42dea1c2d0f4db951a5f : SUCCESS in 1h 08m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e07ce394e6724f3bb56fc6a0d03f0a9f : SUCCESS in 55m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/07ba5dc87d4744b8adb00eee37cf534e : SUCCESS in 55m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0c32453264ab4c4294a1aa41cd58dd53 : SUCCESS in 1h 05m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py","accounts_in_message":[],"_revision_number":4},{"id":"c902a6a030943a189d121d00c94e9a288fdb312f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-27 11:49:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6f2cf79f933d8d0384e9765ef9b5110cb3d41df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:38:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/264f6d4af1ab4b74bab38d02ec8911de : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1776aee648804f98bcb92e368cab57d9 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70affcbdb3e4db1b63078e03f70c985 : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57e10323c92145bda387ac8410fcac1e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04c8d459bc3441d19f57f0c38619ec03 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d356d92d524a3990550aea689413b6 : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86cf1aca4c964d6c946e1aff3c9dc824 : SUCCESS in 6m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/005bbdd475f043e8841b4d7f321a5d39 : SUCCESS in 7m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/30345d06568245ea977925d822cc8c21 : SUCCESS in 9m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d940b59de83435c8e80d43313ba557c : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc8982ee336e40839da69e1c822fde9a : SUCCESS in 50m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cfebab2f5384d5581fd58ee9f8ee243 : FAILURE in 2h 19m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2abce8114154c1489c23aa19193e192 : FAILURE in 1h 01m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3fea0d7a033740b1bfd430e7a46d79a8 : FAILURE in 1h 02m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8b4b819d50e49fc91da6f561ae50d52 : FAILURE in 1h 09m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f4b4217333184cbf85b54566c62029d0 : SUCCESS in 1h 52m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05c5c3e488e947fdb3ef58e123448b6b : FAILURE in 2h 26m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73e5f5998d4844eca8b5e5231648cf98 : TIMED_OUT in 2h 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5dfed8e5c2ed455bb84be3e5a538b100 : TIMED_OUT in 2h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/554b2e12085e41a7b40280af36e3398b : FAILURE in 11m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ef7e04fd4ba54f31ac18201185b1ebaa : FAILURE in 12m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1842af3ba18e49e59b2b2a3e5c9c5cf5 : TIMED_OUT in 2h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e81d187c4d8cecfcaf0cdc701903c366eb9f0499","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-28 10:36:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a6bbaaedac4df5bed0d28c7e185cf3d5a53fa667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:17:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc048ba767a449b93e67d07d717dd56 : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b88f674488845898ff6dc7417336ab5 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac35c84f1e904beb90cdaa081e17da88 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f51f5cb5a04413e849222acef66c2dd : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc1563b8abcb4746a6fb447a999f1c39 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e51ef6a7e74bcbb55b1efa8cfe33bf : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33255cdf97df48dc88622f13f830e625 : SUCCESS in 4m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/697d9c6842c245b1ba3657b5e34291cd : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a2b834e5ad864879b74144e93fd34c08 : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c308a0d2515413ba9a4e2bbba8294db : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd4969424bc94a9090df33c06fd050a7 : SUCCESS in 1h 11m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d265f4c69dba46138ea1717fdd9567f2 : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a79005083af94b02adba1f37f349bad6 : SUCCESS in 48m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ccaa5c4833394b0d817758278992fb56 : SUCCESS in 34m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/633ec9fc8c6b4d1c991859e2a7d1e5fd : SUCCESS in 41m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01bb3b1f18f34fbf97090b277cdcc907 : SUCCESS in 1h 02m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f266fb93fe7d4bc889718806a1a09299 : SUCCESS in 1h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/03f48a4a204d4c98ae016c81ff06ca48 : SUCCESS in 1h 12m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a9827c9af60d4cddafdfe32cbb519d54 : SUCCESS in 1h 04m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7310d0248417415781effe7778e5e7f6 : SUCCESS in 1h 26m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a863fedc832b4f9eafee26ac14bcc5af : SUCCESS in 1h 19m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2e8e55bd4d0b4fcb9ea267311be9a292 : FAILURE in 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ca476e7144907e3c93ab49a9cb41664153bedefc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-03 06:49:57.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f5a8164f92f901c81eaabed0dc660ed775499774","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-03 10:43:42.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"721d3c742d63e121686290c48249c177a5409972","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-03 10:44:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1f7ce937bdf34a4c3462c8042381bf29f0790c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 12:43:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e6b9e9e40b54cbbaee3b06fbf0e2da0 : SUCCESS in 12m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b241f3850f8e4f4c8e28ca43d3fa744a : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cc8038f8d04e2790005222cd034f81 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc45e03c39b4fe5a025bdbab9867344 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08285aef9e514a57a81123b5a2c6597a : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d896599050477fb5776bb62a55cade : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43f4880e60604414b9e3670e3b2bc776 : SUCCESS in 4m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/375b862d055749df9fbcc691db048c49 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/033524f578e4461d93113e6da282710d : SUCCESS in 10m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43839e0558e249f6b6e6bffaabf8a581 : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d5f5875ebcd4d3dbb498481653605f9 : SUCCESS in 43m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9db3e13cdc4466888661def8b7d83b9 : SUCCESS in 1h 00m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5c5a65eeff1b4ada96010245e9890d57 : SUCCESS in 53m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be21a651568d40b2b673a11765d19d38 : SUCCESS in 50m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30ab29684bdd4d48a0fbc3ce1f4699ac : SUCCESS in 54m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55e540adac1c45aca4770f6834114c22 : SUCCESS in 1h 55m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9c41801d0b040c1a25c07fc011b084d : SUCCESS in 1h 12m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e24b9c50bd947c3bf584b24935f39ec : SUCCESS in 1h 00m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4015412972f0483282bea546c1c2e248 : SUCCESS in 1h 09m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f2a6b9265a3b4bd1a571e81dc9312124 : SUCCESS in 1h 10m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a71c457fcaa54630afeceb6cb71e9e5c : SUCCESS in 1h 02m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/77f25c51b94d4fc3afb2d68a0cdd881f : SUCCESS in 1h 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"186c741b886835f51bc9f57d287e80661fadf513","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 16:05:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"8e42a1656f74e94ce7957b6c1abeee5ea6b918ab","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-04 08:34:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"487eb46d7567dd69a09b7c2054c6da96fdb7c47b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-07 13:31:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c21b03f108280ca40a1734ee4ecf115538588c4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:31:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"db21681a86bee4453b0d82b36fdf1185b53bc356","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-07 13:57:03.000000000","message":"Patch Set 6: Backport-Candidate-2","accounts_in_message":[],"_revision_number":6},{"id":"0c0a08b26f6d9d496290c7d80fde696480e9a244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:00:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"4730b3d730c037980fa97fcdf0716125c2b98e32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:00:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bb90fff27f0440da99e253ec0ad30a1 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bb7525fb6a4f0e920f08576a0bbeb7 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58836286e9a44d7387b9973fc5151381 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d3e5761b6cc4633883eca0476ade7bb : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22df9bc50fd045c6842a714cf0ea31e9 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a572f0a5a27942adad7369a7034bdf95 : SUCCESS in 4m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ee880d8cc9d46bb9280c14337544da5 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/479bb12d7c894f8ab25dbbb8171b9603 : SUCCESS in 48m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11c9dff88de74313afbbd41f621f85fb : SUCCESS in 47m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9520ba666c2a45a3a38069def493ffe8 : SUCCESS in 39m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f6f719c137a748168d8d12ba0ec6dbb2 : SUCCESS in 37m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d477b3de7e3445ffb22a99c7a7761db1 : SUCCESS in 38m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2596c69dd994969abc062948ce80211 : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":6},{"id":"405aa8edef7529a696bac19e63b40321f08de591","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 15:02:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6002b423f25140a38327c3630a029dac : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a703f2924f0c40538bf7a12db775fe7f : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"de2fe22f9d94d64a91fdecfcfbe4bb147c686fef","revisions":{"d6c1abc7869d24c5e1dd3c5d08fe3118ad7c5e6c":{"kind":"REWORK","_number":1,"created":"2020-08-05 13:39:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/744905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/744905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/744905/1"}}},"commit":{"parents":[{"commit":"fc973595e54d37ce95141ec98b8c7d1faf57cd45","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc973595e54d37ce95141ec98b8c7d1faf57cd45"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:49:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 13:37:26.000000000","tz":120},"subject":"Add HTTP/2 over TLS support to amphora provider","message":"Add HTTP/2 over TLS support to amphora provider\n\nHTTP/2 requires HAProxy 2.0 or newer in amphora images. The operating\nsystem versions in test as of today (Ubuntu Bionic and CentOS 8) do not\nprovide HAProxy 2.x in their official repositories. For this reason, the\ndefault listener ALPN protocol list remains unchanged, defaulting to\nALPN protocol IDs \"http/1.1\" and \"http/1.0\".\n\nChange-Id: Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6c1abc7869d24c5e1dd3c5d08fe3118ad7c5e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6c1abc7869d24c5e1dd3c5d08fe3118ad7c5e6c"}]},"branch":"refs/heads/master"},"6757827264681fa4012779b1fff026b4504678ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-10 09:17:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/744905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/744905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/744905/2"}}},"commit":{"parents":[{"commit":"24725ec03fcb3bb57340556e0a5bff218a01cdc1","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24725ec03fcb3bb57340556e0a5bff218a01cdc1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:49:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-10 09:16:58.000000000","tz":120},"subject":"Add HTTP/2 over TLS support to amphora provider","message":"Add HTTP/2 over TLS support to amphora provider\n\nHTTP/2 requires HAProxy 2.0 or newer in amphora images. The operating\nsystem versions in test as of today (Ubuntu Bionic and CentOS 8) do not\nprovide HAProxy 2.x in their official repositories. For this reason, the\ndefault listener ALPN protocol list remains unchanged, defaulting to\nALPN protocol IDs \"http/1.1\" and \"http/1.0\".\n\nChange-Id: Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6757827264681fa4012779b1fff026b4504678ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6757827264681fa4012779b1fff026b4504678ce"}]},"branch":"refs/heads/master"},"d8d17b0bf503d81cd901f940a0db2e0f93ddc45b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-18 19:05:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/744905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/744905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/744905/3"}}},"commit":{"parents":[{"commit":"8016651affc087fc90c03b30fc0087a791b8ba67","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8016651affc087fc90c03b30fc0087a791b8ba67"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:49:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 19:04:56.000000000","tz":120},"subject":"Add HTTP/2 over TLS support to amphora provider","message":"Add HTTP/2 over TLS support to amphora provider\n\nHTTP/2 requires HAProxy 2.0 or newer in amphora images. The operating\nsystem versions in test as of today (Ubuntu Bionic and CentOS 8) do not\nprovide HAProxy 2.x in their official repositories. For this reason, the\ndefault listener ALPN protocol list remains unchanged, defaulting to\nALPN protocol IDs \"http/1.1\" and \"http/1.0\".\n\nChange-Id: Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8d17b0bf503d81cd901f940a0db2e0f93ddc45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8d17b0bf503d81cd901f940a0db2e0f93ddc45b"}]},"branch":"refs/heads/master"},"de9997c6b5db46008c2c6c97e11d701b63dea9f3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-25 14:01:26.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/744905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/744905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/744905/4"}}},"commit":{"parents":[{"commit":"dba52766e9c3e52a51cef4005bcac1136766227c","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dba52766e9c3e52a51cef4005bcac1136766227c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:49:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 13:58:57.000000000","tz":0},"subject":"Add HTTP/2 over TLS support to amphora provider","message":"Add HTTP/2 over TLS support to amphora provider\n\nHTTP/2 requires HAProxy 2.0 or newer in amphora images. The operating\nsystem versions in test as of today (Ubuntu Bionic and CentOS 8) do not\nprovide HAProxy 2.x in their official repositories. For this reason, the\ndefault listener ALPN protocol list remains unchanged, defaulting to\nALPN protocol IDs \"http/1.1\" and \"http/1.0\".\n\nChange-Id: Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de9997c6b5db46008c2c6c97e11d701b63dea9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de9997c6b5db46008c2c6c97e11d701b63dea9f3"}]},"branch":"refs/heads/master"},"b080faf59011d285fbf28443a9595fdda19ab00a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-27 11:49:17.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/744905/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/744905/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/744905/5"}}},"commit":{"parents":[{"commit":"a5f0524fd00b1fdb9d47bafba81155eeb769f3ad","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f0524fd00b1fdb9d47bafba81155eeb769f3ad"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:49:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-27 11:43:40.000000000","tz":120},"subject":"Add HTTP/2 over TLS support to amphora provider","message":"Add HTTP/2 over TLS support to amphora provider\n\nHTTP/2 requires HAProxy 2.0 or newer in amphora images. The operating\nsystem versions in test as of today (Ubuntu Bionic and CentOS 8) do not\nprovide HAProxy 2.x in their official repositories. For this reason, the\ndefault listener ALPN protocol list remains unchanged, defaulting to\nALPN protocol IDs \"http/1.1\" and \"http/1.0\".\n\nChange-Id: Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b080faf59011d285fbf28443a9595fdda19ab00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b080faf59011d285fbf28443a9595fdda19ab00a"}]},"branch":"refs/heads/master"},"de2fe22f9d94d64a91fdecfcfbe4bb147c686fef":{"kind":"REWORK","_number":6,"created":"2020-09-03 10:43:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/744905/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/744905/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/744905/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/744905/6"}}},"commit":{"parents":[{"commit":"a5f0524fd00b1fdb9d47bafba81155eeb769f3ad","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f0524fd00b1fdb9d47bafba81155eeb769f3ad"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-05 12:49:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-03 10:43:07.000000000","tz":0},"subject":"Add HTTP/2 over TLS support to amphora provider","message":"Add HTTP/2 over TLS support to amphora provider\n\nHTTP/2 requires HAProxy 2.0 or newer in amphora images. The operating\nsystem versions in test as of today (Ubuntu Bionic and CentOS 8) do not\nprovide HAProxy 2.x in their official repositories. For this reason, the\ndefault listener ALPN protocol list remains unchanged, defaulting to\nALPN protocol IDs \"http/1.1\" and \"http/1.0\".\n\nChange-Id: Id29d0d2cd5ab1ae3c7d26194a235af25c2271d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de2fe22f9d94d64a91fdecfcfbe4bb147c686fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de2fe22f9d94d64a91fdecfcfbe4bb147c686fef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
