)]}'
{"id":"openstack%2Foctavia~744948","triplet_id":"openstack%2Foctavia~master~I400dde533f78bc14ee568bfd9714eafac97e8a39","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I400dde533f78bc14ee568bfd9714eafac97e8a39","subject":"Fix operational status for disabled UDP listeners","status":"MERGED","created":"2020-08-05 16:34:25.000000000","updated":"2020-08-25 22:40:46.000000000","submitted":"2020-08-25 22:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744948-1598395141105-038e877c","meta_rev_id":"62740cbce680231382da91a9059090d74da8ed8b","_number":744948,"virtual_id_number":744948,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-08-05 16:52:09.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 22:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-13 12:35:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-08-25 20:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-08-25 20:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-13 12:35:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-08-11 20:36:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 20:36:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-13 12:35:00.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-25 20:40:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-25 22:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c18aba31b4fe518437762a22080babb5ac194549","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-05 16:34:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2f75d19a1cf29ab7f34b19b69de300133a0022f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-05 16:41:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4f5e23b8d09ea034960fa409dc9e37653004fc1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-05 16:44:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c4e38cd13aba7aec742ff92ee9b41024c987ca5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-05 16:52:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"588f1e43a078c6304d9127f358a766034aad4e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:29:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23206ff2905e462e8ff32c96d9c2d3d6 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54721da201fd4baa9e966a96cc9cfeb2 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c72909310e049afbb8b1463b470bd4c : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/479f44fe6684421bb8b205be81d7f74a : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e325b319b7254cadaac47d80f7b34c9a : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18decdb1c4b24f3b99504254cc7bbfb3 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/054ca46d560749f6a1a5354c310c4648 : SUCCESS in 4m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e8181074cc1f45098e3651e095ef7e28 : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c1033517294a4dd7a98e5bf3d3876441 : SUCCESS in 10m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a1e14d0d3204f118b299326ced2f1d7 : SUCCESS in 5m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf0bac8166c049be9a2d12456ff1410d : SUCCESS in 33m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dc39eb5e2ce422282b01ae2e0d8e3d6 : SUCCESS in 1h 08m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/817a63d0b65544379320e7fb8c914b07 : SUCCESS in 56m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d74369a7efa04ced9b5888dd688847f5 : SUCCESS in 54m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/734bfc390b9d4912b57b0fdf34349d47 : SUCCESS in 1h 06m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d06808da847c4941b093af62492cd090 : SUCCESS in 1h 24m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a38d0c6cff6940199910e78c59b1c37d : SUCCESS in 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cdd3a9daa3554a698665412d9b27e03c : SUCCESS in 1h 05m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb716bdcb5a451da94ca557c4d53905 : SUCCESS in 1h 24m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7727f10d7dfb46f5bacd048c5a1a2308 : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7371a072f0214e1890086928796fc750 : FAILURE in 1h 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb25b16b70058eb4f1611a868f9e6b88aa0d3624","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 12:35:00.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"700b1ec2da375c1a9080a02bcd38f627fcf26305","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 20:40:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGot it, LGTM","accounts_in_message":[],"_revision_number":2},{"id":"2febda13aef71ecb940784319138189a23d2c102","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:41:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1ed2ba20410e8343688c63ad7db56540d1539d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:39:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32dbb771c06243d0878760d95b054da6 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de78cc2a6f554ddfae05efb0f81e094f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c431f28bcf9e499dbb052d4312ff8bb3 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aeb690b82fc4ed69c6955d3acdffbb4 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c40778f16d94412ba189ba2189e0c3a : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d31019b05b440f895c5a58511dcc4af : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8539bd0c78b346f896be24d2c0198625 : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee054ec0024d420db15892758bdb6204 : SUCCESS in 49m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1d5103922014ed39af91b405080885a : SUCCESS in 1h 18m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/655a2b30511547b8871af1cdcc0cd89e : SUCCESS in 59m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5efb8cc2e0854e1d80cf82e3afcc6576 : SUCCESS in 58m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f6180554de846f7b97038d61099c4ef : SUCCESS in 1h 01m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6774e9540c8b4049b4c530dd6a3a19f0 : SUCCESS in 1h 56m 34s","accounts_in_message":[],"_revision_number":2},{"id":"949cb5023d4c4efd30ee34f4fa4dedfffce63683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:39:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"62740cbce680231382da91a9059090d74da8ed8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:40:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4004228d82f9474792b733b9280262be : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710d5c6da80a45e290dfd203d9aa29ef : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb4f50f798dd5619185cebc5f3db9a55e7f6122e","revisions":{"2e838d0213aa755090881fdb14b58dbdc76ca81f":{"kind":"REWORK","_number":1,"created":"2020-08-05 16:34:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/48/744948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/744948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/744948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/744948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/744948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/744948/1"}}},"commit":{"parents":[{"commit":"08fad7496612962418c81c2d981069bde275157b","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08fad7496612962418c81c2d981069bde275157b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-05 06:14:24.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-05 16:31:01.000000000","tz":120},"subject":"Fix operational status for disabled UDP listeners","message":"Fix operational status for disabled UDP listeners\n\nDisabled UDP listeners appeared as ONLINE, as the heartbeat message\nalways showed listeners as OPEN, even if disabled.\n\nThis commit fixes the configuration file for UDP listeners, previously\nenabling or disabling a UDP listener created the same config file. Now\ndisabling a UDP listener explicitly adds a comment in the file.\n\nBased on this comment, the heartbeat message doesn\u0027t include status\nupdates for disabled UDP listeners.\n\nStory: 2007979\nTask: 40605\n\nChange-Id: I400dde533f78bc14ee568bfd9714eafac97e8a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e838d0213aa755090881fdb14b58dbdc76ca81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e838d0213aa755090881fdb14b58dbdc76ca81f"}]},"branch":"refs/heads/master"},"bb4f50f798dd5619185cebc5f3db9a55e7f6122e":{"kind":"REWORK","_number":2,"created":"2020-08-05 16:41:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/48/744948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/744948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/744948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/744948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/744948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/744948/2"}}},"commit":{"parents":[{"commit":"08fad7496612962418c81c2d981069bde275157b","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08fad7496612962418c81c2d981069bde275157b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-05 06:14:24.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-05 16:41:25.000000000","tz":120},"subject":"Fix operational status for disabled UDP listeners","message":"Fix operational status for disabled UDP listeners\n\nDisabled UDP listeners appeared as ONLINE, as the heartbeat message\nalways showed listeners as OPEN, even if disabled.\n\nThis commit fixes the configuration file for UDP listeners, previously\nenabling or disabling a UDP listener created the same config file. Now\ndisabling a UDP listener explicitly adds a comment in the file.\n\nBased on this comment, the heartbeat message doesn\u0027t include status\nupdates for disabled UDP listeners.\n\nStory: 2007979\nTask: 40605\n\nChange-Id: I400dde533f78bc14ee568bfd9714eafac97e8a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4f50f798dd5619185cebc5f3db9a55e7f6122e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4f50f798dd5619185cebc5f3db9a55e7f6122e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
