)]}'
{"id":"openstack%2Foctavia~745742","triplet_id":"openstack%2Foctavia~master~I6d84047b3481a2bf6bf9bd17d482fb504dbc752b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6d84047b3481a2bf6bf9bd17d482fb504dbc752b","subject":"Fix the unit test tree structure","status":"MERGED","created":"2020-08-11 20:42:37.000000000","updated":"2021-01-18 15:23:26.000000000","submitted":"2021-01-18 15:21:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745742","meta_rev_id":"7deb42d942001511d9d7dd8d1af7054f4de8049f","_number":745742,"virtual_id_number":745742,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 15:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-18 13:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-15 17:27:30.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-01-18 13:29:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 08:18:35.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-05 00:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 17:27:30.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"}],"messages":[{"id":"db3c8be3d176e67e02c36f8bb89493b4b032e572","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-11 20:42:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba54eeca9010d3ed5b505e9a7dfd22e776489377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 23:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/57457482967d4191b519f5c74ecc1622 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3229aba5b33c448aa7f68e5e8ac88fca : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c39e213621244820935900c8f942a293 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c3a604b5c04902a612f763dc4bdd8b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f0696fa91e44554bb55f5a283475346 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3cea57230ac4491acebbcc1dca6a3b6 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75ed9c3583c142c3a7af1b4f5a0c5c45 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e445e3745bf414eab0b5261d29ea0b1 : SUCCESS in 5m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb47d7ae37ac4ea78e790ad7303940ad : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7647a0cfbd1f4e75b9154811fa38d028 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77921719dd0a4200a9728b06a09192af : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7947949967b94b2589759ac9fc7606bd : SUCCESS in 46m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8f69b1fc83f4cc384a2f4756888d98c : SUCCESS in 53m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67c1e345f9a24c32bd447becbe6c9004 : SUCCESS in 49m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4ac8e1d333384ec7908d3605454286c2 : SUCCESS in 29m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c53714831e8a446bb328e719619c4eb3 : SUCCESS in 29m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31b0f2eb9e0c4972a931018748b97c37 : FAILURE in 42m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0fdbe657995b4039ac8df29e8a299edd : SUCCESS in 46m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/607795b37e4c4dc79c84b1e566dccbc4 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8138ef0b12f24e3bb7e424cc16caec7a : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6d90ea0860e0416bab3d6b3b0b543424 : FAILURE in 48m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b68b4ca80f35398a5347aa002701b053cf4e0c3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-13 08:18:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01ba4f9125d996c10c418f7e8ee0ac6b878a039b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 21:57:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6ed7d1dae38ed77a8bff80229f737b73a59133d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 21:57:56.000000000","message":"Patch Set 2:\n\nJust needed rebase","accounts_in_message":[],"_revision_number":2},{"id":"cb4fae685e4d25d7e1a130c82cb9339618af0581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 23:20:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f3d78db9adbd4227b4d8afe7f20fec54 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a420f627e361465a8df190260af97134 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20d9a4b211e4485c995b90e335f1b15d : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ddd14901ad4ebfb4cf8dd2f138bca7 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c985601f4920473db53e47465c35b7e4 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c6312da5df144acb8b7a920aef9b0a3 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3533cf75d9842ee9572d47fbe981c3d : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92d29ec65f0f42c98483c44be5fa7b73 : SUCCESS in 5m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/52fb3ecb38f149e5a09ccb9f126cf5bb : SUCCESS in 4m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1edeeeacb7f542bda548c55c47e5c977 : FAILURE in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d0e92a04c194ee48fbb29223280013f : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b7e99f6f01949ae80e868f37f83fc40 : SUCCESS in 44m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec2fdccd1c8f4f08b9655940deaa6f88 : FAILURE in 1h 07m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0c845e3f09a4ad98396dfaaa9a77fd0 : SUCCESS in 44m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/098e0f8ebc4f47b08668d15e04f15c6e : SUCCESS in 55m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da0a57aa1d1a4edd80557ed712142557 : SUCCESS in 54m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d790d94cf024a50b3a3c31d78457ee9 : SUCCESS in 1h 00m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7577c6bd73b4885922ef8085f915bed : SUCCESS in 1h 20m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97773256f8b449d18c3f89bfcccb5366 : SUCCESS in 1h 12m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5daae324e9aa4a00bd23c3d48b339066 : SUCCESS in 1h 07m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ce50918540ba4c4f807971f8b6828c2d : SUCCESS in 1h 14m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f20e704b22cf49f5b6cfc2a2284d27f2 : SUCCESS in 1h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09d99fc416863b8992092db8df13371047d2825f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-28 16:53:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"17f4f1f9fc54a93a4ca11ede7853171700325024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 18:51:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/90e91d7873794573af8ce6cefa33d3fa : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdc2d36e0ff34883875574f9a0ec22d6 : FAILURE in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a7ed576e814bab878968774711b774 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13116d8074464db4b363158b344c9ccd : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/424a7bfaf2804aea82c53112ef32c7e1 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3bee81569984c6e90d20106c1d01685 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d8884a866b4e7abbf8b036e7043109 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0dfba4c19164d90aa310595682d9886 : SUCCESS in 6m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf48b5f594e644638897eebc4789e6d7 : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca6bcb32956e4aa197f8d84edab649bd : FAILURE in 7m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc39822c4c954f9abc20342efddf7dac : FAILURE in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a26504493540480ea0350cdd79d5d3f0 : SUCCESS in 46m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b010ab8959ac43a68d248586060897fb : SUCCESS in 45m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eed70860db67453eb82de0e25c0d2b4c : SUCCESS in 47m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/127e835993d04d1c93b342790b6a1121 : SUCCESS in 1h 09m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb88477cfdf44b2e9e4d4c7e4e0cb812 : SUCCESS in 54m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3975fdee9d92498ba1b6ef4793c89fd7 : SUCCESS in 1h 14m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b5d5fd75a8b47ab9404db6b3d327967 : SUCCESS in 1h 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9778e71e612e4fc686a184266d489cfe : SUCCESS in 42m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/08b07579dccb407d963fa99a5186a400 : SUCCESS in 48m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f74d6a4057ba457080602aa850d7e31d : SUCCESS in 1h 09m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fb1319a30620423796e867fcfe3d6859 : SUCCESS in 1h 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18aceb8124dee30325e8328232406ac6b4029c41","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-18 18:49:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"267184f3eb3c12f863a4e92cf87510ea7e899e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 21:17:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad99ba98b76a4d5c97d7c1b0d9d9776f : SUCCESS in 14m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f828a0ab473d434fb5cb5302f0ec58f0 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5958252b28c54b0997aa4fb83e131379 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf8dffe33b9a4221bdaa686496fee4a4 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebf2e5b8ccd54639bf2b9048d959885a : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d3de2e740cd43c694a427d1531e8117 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb87f0fd12604f25b94fe8dfef66db48 : SUCCESS in 18m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c4ebcb3b60446e2a56be9d9b227d7ab : SUCCESS in 14m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/77d1d40e58e446c580672ab7f81ada1d : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b3ba13533ba04a9e9e1439905f7a3a48 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/394ddefe4c6e4762bb609e5191bb6153 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d83dc644cec4460799fd622d44c508dc : TIMED_OUT in 2h 14m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4c0fbee98db4aaeb00df350f968b788 : SUCCESS in 1h 48m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/859af1ce8a734e978dd1bd41c34e06ce : SUCCESS in 36m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6e245fb8aba1403bac2707a2a343e708 : SUCCESS in 37m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91409ade296d4c428ea76b9a84b5a334 : SUCCESS in 36m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c6e5e391bc184fa682bbd3fc254700c6 : SUCCESS in 1h 49m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22c69d4801484dc7b63c45c6e342809c : SUCCESS in 2h 11m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/813a9ed9ead84536adb0ad466f04aaa8 : SUCCESS in 2h 16m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8633bf3e341b4e45a0086c8500604129 : SUCCESS in 2h 08m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b739410db57b4ea2b643d93559bccc0c : SUCCESS in 1h 44m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5c2bee62d5594611b6d6cc39eb1d597e : SUCCESS in 2h 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ee87bca3d8c56a2f47389a96dd0eb1a22d1e670","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-18 21:18:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"28ff2cdaab1d5ac4b4bd9576c2291217b2e7c6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 23:57:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35659789f4fa47faa8abaada8316a580 : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33a88510120b44c2abb94eafced94d65 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7bea9c332794460a3c3d6f0aa425a65 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67251cc727ea415f968ea891dd42d1e4 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48bed7669f7e4441bf669d421bb670e7 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79b71c948b864a7fb0cc5afb49c5c122 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2df793bf7de416ca5e538aec24ae630 : SUCCESS in 31m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528048dbed97493b8e4f73892e6a46b8 : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/90d49d48af1748acbfca0ce9c407788a : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c94970aa6f3f46ef906688bdbcb98989 : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8addd6aea11443b0ada863c88c70ea4a : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64dfff5d0c1c44bb976d430f0944a562 : SUCCESS in 1h 18m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1374ababbf93433794caa544f87bc8a2 : SUCCESS in 2h 14m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0fd15d0e8ef94206b5928d01b8f2307f : SUCCESS in 57m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d7fbb0b801a54002881783a0f61f6b70 : SUCCESS in 41m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27b8463afe954dcda8d7bbc2f3c05408 : SUCCESS in 42m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3c788e870f634b5da15d91b31b231c52 : SUCCESS in 2h 08m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/433550c4cc14473fbb6518d1c3bfef52 : SUCCESS in 2h 06m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d054130552db4024b32fb0fe6c944c25 : SUCCESS in 1h 50m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef2e37136f5344c69b62716f68e4b28d : TIMED_OUT in 2h 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/af7b8da8d3074f719104820ddf5b1c7c : SUCCESS in 2h 26m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/49f6ddf3d3b14b5a8c1b31fab206d085 : SUCCESS in 2h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d655cd857ac3abf99a597d8f9acbd029afd7f0d7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 16:31:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"acf8d6055cebf8b6a329ab61a26b6204e99cb617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 19:35:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/309198ca88fe4f3690f2a5b08b84c8c4 : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cad7b49a89894986822934b9e1e9736c : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/167e6df1a55847d19ea09a7b3d72a6c1 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc88171c5ea541c798530f328c604b6e : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def704f04f0e4675aef1e5bd7e5f2725 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ede0bd7034a403995267032e6574b98 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a426c2fb0e47429f37d407f32c1fe2 : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8ab988f439e4f80bb34daf7448f44f0 : SUCCESS in 7m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f9e79fa76c644d879cbaf06b7972e0fa : SUCCESS in 7m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68ca446670fa4fde862f320b82aa0c4a : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41b5df4ec9c9444caeb1b65cbb9dd11e : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5f13364b8a24a5aa87a89f98f23af5a : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4076481283d24882ae9a9823e8c130d3 : SUCCESS in 1h 38m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d11815f0c5da4579beb70c03356f1bd7 : TIMED_OUT in 2h 33m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bdcaacb83de14e22a8fea0bfc4e9529d : SUCCESS in 41m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/70d172d742d44888927032d44dd12a0b : SUCCESS in 40m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e6dfe51de3645b2a34ead0f48eb7c00 : SUCCESS in 33m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/790e94d87619409b8512c42b15af27ec : SUCCESS in 1h 37m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96d5b05a75734019bd5c2cf27d9d3245 : SUCCESS in 2h 25m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae3c92049bf64a0fb8e218c528b394fc : SUCCESS in 1h 58m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/da6fd1c605f24de6a531de6708914bb2 : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/827b2c01c7a94b3f88e921eeca2d1839 : SUCCESS in 1h 48m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3ab13e464a3d4041b41d61bc54c135cc : SUCCESS in 2h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e8f9febf8ee836546dc5d6d36d93a2bde3b5f63","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-26 19:53:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"938effd0ee6815196a5efe72f634fb2c7e7e04c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 21:33:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b5abad826244f609c9d54c9a4c45093 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3eb57a436c2f434f965db8fd9141ea2d : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66463143d4bf4894b5a533bed3db0f2c : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314f733c825c45c8906520870dc41a28 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53a575adc89a48458d46cbcd98a3ea5b : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e1c6d44be849c4b427c096fbd4a9e9 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0180410bba2148a494f7febb07a55a0b : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e0a7e66f2b4a61b9bd2e52444493d6 : SUCCESS in 4m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/009664ca28f04269a64af4d52853e028 : SUCCESS in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/efef3ab86f1f48b3a2749d7d03483405 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/491ca76090794af1a0b368a7f369eb48 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/766cae201b0845d2bc5fb915ed908734 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40b189a845054b3a852a471690a6a111 : SUCCESS in 1h 30m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8709c506fcd43949da3bed34eead5eb : FAILURE in 33m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2984d74d8d2746e5a757c5b0ceeb15cc : FAILURE in 31m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e23c68617a24247b416040025e8afa8 : FAILURE in 27m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84bb555313a04dbebd2dad32742eff5d : FAILURE in 29m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/166dce4c606849968e7946903f4ba0ba : FAILURE in 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f42c7b76fc544dfd8e271892ad28c685 : FAILURE in 28m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9422363778584ca5960c3c57cb672902 : FAILURE in 54m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/894604e4fdf64f2498ae07a8c092a5ca : FAILURE in 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/864729ad83ce464992203928d4c6d53d : FAILURE in 40m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/29510dc3791c4cbca06f3f0cd1c47da1 : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f23e540b8f3166e722d33f268d8df78d9d0ab2a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-04 22:21:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5166ab08b5cd8632ff643fe3550e229de5010002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 00:30:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e073ed12238c496c82d5f2546e69bcb6 : SUCCESS in 13m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dcac70bf63c45289f9ebfa2714309b9 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0bdd36b80b64c5595c24c7351c10215 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0b2e08982a4706aff0cdc2a09e1bae : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b4e3a6e5d6a40e280058a15577fa034 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/240e86f1a8ef43c1b76c2305ee65ea82 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef41c3eaf074d52841691109410e80d : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd5b631e66c4d43ac486a22dfa49c4f : SUCCESS in 17m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6d0e3fce6694d0f97561106ae8f84f5 : SUCCESS in 4m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5f6107220f97487ba3cd512099d84158 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4e4d485b807942f1b0e58d2253cf04ed : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a9d071293724382bf30c9b2e0908079 : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddddeb0fd5f34459bd7f8a47d88ab5b9 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d66f9c8e881d493c8f53bd2714a88893 : SUCCESS in 1h 15m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/234251d5db0941e99906308a801f295f : SUCCESS in 1h 56m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/158b841214da4ac6b195dee7a217791a : SUCCESS in 38m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e26ec926bb7452fb0a4ed1022ac7bf4 : SUCCESS in 36m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27ffbdcc072d4d17894d57229acf0476 : SUCCESS in 37m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a35d59bc8f7c4129ac0d4652b9881e14 : SUCCESS in 1h 48m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd8898ba02624554a162bbf43e64ec00 : FAILURE in 2h 00m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ef58e24a2de14f49b3ff36026a82c768 : SUCCESS in 1h 48m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/05523c8fd03c4767a6d311ffa2eff33e : SUCCESS in 1h 45m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e8730f19e9e048e9bceec9bd0419300a : SUCCESS in 2h 04m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8c0881724b0b4d2894ea8345b876c440 : FAILURE in 32m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e11613a23cd140c7f0313b4c37e3f899eab47f8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-05 19:50:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"23e33fe07f56de5f38927171cad6b67fc02acd50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 22:06:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0de5e9a3e814dcea408a850a59f433d : SUCCESS in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef9da0a044ec4f2c8130adc26afa55e6 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c05a00b6d764f508470e35eefbb0ef2 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c9b7a96fbdd4fefa55199381467e348 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10455065bd254a37a1897dfcd5d8d278 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc85f3b687c482fb34bf358e471a1b3 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89cccc51719b4997966e0ec00cbf38a3 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8a6a1b77c04b6fba375b3b0b1466ad : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c0457c8252549be81a3226945b642fa : SUCCESS in 4m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/499c4a1bb7b74a7e903fc5f4936126ac : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f3388871a1c44c16a61738acd1630d93 : SUCCESS in 7m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41c31dd2940847bea63e45c5dcc4434b : SUCCESS in 4m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6ca81e26d53463181dd6842d3ca4da7 : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c23bd1efc47f4dfb80e016f38543a4d8 : SUCCESS in 1h 22m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/141db796d5a348e7862b39694fd28a5e : SUCCESS in 2h 13m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ecbb794ad2b4570bddf3d00e399ca1a : SUCCESS in 42m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f7bdc473263849ebbed2f949dadd5c9b : FAILURE in 8m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f625904714a4afdb0d83668221c571a : SUCCESS in 39m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54d0eeff54de41c68186b46bf7ccb96a : FAILURE in 2h 11m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05d5cf45519d45daa5e2a209c605776d : FAILURE in 1h 46m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fbeef144b93e4577a798ef6b1a88a493 : SUCCESS in 1h 52m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7097b24170bb480a91a06448e8739503 : SUCCESS in 1h 54m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f24d005d5259425ea7fa937db219bb6b : FAILURE in 9m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/800461ed2362434a992bfd35acd8a972 : FAILURE in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fcc59f1261e076b1c9b034b108c877442310f7fa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-05 22:16:04.000000000","message":"Patch Set 5:\n\nrecheck\n\ndpkg issue","accounts_in_message":[],"_revision_number":5},{"id":"4456b97de555cf24903c07be318ec869402403db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 01:12:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bc72f8e07daf4d3a8674e9ff996d5be3 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9cc1d66ff574a67a6e5b904eb8e34c0 : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72c66bc900b44d8e91b457eb26ee1b63 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0199c561c97b43bcb3a4cb1c6d301abb : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/913569b80d2242e98a4dcaaa2c850340 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35c70257964f4240b673073993b0f57e : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f60bcc8f17d4fd28de0abc268a8c4ae : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8747010cca24933809b0e99240c8a33 : SUCCESS in 14m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c0ccb1dd3924a11aed0a7fb19988447 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e123b7cbc2b9468897ba4bdef591cc72 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0a6dfd242c72405a824bed8a63206a4c : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5d4a7324ce74162a6927b0bd279c9ab : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d36ba0d88f38488d915783496acf1862 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c9e698474bf4f4dbc7c2387de51ae10 : SUCCESS in 1h 19m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56da412777d5471da2bc807a574c2386 : SUCCESS in 2h 03m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52beed0308164bb9a72a015981f454c5 : SUCCESS in 51m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/339755c1b696416087786d8010158308 : SUCCESS in 45m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac79aabca085416d810942348a376c09 : SUCCESS in 37m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/81f8c63a5b0846209a362c7c32cb401f : SUCCESS in 1h 59m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cccf2a86786b4553abf946ea10bde4fc : FAILURE in 2h 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/487e3f3a4a5349c0ae64237953e1ab96 : FAILURE in 8m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2ccbe7a72ecb4c32a0e8597026f56354 : FAILURE in 8m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/58238814f3e949a28b1363732f4b9c85 : SUCCESS in 1h 52m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3c50bacca69146f5b93b8f37e78cbad9 : FAILURE in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f1eaaddbe2cbf654d958045993582878ee68ab3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 21:00:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"153f312fa38dcff294f6f0a38b040283ffb5b5f9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 21:00:29.000000000","message":"Patch Set 6:\n\nmanual rebase","accounts_in_message":[],"_revision_number":6},{"id":"c33a944f3b3ad80f29d5e1b81baf92039fb987ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:29:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c80337e5058f4b57a703a12502847d14 : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671f2604b19048f49cbd4b767ff1115b : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f4e297f967e4802821eebbed3c08969 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a113505fc7745a59f13cbd63936503b : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e487186d16374ecd9152cf4277b63664 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8166c572fb394e8eb4fe5a003643c78f : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29938db980b4be1a75ccb2a2661254d : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8ef8cb6af04f8a8b81214f00e892db : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/767734cd36ed45bca1e062850a8e9f43 : SUCCESS in 8m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/410acb37390041a694199d406920e8e9 : SUCCESS in 4m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aa562ae54b2247409bfa098cda3447e5 : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae04dddba0cc46caa4176a2f71a14c52 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9307efb284b48eda40e56fd4bdeaa63 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c2ca65d512994f0ab78f0234d36ffd45 : SUCCESS in 1h 27m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38adb7814a2c44948e9cf7628c7ee99f : SUCCESS in 1h 51m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5b12bc112384697ae06581efe16b79b : SUCCESS in 43m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e94347b316fb47a89c6eb9495bdec524 : SUCCESS in 1h 06m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/751f4f977fd44e789b1f606fa8917266 : SUCCESS in 37m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26e5aabadaa04208b6f605fc20a6dd81 : SUCCESS in 2h 03m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf523d78cd7e4c52b3560de6b56a0deb : FAILURE in 2h 05m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23de3079e27641f0b123e20ae9ed3842 : SUCCESS in 1h 51m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cf8fa065e1aa47c7804f83ea6acaa3f3 : SUCCESS in 2h 14m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/87355b76be684bedb19b78f1094483e9 : SUCCESS in 1h 55m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0b2bd20c2b3c4903977807f1afdf8bdf : FAILURE in 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5679235db957775b54d8f1e92b213bfaf35d8313","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2021-01-15 17:27:30.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c05b7214199134220177b6389f2613bc7e3303b0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-18 13:29:18.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1aefd79a713409db171388fc61a41e8ccde9e071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 13:29:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3aee705e4cf3ef8663e7d565a9003a5e0c3ab68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:21:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/762dec51604d4378b8fd73e1b50786a8 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09248a9ecdf64835a58d4853caaee0d7 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b244f1c1bc9b4776abb542e5c82a1518 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0587ffffffa14e69b9ed965ac164f1db : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/277c8014a50f40e08a9574182e20e15b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6b56f1f25a4c1bab28bbd3704db000 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8415d52c8234ff39c6293106e8844fa : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d2fa1ce9761d4c4cbffa88688dd8983a : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11e329771fa64e2bb9729140a262fe40 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e111a9f78bda4da8beffbd100a1123eb : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c657fc62b15144c78b82dde4b2cbed16 : SUCCESS in 1h 47m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e927f8195a943dd918fe30720412e98 : SUCCESS in 43m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/480b10eb2624496e85f582e22560a102 : SUCCESS in 41m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/571bbffc95614061bcf32858ba6d2a9a : SUCCESS in 42m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0f0a17dff0e4eb4b74c0a7f506fb931 : SUCCESS in 1h 16m 39s","accounts_in_message":[],"_revision_number":6},{"id":"ddfb9cf858b5e4cc0f6ba4441b8ef847ae03a06f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:21:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7deb42d942001511d9d7dd8d1af7054f4de8049f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:23:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/633e38bd43224157946dfbac305db1c0 : SUCCESS in 1m 12s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d13fb07c102404a994bb812ae51c6d4 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4348dac61f94c34a12d9a5b493b8d01 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a4aa03d3bc98eb27cc353140cd998a623baa505f","revisions":{"3051697799202543743da9a812b278fa74fa983b":{"kind":"REWORK","_number":1,"created":"2020-08-11 20:42:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/745742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/745742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/745742/1"}}},"commit":{"parents":[{"commit":"d78e992a9ed28f5dca36ae38efc5352b5c11cf32","subject":"Merge \"Fix haproxy user flow log format substitution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d78e992a9ed28f5dca36ae38efc5352b5c11cf32"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"subject":"Fix the unit test tree structure","message":"Fix the unit test tree structure\n\nThere were a few unit test files that didn\u0027t match their\ncode counterparts, so were a little hard to find. Moved\nthings around to line-up better, leaving only a handful\nof exceptions to ignore. Added a test script to check\nthings so it won\u0027t happen again, copied from Neutron.\n\nNo actual code was changed, files were just moved around.\n\nChange-Id: I6d84047b3481a2bf6bf9bd17d482fb504dbc752b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3051697799202543743da9a812b278fa74fa983b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3051697799202543743da9a812b278fa74fa983b"}]},"branch":"refs/heads/master"},"744d21d2265d0b3a46c9ec3d4c6c109e8fe255d1":{"kind":"REWORK","_number":2,"created":"2020-08-27 21:57:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/745742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/745742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/745742/2"}}},"commit":{"parents":[{"commit":"0b1d8dd5e76edc4b3058a36b47f6c059f970e516","subject":"Merge \"Update devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1d8dd5e76edc4b3058a36b47f6c059f970e516"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-27 21:56:21.000000000","tz":-240},"subject":"Fix the unit test tree structure","message":"Fix the unit test tree structure\n\nThere were a few unit test files that didn\u0027t match their\ncode counterparts, so were a little hard to find. Moved\nthings around to line-up better, leaving only a handful\nof exceptions to ignore. Added a test script to check\nthings so it won\u0027t happen again, copied from Neutron.\n\nNo actual code was changed, files were just moved around.\n\nChange-Id: I6d84047b3481a2bf6bf9bd17d482fb504dbc752b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/744d21d2265d0b3a46c9ec3d4c6c109e8fe255d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/744d21d2265d0b3a46c9ec3d4c6c109e8fe255d1"}]},"branch":"refs/heads/master"},"4ae4e46e54ca8aaf75a76b6d46fe3f0b20b77361":{"kind":"REWORK","_number":3,"created":"2020-09-18 18:49:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/745742/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/745742/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/745742/3"}}},"commit":{"parents":[{"commit":"e3085940c61b53f37e61ebaf3fe78e6777fbb8dc","subject":"Merge \"Fix amphora failover when VRRP port is missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3085940c61b53f37e61ebaf3fe78e6777fbb8dc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-18 18:48:51.000000000","tz":-240},"subject":"Fix the unit test tree structure","message":"Fix the unit test tree structure\n\nThere were a few unit test files that didn\u0027t match their\ncode counterparts, so were a little hard to find. Moved\nthings around to line-up better, leaving only a handful\nof exceptions to ignore. Added a test script to check\nthings so it won\u0027t happen again, copied from Neutron.\n\nNo actual code was changed, files were just moved around.\n\nChange-Id: I6d84047b3481a2bf6bf9bd17d482fb504dbc752b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ae4e46e54ca8aaf75a76b6d46fe3f0b20b77361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ae4e46e54ca8aaf75a76b6d46fe3f0b20b77361"}]},"branch":"refs/heads/master"},"63496e18e2d85d0f23a40ae5414d2060fb813951":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-04 22:21:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/745742/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/745742/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/745742/4"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-11-04 22:21:05.000000000","tz":0},"subject":"Fix the unit test tree structure","message":"Fix the unit test tree structure\n\nThere were a few unit test files that didn\u0027t match their\ncode counterparts, so were a little hard to find. Moved\nthings around to line-up better, leaving only a handful\nof exceptions to ignore. Added a test script to check\nthings so it won\u0027t happen again, copied from Neutron.\n\nNo actual code was changed, files were just moved around.\n\nChange-Id: I6d84047b3481a2bf6bf9bd17d482fb504dbc752b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63496e18e2d85d0f23a40ae5414d2060fb813951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63496e18e2d85d0f23a40ae5414d2060fb813951"}]},"branch":"refs/heads/master"},"bf604c39dc48823c6860f15b821841d8a6a0002c":{"kind":"REWORK","_number":5,"created":"2021-01-05 19:50:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/745742/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/745742/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/745742/5"}}},"commit":{"parents":[{"commit":"4704e458a8f0c5886977aa4c87a6c805a518c43a","subject":"Merge \"Fix periodic image builder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4704e458a8f0c5886977aa4c87a6c805a518c43a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-05 19:47:52.000000000","tz":-300},"subject":"Fix the unit test tree structure","message":"Fix the unit test tree structure\n\nThere were a few unit test files that didn\u0027t match their\ncode counterparts, so were a little hard to find. Moved\nthings around to line-up better, leaving only a handful\nof exceptions to ignore. Added a test script to check\nthings so it won\u0027t happen again, copied from Neutron.\n\nNo actual code was changed, files were just moved around.\n\nChange-Id: I6d84047b3481a2bf6bf9bd17d482fb504dbc752b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf604c39dc48823c6860f15b821841d8a6a0002c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf604c39dc48823c6860f15b821841d8a6a0002c"}]},"branch":"refs/heads/master"},"a4aa03d3bc98eb27cc353140cd998a623baa505f":{"kind":"REWORK","_number":6,"created":"2021-01-14 21:00:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/42/745742/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/745742/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/745742/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/745742/6"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-11 19:06:51.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-14 20:55:35.000000000","tz":-300},"subject":"Fix the unit test tree structure","message":"Fix the unit test tree structure\n\nThere were a few unit test files that didn\u0027t match their\ncode counterparts, so were a little hard to find. Moved\nthings around to line-up better, leaving only a handful\nof exceptions to ignore. Added a test script to check\nthings so it won\u0027t happen again, copied from Neutron.\n\nNo actual code was changed, files were just moved around.\n\nChange-Id: I6d84047b3481a2bf6bf9bd17d482fb504dbc752b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
