)]}'
{"id":"openstack%2Foctavia~746771","triplet_id":"openstack%2Foctavia~master~Id18783acefc5ba9acb1789a2fd3d52df603f292a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Id18783acefc5ba9acb1789a2fd3d52df603f292a","subject":"Add a message for the amphora image size","status":"MERGED","created":"2020-08-18 17:08:52.000000000","updated":"2020-09-02 10:11:20.000000000","submitted":"2020-09-02 10:06:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746771-1599041173158-9d651cf1","meta_rev_id":"38e99308947e108ed0cfef274adbfb38c9d3fcdc","_number":746771,"virtual_id_number":746771,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-08-25 18:18:55.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 10:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-02 08:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-25 18:46:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-19 13:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-02 08:33:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 13:12:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-25 18:46:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-02 08:33:58.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-02 10:06:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1b856b32b853c8c96155d05bb22e56f3705974b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 17:08:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc149238b3864184b04d33ed83c1af0ea35a796c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-18 17:10:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bb8c8991f112eab7c5dcf5e87556a83c0944ff2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 17:16:20.000000000","message":"Patch Set 1:\n\nI think the in-line execute will handle that correctly. Let\u0027s see what is in the check job logs.","accounts_in_message":[],"_revision_number":1},{"id":"67b3d0d97ad5ed07f70ca7ee2a45a6462d99ca1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 18:09:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe019f58a7747b0bd6621b818a61ec3 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1075064013d242f0ad4e93c18e609c27 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e5965259e64868af70320fda3b8acd : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08a925ec49524ab5aeed3cb1d0144c2e : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59c1e1fef236462d8e112c93d9e125b1 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9013402071564b9ab6ed39f73478a810 : SUCCESS in 9m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c1a7032407fe4bd6a31ca7befb5ce52a : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6ba2c73187d04b519e3db5a3bb2c2f13 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/233dd46521ba47378c34c7da5f4cddfd : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b69f6426ffa4ee5906cde3bae01da5b : SUCCESS in 42m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3583fdc503204e94bf230386475642a7 : FAILURE in 20m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d30dec0d4b3246859f4c316892699e70 : FAILURE in 44m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8ce8e5cace004ddb98ecf45b933a464c : FAILURE in 27m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54b523e9b7f94782b0585800396dab94 : FAILURE in 20m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0206643efe314eda9a27ae69af25e840 : SUCCESS in 58m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a13d27a25d604f43a9aeed9d49b5e48c : FAILURE in 22m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9cfe49e87f454ed690a281743a981012 : FAILURE in 23m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98a7f484fa094c9593843fcd1d5c54cc : FAILURE in 21m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d8ecb97a16f84858a7c36d7a94bb6b5e : FAILURE in 20m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/89c818437efc410ba818016b9c5eed19 : FAILURE in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15b17a1f883b6f7832ebf5aa51685e43461f1e6a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 18:16:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3d654de9de4fc05c24483cfe5c066cd2041337db","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 18:55:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a9dbab578888496c066511e226c2b2c589b41ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 20:52:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/538907d3b16b45fba2ee0e0967596e73 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a4ac0f2bb44195ae231128ce6fc03b : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d105bdfbe0043abb1e771449116363e : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/738b66a790ce437792f7d2a680e89cd9 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/199cf32172894cea8addfd823e954784 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760efebf3b3c436cbdc6aaf660c739a5 : SUCCESS in 15m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5c8b3994339a45988eb4e4e9eff04fb7 : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9c47565062b343beab04baaa19581840 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/714be60cacf4497993a033d732ef8106 : SUCCESS in 9m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6be4dc9153949b38e3f38c13b399b60 : SUCCESS in 56m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b580e25fbd7e49729e9c3bdfbbb30396 : SUCCESS in 50m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/639baf12811941e4ac107a483c429d9d : SUCCESS in 39m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0351d302c9cb4ff6b992dc40ef3a6a82 : SUCCESS in 35m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d24e3aada1e46d7b16feb1b0d80e934 : SUCCESS in 1h 03m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0879137da8d4bf4ba8fbd6206ce62d9 : SUCCESS in 1h 54m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7809a10aa4e5430a862a519284e8914c : SUCCESS in 54m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fda35ac27be94220b7051965a72c042e : SUCCESS in 1h 30m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f38c1091dc3040559c14295482bdb738 : SUCCESS in 1h 05m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0974951477b943d0b3dba6cbdad813dd : SUCCESS in 1h 10m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/61d99b22f2e84931bfd0ab4c58e0fba1 : SUCCESS in 1h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12527034d5a2798e5ae66a01d1ff2a4c9b650774","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-19 13:12:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a14bf1870e18db317e628e03ad04fd2aca21c4e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 18:18:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17175d9a131235d3542e484e1bd034d05f876a80","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 18:46:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ad696e92b8b940929d6f11e14e37792a5bfec60b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-02 08:33:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"463c82b2aba75c938695e0594729dc5a2fbbc290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 08:34:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e850bf7761c8c4967f01aded28ad6115d9ced4d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 10:06:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d07e00efb8349c8a253155d8ff0a4b8 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f22b406a1f4eaa9956a8206c4c9239 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f76c9b7e9fc454ebca8ad9526a64584 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c62a3fdbc4437db2e7cc9a69d530bb : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d84c123f3c49b59c1b9e7f23ad1a00 : SUCCESS in 13m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dea2adf62e5d48a7af71543584c00262 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1436ff36d2d4e2995046020a5af28fd : SUCCESS in 41m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4a7c3015b844478b7694601a37fdfc9 : SUCCESS in 1h 15m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/710d09ba203249b591854726bae5c632 : SUCCESS in 57m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f843d2ca7e31476690a2b0ee1b01f7e1 : SUCCESS in 53m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4494783821648c6a5f38ced202e9dd8 : SUCCESS in 55m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcce48ae93df40aeb3706d2a155506a9 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":3},{"id":"40583d037481e78708566bf90a2f83ff20b35316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 10:06:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"38e99308947e108ed0cfef274adbfb38c9d3fcdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 10:11:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210c802a732d4f2f9e634d5017bc643b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2cda017dc1374fa5083b4b1ebce9b1d87f9934d","revisions":{"b32eff492d7c339d2d465a1ab66c7eff1b6ed466":{"kind":"REWORK","_number":1,"created":"2020-08-18 17:08:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/746771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/746771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/746771/1"}}},"commit":{"parents":[{"commit":"e2e873fc9d99b902ae5373baea2ab495bbabc409","subject":"Merge \"[AmphoraV2] Fix l7policy/l7rule quota decrement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2e873fc9d99b902ae5373baea2ab495bbabc409"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 17:07:10.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 17:08:51.000000000","tz":-420},"subject":"Add a message for the amphora image size","message":"Add a message for the amphora image size\n\nThis patch adds an output from the diskimage-create.sh script that\nshows the generated image size.\nThis is handy for seeing the impact of a patch on the generated\namphora image size.\n\nChange-Id: Id18783acefc5ba9acb1789a2fd3d52df603f292a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32eff492d7c339d2d465a1ab66c7eff1b6ed466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32eff492d7c339d2d465a1ab66c7eff1b6ed466"}]},"branch":"refs/heads/master"},"174191076b2444ae619ef22df18d91857cee2a74":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-18 18:16:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/746771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/746771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/746771/2"}}},"commit":{"parents":[{"commit":"38e4075c970599c6e1f992c71e4b339b41aff60f","subject":"Fix DIB git breakage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38e4075c970599c6e1f992c71e4b339b41aff60f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 17:07:10.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 18:16:48.000000000","tz":0},"subject":"Add a message for the amphora image size","message":"Add a message for the amphora image size\n\nThis patch adds an output from the diskimage-create.sh script that\nshows the generated image size.\nThis is handy for seeing the impact of a patch on the generated\namphora image size.\n\nChange-Id: Id18783acefc5ba9acb1789a2fd3d52df603f292a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/174191076b2444ae619ef22df18d91857cee2a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/174191076b2444ae619ef22df18d91857cee2a74"}]},"branch":"refs/heads/master"},"f2cda017dc1374fa5083b4b1ebce9b1d87f9934d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-18 18:55:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/746771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/746771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/746771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/746771/3"}}},"commit":{"parents":[{"commit":"7d6322ae08eba23d7fc02f4141d27d51f9433a7c","subject":"Switch back to using git for the agent ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d6322ae08eba23d7fc02f4141d27d51f9433a7c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 17:07:10.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 18:55:35.000000000","tz":0},"subject":"Add a message for the amphora image size","message":"Add a message for the amphora image size\n\nThis patch adds an output from the diskimage-create.sh script that\nshows the generated image size.\nThis is handy for seeing the impact of a patch on the generated\namphora image size.\n\nChange-Id: Id18783acefc5ba9acb1789a2fd3d52df603f292a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f2cda017dc1374fa5083b4b1ebce9b1d87f9934d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f2cda017dc1374fa5083b4b1ebce9b1d87f9934d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
