)]}'
{"id":"openstack%2Foctavia~746782","triplet_id":"openstack%2Foctavia~master~I9a346b626f852171ec00d2868ccb6f98b1fb8994","project":"openstack/octavia","branch":"master","topic":"git-ref","hashtags":[],"change_id":"I9a346b626f852171ec00d2868ccb6f98b1fb8994","subject":"Switch back to using git for the agent ref","status":"MERGED","created":"2020-08-18 18:11:53.000000000","updated":"2020-08-20 09:38:00.000000000","submitted":"2020-08-20 09:35:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746782-1597916112406-e6cdcfc5","meta_rev_id":"2dea9787238d7b8e16729c68eb1c127d693c9a44","_number":746782,"virtual_id_number":746782,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-08-18 20:07:54.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 09:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-20 07:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-19 07:59:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-19 13:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-08-20 07:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 07:59:11.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-19 13:14:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-20 07:25:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-20 09:35:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cba539b352e2b474949609df3452617dc9f569e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 18:11:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2eb8eb55ca926352065fefd8db34b5add613cc81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 18:27:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"daf7f45ff82c4d0ea5d65f13496c3fdc0029a0b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 19:53:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abfa6c1d1b6a48a990f1d688b872207b : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95d20dc64a6b4f3f9c6232e305de5487 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0721eb50944e3caecf3e40ada72c46 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b84f082859d4562a26815b24fbf0a20 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14bb8548c8c04ad2aeef51a9182e0ebf : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f857ec1f20a49ae9b0a50c2f1d2d2de : SUCCESS in 10m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/708aba51073e4aa1b6eb43ff2a7ee368 : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2dd0d450d8274dec85f4ae422f1683d2 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7118a66356e543a29aecbc3dabb7164a : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4fe060e4a4524be9af0c08c8b5b74f14 : SUCCESS in 47m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45c8672ea63f49eeb981d7c88d7ef112 : FAILURE in 1h 06m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92d671e83a6f4d96b280c1ea4d0e21ac : SUCCESS in 53m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2f691bad9f62425991a796dec42d8c87 : SUCCESS in 38m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ef1955dee604465a3cab7c2584679c0 : SUCCESS in 39m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9908543d3de9443f9ed1457de00d50e0 : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/74b43aef44db49b8b567b41d17568558 : SUCCESS in 56m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43bd00e5b81e4ab2a8fd5e65f20af524 : SUCCESS in 1h 03m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4bdb478dd6f34afbb33ee1fe67e891be : SUCCESS in 58m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dc879d9181a74933917e67fce11f2568 : SUCCESS in 55m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ea72968323af400c9158f7fcfec41c7c : FAILURE in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16d20b935d521d202440644832af2224f146a6e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 20:07:54.000000000","message":"Patch Set 2:\n\nrecheck\nnova failure to go active","accounts_in_message":[],"_revision_number":2},{"id":"18a22bc13d1004a9d8d4d9e7d4b8bdc2ad137680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:51:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e4e47c8d80642bd9364471804261cbb : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13fd0dba40de4ac3ae1fc2e7f020e196 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2999fb18367d43cbbe1fd6d4fdb6a26e : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3cf675ccc1b44aca04e5072cf7cfbcb : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c02535f6c0c408f89dce4dbb9fd93c1 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fe91bbd4b84081873238e34e2706cb : SUCCESS in 17m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce6c693e99ba490fafb592e82152d63a : SUCCESS in 6m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/300e249dd02f4566a4714065b8542978 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ede7fa4957b4ebb8d3cf11d0613216c : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0c6a1fd0e4c459f8288872b7d3647b7 : SUCCESS in 46m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/623d6eb5e34f45e9bf955af7748f73ed : SUCCESS in 55m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0036ebe15a244e298b1a9e4235b07410 : SUCCESS in 41m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cbfae734bcd2414b82581f378b90fb7d : SUCCESS in 1h 08m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23441506aa2b44c6b4dd23b7d90cffce : SUCCESS in 38m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a22bd62cb8b74d42896830712d92e19a : SUCCESS in 1h 23m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fd467070da5b4c6fb7294dfaa7ed78d2 : SUCCESS in 1h 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/06cdf1813a1740b385851fb041902469 : SUCCESS in 1h 06m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d3ae8e5f40fa43b3a0cff52354a40caf : SUCCESS in 1h 29m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7e41705c45be4d83b53ff4a9991b16ad : SUCCESS in 1h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2c42bbfc498c4a64bfefb9e27dbc119c : FAILURE in 1h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"830f6b359463b0658f3fcee98e700b541a2daa6c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-19 07:59:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebfd98f64a77fa9e9c8e58a7a82f09e199d879b5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-19 13:14:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5e8cea8e3fa4f6a1cb0d091df6984d987414faf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-20 07:25:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7122066f5732571e22c697bf47a7a004d87ee9b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 07:27:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e51f63880d09e8c0bb008202afc21cd2e521128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 09:35:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/015992a43d704372859963bd740f6624 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2dfdb9b5f1421db97016e4fe706454 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbcc56ca58024092bff394ec1f3355c1 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994718f13fb0410cb9ff8602f574a079 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167d565eba05486f8db4512a4ac0cb5f : SUCCESS in 9m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8bef3ac21a844a108e038b015de41aac : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7ccb7debe1e479082adc2a0c0ccb3fb : SUCCESS in 42m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42ecb256574949a6a7965d0e546f6dc7 : SUCCESS in 53m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff83fcea00234028b743a04e9ab8ba31 : SUCCESS in 38m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5cf2d6a059cb4370abb52a7542db64f9 : SUCCESS in 56m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb2d259ae0014a6d851cef098d9336b3 : SUCCESS in 39m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c2acc24792b406c8acdce781aadafb0 : SUCCESS in 1h 57m 36s","accounts_in_message":[],"_revision_number":2},{"id":"570c0b99cb3c10b3fd2c5ad492d21401b2cd3084","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 09:35:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4bcf85191af967cd26a026910527dbc9c624a3a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 09:38:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd98e7dbaa242f79a95c837c035e27a : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d6322ae08eba23d7fc02f4141d27d51f9433a7c","revisions":{"38e4075c970599c6e1f992c71e4b339b41aff60f":{"kind":"REWORK","_number":1,"created":"2020-08-18 18:11:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/82/746782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/746782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/746782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/746782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/746782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/746782/1"}}},"commit":{"parents":[{"commit":"e2e873fc9d99b902ae5373baea2ab495bbabc409","subject":"Merge \"[AmphoraV2] Fix l7policy/l7rule quota decrement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2e873fc9d99b902ae5373baea2ab495bbabc409"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 18:09:30.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 18:11:52.000000000","tz":-420},"subject":"Fix DIB git breakage","message":"Fix DIB git breakage\n\nA diskimage-builder patch [1] broke our amphora-agent element by\nuninstalling git prior to the post-install.d phase.\nThis patch updates our element to finish using git before the\ninstall.d phase is complete.\n\n[1] https://review.opendev.org/#/c/745678/\n\nChange-Id: I9a346b626f852171ec00d2868ccb6f98b1fb8994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38e4075c970599c6e1f992c71e4b339b41aff60f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38e4075c970599c6e1f992c71e4b339b41aff60f"}]},"branch":"refs/heads/master"},"7d6322ae08eba23d7fc02f4141d27d51f9433a7c":{"kind":"REWORK","_number":2,"created":"2020-08-18 18:27:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/82/746782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/746782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/746782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/746782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/746782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/746782/2"}}},"commit":{"parents":[{"commit":"1d1142a13d528b5724d323c4d42a004e34ffeabc","subject":"Merge \"Clarify the current status of Octavia in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d1142a13d528b5724d323c4d42a004e34ffeabc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 18:09:30.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 18:27:51.000000000","tz":-420},"subject":"Switch back to using git for the agent ref","message":"Switch back to using git for the agent ref\n\nA diskimage-builder patch [1] broke our amphora-agent element by\nuninstalling git prior to the post-install.d phase.\nA followup patch in Octavia [2] changed how we get the git\nreference to awking an git internal log file. This seems fragile\nto me, so I am proposing this alternative.\nThis patch updates our element to finish using git before the\ninstall.d phase is complete.\n\n[1] https://review.opendev.org/745678\n[2] https://review.opendev.org/745683\n\nChange-Id: I9a346b626f852171ec00d2868ccb6f98b1fb8994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d6322ae08eba23d7fc02f4141d27d51f9433a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d6322ae08eba23d7fc02f4141d27d51f9433a7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
