)]}'
{"id":"openstack%2Foctavia~746802","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","status":"MERGED","created":"2020-08-18 21:21:42.000000000","updated":"2020-08-21 09:41:40.000000000","submitted":"2020-08-21 09:38:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746802-1598002730669-7a2d7593","meta_rev_id":"1214193c8451dd6e4005f57eaacd904e3330a334","_number":746802,"virtual_id_number":746802,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 09:38:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-20 10:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-21 07:56:22.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-20 18:12:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-08-21 07:56:22.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 10:39:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-20 18:12:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-21 07:56:22.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-21 09:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02e3339c9d3bc14655527aa763342771f9657fe2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 21:21:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"796e4b9c146eefe8051b4d2bda243cf082868c78","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 21:30:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcab6ec387fe3cb819f0a41e5fafcce43e36e8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 23:21:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4fa150abd47446fa774c4be17ab48e9 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ebd4fe7555f42b89300d029316733ce : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3caf37709e34e5a81991b0cc488f80e : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca210eae96e54d69a7135d16a5fe613f : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b634ab9acd9447d82f73ae6ec6b7de8 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4d2cd031c04b4bb8187ec2b58db0ee : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd24a2cdeab041aba7c4223c6d2eab87 : SUCCESS in 4m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/50c4c3ea9bc6415fa293d16f9b581f69 : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c8057528b4804fceb28eec8f58337f36 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7160966cdba94c81a3cdc28b70c1452c : SUCCESS in 5m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/864dbb03230b416bb7a5e99702e92bb1 : SUCCESS in 1h 14m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afc051fd411e43c7ba5830f34e8e1fb2 : FAILURE in 43m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84c6ef84e746472c8153f8a1e8bc532a : FAILURE in 35m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0410476fddfd401d96fa9fa3751b15e2 : FAILURE in 47m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5eb72a6b156d4392a8b86f9845a95a16 : FAILURE in 33m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/674cf32f46ad4f4b92015a38e41db187 : SUCCESS in 1h 44m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed7f9df86dbd4374bac899c0c27b90e5 : FAILURE in 36m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c7e41998b95498a86ff6930bf8f8911 : FAILURE in 28m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/424451489c6a40b4bfdf4d4fe5074a4a : FAILURE in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3dce2d37534bb60e1b4b5f481413abebe278ea9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-20 07:24:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f5e30d406a0a828877caba2081f9cd283dadae43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 09:52:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/601d08fdd1b240ff9b8bed334f3cff37 : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/632cdfdb860e4bfa8a58fe5f9cc43211 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b57f87602140738d16e809b434f8b9 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf5cd01a45448f19567820e26b0df0b : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa7cab1891d0436a9e6c9ac34e511797 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809ae1e44c6d4587ae4512ac22266194 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bba74581b80649f3aa8ceb0ea2b06ee1 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f87d4e1ae2a404f8d32e701798315e0 : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6065e03b61184eb99c03b7bbe7bea269 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd3b38208dad407d9b5fde7af3f4a392 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2235115427634472983bb877dc181e4d : SUCCESS in 42m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbb497ba09ee40da9d01dd426dd9d042 : SUCCESS in 52m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca1bc964ea7f4a64a4385428acf78961 : SUCCESS in 41m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ff09e48572f44315aeacdd827beb7b81 : SUCCESS in 39m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/350c766376af4474962592eb10d57b04 : SUCCESS in 40m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f1ec9285b5641089aea4ba836c27393 : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1284cb3e0d66490f93a122ab56ba42aa : SUCCESS in 57m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29e5e6e192304bafbd7b16327e916cde : FAILURE in 52m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c98e4002d17e40c6bdd9c57a659d8b4c : SUCCESS in 41m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30b5da2faf1b15ed2ea912c72546824bd577ec41","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-20 10:39:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1573c108cc49100f6b27c52842c2827848f4232","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-20 18:12:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0bcdb9d81e089d6507a7fe67a9a012914ff3b1cb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-21 07:56:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3c1fec41ff3310c87a86d48e077fd61eca83fc14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 07:56:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04ea633a1025712d7b07868c716c44e07bbb6f79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:38:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"21d8b1697cf2038e3867cb0e617d26c09e04c250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:38:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1437ec78e4084ecb83ff8aaca6cc3055 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fee371cd564965a73527cd380d4340 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d2d5806d3cb43f6a4538d2efdb6f55e : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3be2d6919cd4d148746e66576a52f4c : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a56594fe4e64afea8f49fcba4a71686 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27e10df28c2349ab8b08ad0cccc18f9f : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08553c5d7a15489291916ef8d316cba0 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c1ce5129f3a45db97f1cc417df24b39 : SUCCESS in 50m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b78c963326d49eb849fd1c8063706a8 : SUCCESS in 55m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e61df4269e74b1e900ad605d9a5a5af : SUCCESS in 53m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fa3e56dd980543e59c25c19980a778da : SUCCESS in 1h 00m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ebc1337ab8d40bfa9aa5e3d1d01f5a6 : SUCCESS in 47m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0375d22ce4574540bd6bc18f20e40113 : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1214193c8451dd6e4005f57eaacd904e3330a334","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:41:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816ac19f7b344679aba0ba74c7cb2040 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c09fb21d596b4ffa934a4c9022e14157 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"621acb4278d337e379bdcb0ed65e434bd94db11b","revisions":{"294193bd6ee2be51712f538b45f716f4f4e3b197":{"kind":"REWORK","_number":1,"created":"2020-08-18 21:21:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/746802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/746802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/746802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/746802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/746802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/746802/1"}}},"commit":{"parents":[{"commit":"fcbbfd819d10a3e391c2a9070e097a1f384a586b","subject":"Merge \"Migrate grenade job to native Zuul v3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcbbfd819d10a3e391c2a9070e097a1f384a586b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-30 20:32:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 21:21:23.000000000","tz":-420},"subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","message":"Fix accepting \u0027insert_headers\u0027 when unsupported\n\nThis patch fixes the Octavia listener API from accepting the\n\u0027insert_headers\u0027 parameter for protocols that do not support header\ninsertion.\n\nChange-Id: I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb\nStory: 2007967\nTask: 40464\n(cherry picked from commit 08fad7496612962418c81c2d981069bde275157b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/294193bd6ee2be51712f538b45f716f4f4e3b197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/294193bd6ee2be51712f538b45f716f4f4e3b197"}]},"branch":"refs/heads/stable/ussuri"},"621acb4278d337e379bdcb0ed65e434bd94db11b":{"kind":"REWORK","_number":2,"created":"2020-08-18 21:30:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/746802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/746802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/746802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/746802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/746802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/746802/2"}}},"commit":{"parents":[{"commit":"fcbbfd819d10a3e391c2a9070e097a1f384a586b","subject":"Merge \"Migrate grenade job to native Zuul v3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcbbfd819d10a3e391c2a9070e097a1f384a586b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-30 20:32:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 21:30:38.000000000","tz":-420},"subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","message":"Fix accepting \u0027insert_headers\u0027 when unsupported\n\nThis patch fixes the Octavia listener API from accepting the\n\u0027insert_headers\u0027 parameter for protocols that do not support header\ninsertion.\n\nChange-Id: I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb\nStory: 2007967\nTask: 40464\n(cherry picked from commit 08fad7496612962418c81c2d981069bde275157b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/621acb4278d337e379bdcb0ed65e434bd94db11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/621acb4278d337e379bdcb0ed65e434bd94db11b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
