)]}'
{"id":"openstack%2Foctavia~746807","triplet_id":"openstack%2Foctavia~stable%2Fstein~I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","status":"MERGED","created":"2020-08-18 21:44:28.000000000","updated":"2020-09-22 13:54:54.000000000","submitted":"2020-09-22 13:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746807-1600782775136-c6eda660","meta_rev_id":"92e135588a7f0b34a1498bb5e9a0ccc5ca5181b5","_number":746807,"virtual_id_number":746807,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 13:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-21 09:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-21 07:56:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-18 18:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-21 09:21:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 07:56:52.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-18 18:39:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-21 09:21:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-22 13:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6bbad8407799a60f0eb5d1626d18a17123cc719","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-18 21:44:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab28f39762e6b303c2bd1d30388d4a3614172da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 00:41:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e86e341815db47d3903fbbe2a560dca4 : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a870da1aba649868551883d4eefc206 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb2943524de420fbc39acfe2b532280 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f82c575554f64dbd9e37f29f8340a40b : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/769a75fc6e164d83a5c57836a6c70a34 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d6c698834964718869497b6ba84def0 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1412e66c1c9e4b9c9bef6e271f271321 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f020f8ecd1642fc898352ebd81d185f : SUCCESS in 8m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a6a1a909c318493aa4b21969bb7b345a : SUCCESS in 1h 24m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/198e4407f873433aad0dab73abbd1121 : SUCCESS in 1h 22m 46s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40e2ec6f92d74d279218b8e9aaeb2989 : SUCCESS in 1h 28m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3c388c03cae84387ace3cf2f2a9cb585 : SUCCESS in 1h 28m 19s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ca23d64016fe43ad8f67a271ca5e5e3b : SUCCESS in 7m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/61f8b00ed60848598cd648ba6170458e : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/477b6ca5235a4906a8323a90f4edf395 : SUCCESS in 6m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f7328cc18b44a68a458dfb20900bee4 : SUCCESS in 1h 07m 22s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/38898eee64cb4d4c995cadbe358e20db : TIMED_OUT in 2h 11m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03714536fd5c408ea568fbb54b437dcd : SUCCESS in 46m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ca294cbcee3747d3b97456b95ee15f47 : SUCCESS in 33m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b078d5871a274233b852039627aedbdc : FAILURE in 43m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/714bb6323fe642708344dfa3f49c48a3 : FAILURE in 1h 03m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b046e3c610124a1883c174dc10c443f1 : FAILURE in 14m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d46e1e33ed8947b8877f626ac40057b7 : FAILURE in 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e287e480c2149faa58d078f2492e276 : FAILURE in 46m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3fe6c77837d4ba196e81f4ce4605881 : FAILURE in 44m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50dc6358a00a45b8a5d548349a874497 : FAILURE in 58m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0fd9c2e43d684c3daff42017e5c513f7 : SUCCESS in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21cc43e2f1b604175cf72fe449b24d27e9a135de","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-21 07:56:52.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ad54474d6b47e114100e8b6eae1ae582f808be84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:12:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfd41eb395834a5abb2143286591e552 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ace658ad0af24ea59d18a9bfe14a4a2a : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ec9d3259cf49879e40a54373754273 : SUCCESS in 7m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/764d50a2d90c44cdb3001be738dfae11 : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bfa4dd5d6fd64e10887686f0132d08bc : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b2ed31532794f43a62d5f99d8ed5467 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba02740fc9d4223936e38cd2626d05e : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e06bf1726b64a0facfd3243c037e636 : SUCCESS in 5m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7356f64ae3a04a648efce3563a010501 : SUCCESS in 1h 14m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/25c3e6de932a48939b413862bf831bff : SUCCESS in 1h 15m 04s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57140edd261e4c5d83fed34666498a90 : SUCCESS in 1h 17m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/42d54764ea5c4ea992dd1d5f40927767 : SUCCESS in 42m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a281520f1852456d85d6993e7e30623a : SUCCESS in 6m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/875f79f853dc4bdb9facac335311b893 : SUCCESS in 8m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cae14329768427ab3fee2a7ded688bd : SUCCESS in 7m 52s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/857e000f2bf743c180f3b4d8b3ca2795 : SUCCESS in 1h 09m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2ca80ee2308e4988bac130fe8e51225f : SUCCESS in 52m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc5e95aa9bc740a1832aae2980f4fc23 : SUCCESS in 46m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d20949b5c11f4d94b290a63bc7008b9a : SUCCESS in 39m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fb1769a7dd4441fb244c9f5c1e1b4a0 : SUCCESS in 1h 12m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/caa7d6dc514848e3a667387c0d0bf1cd : SUCCESS in 1h 11m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a338de4182d34196a93089076416c3ae : FAILURE in 16m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcbfbfdd29474bbeb07ace5b5dbb1101 : FAILURE in 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23d7d179d79a4a53bb0e8e0d0e335c0c : FAILURE in 49m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/41b433a4b93c40009c0b42afdf7dfa56 : FAILURE in 45m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1c7cceaebc96451e828cb21b2043615f : SUCCESS in 1h 19m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7324f39e05624b69a1eb9804da35969e : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a81d3aa88667b9ffe1535aca3b79faafeee3090c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-18 18:39:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"59304c7e17577b0e632f7c757f08c14d4e6f3c72","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-21 09:21:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f9e431ded1e6d862e7175d2e5c2cba9d38bbc1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:21:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6e05573654d6459feccaea8e75ce3b1b8345928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:56:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35c97f48875a4cf3aa574c1f70764bb5 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4a46891ff74689a128f2936c721fb4 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/792b7388f8404b7aafc43aadcfd0a9a5 : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1bc98c756bfc44b29933cdb4507d96a8 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7630933c278e45fea4a1bb6060077f20 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215c035276eb4eb2bfb2ee2dee0bea4d : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53462b5d680945569b81d2d920b60800 : SUCCESS in 7m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aab6f76a9b124aec9c360acda69cf2e9 : SUCCESS in 8m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/713dcbfc9d22427cb395be29ac879420 : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bce51af63a444b9afc15b29f949b494 : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74e4790499b74bd794c857ebfc2dc239 : FAILURE in 1h 28m 42s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1edf8773436d4b8e9de9fb82d0dae75a : SUCCESS in 39m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/be3f5282fd6c40b38fe0c34735458ea5 : SUCCESS in 1h 09m 06s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7a14a02d07604e7ca5d74ffb43232ddb : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f861a33d033d444ca4e71be4b8702ef8 : SUCCESS in 1h 19m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5972898088b64a7d9ac31191c5abc08c : SUCCESS in 1h 22m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72dd651652dc44e093ea589d62ffdbcf : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b05be8dfb3ea4a5392402f7b563ba469 : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"13f249ad97adf3752d716320c725f5781e7fe5a1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-22 07:38:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3781325c86f03050ea83ae6f7d45446afe18a8cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:17:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58e6ffd78545495c99446d6e24df1e90 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3293b26c9275431c8aa3f9df8934789b : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f394c8072a4536a6b8315899b18c6d : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98ecb79f67a24224a6f4ff26ad40b911 : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ca391e0549742ac949b7033c1201958 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8edfae0da0d452cbcc6268b4cb1e00b : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7718dc39d02a4a43bc042249bc0aca59 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97c04ec48e3d422a8a90aa67a8833894 : SUCCESS in 5m 58s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b72ed503a0014fc7a5a858b137db0476 : SUCCESS in 1h 13m 25s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3176f08df4c2464c92fa6bdcf96d5ea8 : SUCCESS in 1h 12m 27s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc0abec69d944f30a48d2bdfe840cf13 : SUCCESS in 1h 09m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/32a5f85388d24429b496d126da65a450 : SUCCESS in 1h 12m 10s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/51b77d96020d43d88baa9b2cf85a72b1 : SUCCESS in 6m 51s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/05b1ae02d3f5423dbed6baf3fdebc4de : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/41c68f8ca77d4547833d71ffd98d5034 : SUCCESS in 8m 58s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fe7b417ced94cdfb6c295f98b41eddd : SUCCESS in 53m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a54af168e4994588aac8a9a0e296944c : SUCCESS in 1h 02m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1348c562bb714e62ae020eb57eb74802 : SUCCESS in 1h 23m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c328a666c2b64164b29ecaf5591c3843 : SUCCESS in 1h 23m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ea4db7bdcbd414295586a273983ae7c : SUCCESS in 1h 28m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2a5d1123067249abb4a1bd7572f1813e : SUCCESS in 1h 13m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5d4a4cf81d1e4b1c8343adbf1f5f7987 : FAILURE in 13m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e75a35d05e54155b4110b487f385c22 : FAILURE in 49m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/75b5cc63c496452cb33a53fc239c657c : FAILURE in 43m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49fa6b03cff0493f81e54a201de0981b : SUCCESS in 1h 16m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f73f5b2eff948dc9c3d2e6f0c24ac3b : SUCCESS in 43m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb6ff97bcc5abedae130e175c80c46cec6c318c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:17:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42f9884a8fd3530ad07209f0545dab303b8a7003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:52:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f684a304fbcc47c29e996d5638b7b6af : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34f0e2d6213448ebe52b7a001970521 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4d86540396e4778823d4e5aaf5e737b : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/424852b2a7054d148322584b5f53072a : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49aee991cf2a43678a3050638bdefecf : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad2de0c33724bff959e27e9c25f5d16 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93527975f6984ced9c9d1494989414db : SUCCESS in 9m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/957669a4559146fcafcda7961099b68c : SUCCESS in 7m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/940b5335a93640c3b8e7b04e4bdae9be : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c1cb498e66c47c0b4c81239a851b4ed : SUCCESS in 7m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fd3a08aea6f40f793e777d47d59b813 : SUCCESS in 1h 06m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2d16c0a581b84d97a715f5c26703dd85 : SUCCESS in 50m 15s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/83673295498644049c2a40d591047933 : SUCCESS in 1h 12m 09s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/50276a82ca244a84ae00ed4fa6bb84e7 : SUCCESS in 1h 19m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38d20e463310415ca5d35b56f8e59c03 : SUCCESS in 1h 42m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/97cd95ff0ee64c81b6d4598f08cb2479 : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff0ef49ac78a42fab553c65763310029 : SUCCESS in 1h 32m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8b2407d6e5d745a6b1071cb37980437f : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2c328466f375f3cd386bafebd261b71f0c7551ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:52:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92e135588a7f0b34a1498bb5e9a0ccc5ca5181b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:54:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b70d2b7af5d420c92af7d85bdca222f : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/402ded178ada497fa49c46dd4ed14736 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4031370f4da826aa547142cee604aaa14c67efb","revisions":{"e4031370f4da826aa547142cee604aaa14c67efb":{"kind":"REWORK","_number":1,"created":"2020-08-18 21:44:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/07/746807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/746807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/746807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/746807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/746807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/746807/1"}}},"commit":{"parents":[{"commit":"f42efd0e6e984b2420d810ead92c3e3182cdb288","subject":"Merge \"Fix the interface filenames for Red Hat amphora images\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42efd0e6e984b2420d810ead92c3e3182cdb288"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-30 20:32:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-18 21:44:27.000000000","tz":-420},"subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","message":"Fix accepting \u0027insert_headers\u0027 when unsupported\n\nThis patch fixes the Octavia listener API from accepting the\n\u0027insert_headers\u0027 parameter for protocols that do not support header\ninsertion.\n\nChange-Id: I4ec2299b64b180f8b2d8f0b8485a6be9fe32d2eb\nStory: 2007967\nTask: 40464\n(cherry picked from commit 08fad7496612962418c81c2d981069bde275157b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4031370f4da826aa547142cee604aaa14c67efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4031370f4da826aa547142cee604aaa14c67efb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
