)]}'
{"id":"openstack%2Foctavia~746822","triplet_id":"openstack%2Foctavia~master~Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4","subject":"Create InfluxDB driver for statistics","status":"NEW","created":"2020-08-19 00:49:12.000000000","updated":"2021-04-17 18:28:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86a2292d62831f248ee575730f2e4c4f677645e1","_number":746822,"virtual_id_number":746822,"owner":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-17 18:28:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 01:50:01.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2020-08-20 23:16:50.000000000","updated_by":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"reviewer":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"state":"REVIEWER"},{"updated":"2020-09-07 21:48:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-07 23:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-17 15:37:09.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"87c356fed6d117cdd76ca11a63e9a86228cbf337","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-19 00:49:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38f16b5cca65d45ce6f57204a5a67b42f38aa585","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-19 00:57:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6f90b01d49d273bebf2d22f26f487bc7991a0ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 05:34:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ddc6b4a5bf6c410dad79ea1e2c63a3ec : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/101eea80ba374eb98cd8b7ccf600682c : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e98eb3d245244d0ad2d2903076be1f0 : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226b109df49e406abb3e0add55e7a5ef : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2f5c0ccee8415b92894b02e0b04794 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ee77561fa66489391695e78ab96b6c0 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d1a445c0b54ca1b1a48dae9c4b101d : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7cdee5376ce4f21b48863c77601aa6f : SUCCESS in 4m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/80231c7bf0ae460a80b005def5788151 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f61461be20f4470e9e1554c8a62a8a49 : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6bd09c84149f47a28f743ee01764c626 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2173c503ba86490b870dc1cac928fcf5 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f3bf48e8df44c4bb75fc87c75ab793f : SUCCESS in 31m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5dca2863a4c3493d843c4eda21098cea : SUCCESS in 59m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74e91d5b7d0b4fce9f50f819b9544b0a : SUCCESS in 1h 01m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e1268a966b3a4c22a3422db94e618588 : SUCCESS in 1h 41m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4af9a8ff08cf42a58eaf1419d82a1e61 : SUCCESS in 1h 16m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4dd626851dbf4daf9a09ca1fd1b239b8 : SUCCESS in 1h 29m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0fcb038999342cb8ea8cede24783562 : SUCCESS in 1h 16m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d107a09f21e4473b2e7be22ef79e4de : FAILURE in 41m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5f0ded5a71e74fb381b089c0fdc77e32 : SUCCESS in 1h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30e982787c5a6db6dead456b11d0087b7c131812","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-19 16:07:53.000000000","message":"Patch Set 2:\n\nRecheck","accounts_in_message":[],"_revision_number":2},{"id":"58a7e8b26912c0cfd230946550f0ab237fb47bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 17:09:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c924a92e5bd541fb8e770cd450a94da9 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4976a4997dcb4dd3ae75f791f07d3d18 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15095bac42e549ca8653e781f0490a3c : FAILURE in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3599cd83df9f4c54a887ae985484f5a2 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd63bfbf3254b8aa5fe999864136b80 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8b42aaedbc349719a17002a76f80a1a : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a414dfb966ee4d8b8043e48ce4b96f10 : SUCCESS in 11m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0e4601d84d24d8d9bff90edf068c264 : SUCCESS in 6m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b94b24f6fa614c949506ac0eba8f1245 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b873370de26f46318a76d9c45a05dcdc : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f726d32b9d747ed80172008fe68c19f : SUCCESS in 40m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cca33c9048844fcaec16a29fe0e67d9 : FAILURE in 9m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c5b2f3470514792becfbe7fa9648a0d : SUCCESS in 57m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/618aa548b7744ae28289a4df4aa903a3 : SUCCESS in 42m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d98cd2f476f34deeb6e70fc7ec2b43c0 : SUCCESS in 55m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a57e009c58e441fa6b641285fd21498 : FAILURE in 34m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b5e64c4411994953b72c1239a9c082f9 : FAILURE in 10m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a07b168488045b9a05c804e5276fa92 : FAILURE in 12m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/831ab87d3eff48d6b0f96714ad7d5d69 : FAILURE in 14m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/766cae4578ac4058972d75ae5b3870a2 : FAILURE in 14m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7ae7248dfa944415a3c61f8a8ecea9d3 : FAILURE in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d3f6623c5ba4e2dc70f0edf7141681a4fbd79c9","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-20 19:40:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fc8047804fb077beccc8e08467c7d15fc50c5aa","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-20 20:08:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7681809151531ccec77fdb96587c02878d8f50c0","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-20 21:25:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d738d3753676a9c9df6cc0c4306f568b1505dd7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:38:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8b7f4a967fcd4a0091a01745aa891c8c : SUCCESS in 7m 51s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/632751d226744b55850afc3a4ac5a1f1 : SUCCESS in 3m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3100654323a84ee08a33d5bfcb381b7c : SUCCESS in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84de9ce6fd76483ba239e5003d54a846 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f08f0ced54f493ab076285dd098ef27 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386eccd00ad849c08d3ec01ceef5691b : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed2b57891adf486ca1df84d7f9c25e02 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be73f227fb8343c097af69fb669eceda : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662d3a5d129744d395366ff4e0033360 : SUCCESS in 18m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b556233299fa4fdc9a35294328797414 : SUCCESS in 10m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30c22a1d31834d2297b67aea67c548ed : SUCCESS in 9m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/510d342eb9a340b09ae5f0a5d1411174 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9991ba9374840b899e8d28c7547e006 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/331934b1d60843399f8d2fff17582c54 : SUCCESS in 40m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6d5c9d0565a4065940931cf560d6a83 : SUCCESS in 52m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd971ac99a4246e4838c5066c41f848a : SUCCESS in 39m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a3ec0c1eef2141af9b897747c39fe3ea : POST_FAILURE in 48m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6f5f76e51084122a6f1d3bac7ce1281 : SUCCESS in 42m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/241e1dcc8b9740ea970567fedf4d05cc : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f1c18fc874844dec99e1cf8bbaaca78b : SUCCESS in 58m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/12e6cddae30b432c9809f23e63d32a3d : SUCCESS in 1h 08m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4c11030570fc47fa8c485560a099b283 : SUCCESS in 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5428e31b95d4457822d75654803cc2a : SUCCESS in 54m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3249027f6e9141e793d6ab7a212206f2 : FAILURE in 1h 06m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1cea5db314191202dc1a34fa4f83b4bfb9e4021a","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-20 21:43:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b17ea5814d7650aa61c180d3b1b390e148e1f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 22:53:34.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/716a2d4597ec40b18d3b3360d92826bc : SUCCESS in 4m 41s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e8a24672e8754c7482c9aeef531e3dc2 : SUCCESS in 3m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75e3991e466e4ff0b678dfbe2ad4f7d8 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9278102b3e0a434a89806b56a2df838b : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c7f63fd319d4f9b983c8f9fe2c54e34 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17186aba75245a98b6ed37fb1f5efce : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32f0323f567845cca9f572f0dfcbf8d6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c0e8ea75bae4537a7f785711c99b876 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e463bd269b94f06a50f892404d07f29 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5729ffc9584455fbbb797aaf6278002 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1529ec224e1a4fb5ba0181fd65d669d9 : SUCCESS in 7m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bbba068e0c844bf7813487e5d3fb4827 : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b5b4b6cbc4b4d9e9d637bac76c6a122 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9dd17551aa44412fa47c691e2bc8a451 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3a3511f53d0441292d12f3869ab4228 : SUCCESS in 51m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e6358c999aa4b3086e16f8178b11216 : SUCCESS in 45m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e591adea42cc402d95c60e38c515299a : SUCCESS in 33m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14799abd531b4e85adc9c9e6ac7546c8 : SUCCESS in 38m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60d87bc0e6e640b9b9d17f87a975877c : FAILURE in 34m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/564d8ee0984b4e6aaf48a84f1b927ffe : SUCCESS in 58m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc8bb2628e98451b88c431c1d9cdf5cd : FAILURE in 26m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/219002cdb1784d3ababb9573497e3bb7 : SUCCESS in 56m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d91fdd264b514ff68880b1d3120999cd : SUCCESS in 1h 02m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/225bd3efdad248a0951455aa25d4746c : FAILURE in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16fed4ef3592ba2f3f35e4d754dd5b32a6917b34","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-20 23:16:50.000000000","message":"Patch Set 4:\n\nRecheck","accounts_in_message":[],"_revision_number":4},{"id":"a6f052ffa71189e6157075aff96e199c728a8399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 01:14:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/746276db490e41809f873881c6d64b46 : SUCCESS in 8m 19s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad604cf46b2a49afa746b49e8e6c124e : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb0cb187590f436ca6f878860b35ef56 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caf09530b87c42639561bdee9976fe46 : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebbc336c9e3e48a2ba44151137cf2fe4 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f59472d687428a9ac0aa51a0ef529d : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85baa14156f847339532b5c3ed140da8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f738b15cd08c4a69adfcc2bfed9ca99a : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86923c9aeed344ac835a9d95229a2126 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3148b665bef0441f9c520a367f47be75 : SUCCESS in 7m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/20f7fb588d304f9faf3dd70276eaef69 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c851ecb6df744c9a51c0e566cace8e9 : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3dfaa0d120284c608eb96217a60f4ec0 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c81152690a04cbca54672d6b33bc872 : SUCCESS in 48m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b456f66e30c4919b67fc84ce863d2f9 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fbd3c8e118224cc08fe2c5c639853c2a : SUCCESS in 56m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/166b707b4bf848a481eb7eb5a901199d : SUCCESS in 34m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0b0697f802643709656fa575d56ff92 : SUCCESS in 32m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b2982c7dbde4e99aa47dcf59a4376c8 : SUCCESS in 1h 41m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/647e69b90bcd4afb91ca9de727b99c57 : SUCCESS in 1h 28m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d719d08c82148c0823b7955816ba315 : SUCCESS in 1h 28m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b552b052aaf4b15bb6e7f0edbfdcd6f : SUCCESS in 1h 22m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd24ee4f72a54eee842072902a58db9b : SUCCESS in 45m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fb13da9621c14449b188fbd20ae30704 : FAILURE in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f23861a41fe9f80d91f19a95760e3fbb59d3686a","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-09-04 14:11:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8a1cc16a9df3fee828d3ad0fb4975c1a94cf6c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:58:31.000000000","message":"Patch Set 5: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6df08dd4b8549e382cefeac57570bfe : SUCCESS in 6m 25s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/14d757f9f4f64a31977da6866b6a61cb : SUCCESS in 2m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2b60e266b1b249d29442345d240e88e2 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6268ac4b20a143db83a37f5c1f149f9b : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68ab47f5a92c44b98a6f3765b8363bb7 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa61b4b30d524e00a4886c8649d015e3 : FAILURE in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d35b27f82b462fb0746e99c097551d : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d169588e8b84b7a91c1741af9a4672f : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a35e0b28cd43048816b3530278dbbc : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97214461a77445ffb3c9e2d0b2da8aa8 : SUCCESS in 6m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a0e807ddb6da4bbda61906778f6ffed8 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c7688e8836a2498eae6e8dc608be3c20 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d63dc05a22434b27893c65ff14e33b6f : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97308a655e3148d5ab0e33c1d8d56ac8 : SUCCESS in 46m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7a4a24a19cd4cb69ded801ce6ab6742 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4d0fc7e2b974857b132e09776b25a6e : SUCCESS in 54m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8e169cdf62b482a95e737e05020c271 : SUCCESS in 31m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b887794d0d9430ba8e24366711b5564 : SUCCESS in 1h 05m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb1710c1894e4c31918a6e0d8e09679f : SUCCESS in 1h 08m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e39d3bc0b46146b3bdb40b9c8cd3b0a0 : SUCCESS in 1h 04m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/76234bad56bf42cea6b4f737437303ef : SUCCESS in 1h 39m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6df2587c7337403ab660c28a3bf663ab : SUCCESS in 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f29bd1196ec482abed4a95a6e8ac9ec : SUCCESS in 1h 04m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/097524c341e84728b3ce836dbf0677f8 : SUCCESS in 1h 17m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b75c952c3beb489e9d690d410e929bde : SUCCESS in 1h 22m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2066b706e2f5187cdc9a3dbec68f3a50fc523ec3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-07 21:47:36.000000000","message":"Patch Set 5: Backport-Candidate-2 Code-Review-1\n\n(6 comments)\n\nThis needs a release note as well.","accounts_in_message":[],"_revision_number":5},{"id":"9f51f98f7a6c32414ad87a03cd6dee62374a8c81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-07 21:48:03.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f4b76bfe6bdf983e71ecbcfd2e461f79d55c5359","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-07 21:48:26.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review-1\n\nPulling forward the -1 from PS 5","accounts_in_message":[],"_revision_number":6},{"id":"387bcb1489b0b3387c49f7bcb6ea09d2f90e0cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 23:15:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/926b95eaa6684bc38d04173fde55e28e : SUCCESS in 6m 34s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/08e9ef958e1e42889a2b3910646c0649 : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1792aae8adab4a6f8249f715a2bd4ba0 : SUCCESS in 2m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f057b3968e9b4709b6707fb992f51474 : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/852cb939c341479cabc93f6474767c5a : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a608b2a428c44d9b37c0625d7a67774 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5571aaeff4c46d48f799259c9d30527 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d63d07c1bcc3428a859bddda997660b9 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576a4860435a4ea78c7a373220794a6b : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea605de9376d4cf8945b25ad74a045d4 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ab3603de004b4c948c70af695d8919a1 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c5d7153153443069a314a25c30ac2c5 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5927f8669944dd58c7184fce297d0bd : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/934c8a36fa034815a9f71cfa38fdac09 : SUCCESS in 37m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d08a3268e254e11a77c0e864974b288 : SUCCESS in 52m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df038ad2f74441a0b67f36b3f45d7f3d : SUCCESS in 38m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a78054cf88ed428c960c0d903bd918f8 : SUCCESS in 36m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92fc7f90e7d84739a8737766042b8039 : SUCCESS in 37m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2704fd97611437eb90e3f3f14acc0d1 : SUCCESS in 1h 25m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/126a1f7e10ba46b8910cfbc2dca51869 : SUCCESS in 55m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/34fb797bac5c48f5b2ab4c5db65b2050 : SUCCESS in 1h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/00f24f0dec65401aac7edb162353a0af : SUCCESS in 1h 00m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a833c5b43f94b78b8c00dca29b1bbe3 : SUCCESS in 48m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9159db66e4244b1dabb492350431d468 : SUCCESS in 42m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a84305e440ab41c0827cfdd4e1b0da0e : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d400919bb34479379ec4b462ea0b410b9cf409ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-17 15:37:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"86a2292d62831f248ee575730f2e4c4f677645e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 18:28:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d3dd50b783dd438cb8a8f6a4947b529d : SUCCESS in 5m 30s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/17d189fae9f245f6890a846d24575539 : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/097637e85cfa412f851017d7ef3b2f6e : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb8bbf1cb73d4cd5a05d2e83a4985f96 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/101f314ed47642458c9405c8345bbcae : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b350f466c124fdf97838e97f4e69772 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a321b060c8be4fd2863825c84f9d0bc0 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dee7a51d472f455d865ede85f6956d25 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87630cf5175f43edaae985c1f2bd38d9 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc7cfee02f849478a25f2485af25e41 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10029ae1821c4d59bc3042a4ef9be7bc : SUCCESS in 5m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9fffa3e6b0424a9bbb2c36c26845a349 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9ce22d96aa39497aaed12547ba8bcd96 : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1962cb5eec8b4ee69f781d74a15f9b31 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8441bfe540664fc4a5f716730a12c26a : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c83fd0a916a742bcbe7b166d54c4bffb : SUCCESS in 1h 38m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbc080df3dad4ebaadabbc88eaae905c : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c761443cf2ad485daac1cef530203468 : SUCCESS in 45m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/55f30207e21b4cce8f43c61521d54f1f : SUCCESS in 32m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/413b7de31fd74ac7bd41604a747b728d : SUCCESS in 1h 08m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85660e311a2c4a62987bfb492f1f6248 : SUCCESS in 1h 08m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/11722131d6694870acbe50e453d106f2 : SUCCESS in 2h 27m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b65489919a1645c595199147777fe8c6 : FAILURE in 40m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3d04dbbf24c24948a4d5e91555a32b4e : SUCCESS in 1h 35m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c91f19a069546969454800479d03fd2 : SUCCESS in 1h 58m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bc5511847f6d4c9fa792f66be2a7389b : SUCCESS in 1h 16m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/47fe7f4090734568865537562a219b4c : FAILURE in 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"91bf59ac378f1352e2c78d844b8174ec18018823","revisions":{"dfe5d4aaa25216aa1058386a978fde6b4d8ff69e":{"kind":"REWORK","_number":1,"created":"2020-08-19 00:49:12.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/22/746822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/1"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfe5d4aaa25216aa1058386a978fde6b4d8ff69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfe5d4aaa25216aa1058386a978fde6b4d8ff69e"}]},"branch":"refs/heads/master"},"b81ca7e0e536f6bb1470332af38a0f66e010421f":{"kind":"REWORK","_number":2,"created":"2020-08-19 00:57:37.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/22/746822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/2"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:57:19.000000000","tz":-420},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b81ca7e0e536f6bb1470332af38a0f66e010421f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b81ca7e0e536f6bb1470332af38a0f66e010421f"}]},"branch":"refs/heads/master"},"9de81f29dee0123df5e2abe234b5b5d50d1a7996":{"kind":"REWORK","_number":3,"created":"2020-08-20 19:40:02.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/22/746822/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/3"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-20 19:37:22.000000000","tz":-420},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9de81f29dee0123df5e2abe234b5b5d50d1a7996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9de81f29dee0123df5e2abe234b5b5d50d1a7996"}]},"branch":"refs/heads/master"},"7c278c49f4014c37ad254c4e50a26654bb70f786":{"kind":"REWORK","_number":4,"created":"2020-08-20 20:08:52.000000000","uploader":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"ref":"refs/changes/22/746822/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/4"}}},"commit":{"parents":[{"commit":"b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3443a81cf78c5b4bdfc93db4ac1772e4fc99b12"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-20 20:08:33.000000000","tz":-420},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c278c49f4014c37ad254c4e50a26654bb70f786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c278c49f4014c37ad254c4e50a26654bb70f786"}]},"branch":"refs/heads/master"},"25f17952024658e41445d9634b133f9e519f933f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-04 14:11:55.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/22/746822/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/5"}}},"commit":{"parents":[{"commit":"0e6a6532c3ecaf1b490227add41b8f552b5467ea","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e6a6532c3ecaf1b490227add41b8f552b5467ea"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","date":"2020-09-04 14:11:55.000000000","tz":0},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25f17952024658e41445d9634b133f9e519f933f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25f17952024658e41445d9634b133f9e519f933f"}]},"branch":"refs/heads/master"},"e7e23724ca702e6306507941388169b7663114bf":{"kind":"REWORK","_number":6,"created":"2020-09-07 21:48:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/22/746822/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/6"}}},"commit":{"parents":[{"commit":"0e6a6532c3ecaf1b490227add41b8f552b5467ea","subject":"Refactoring amphora stats driver interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e6a6532c3ecaf1b490227add41b8f552b5467ea"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-07 21:48:00.000000000","tz":0},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7e23724ca702e6306507941388169b7663114bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7e23724ca702e6306507941388169b7663114bf"}]},"branch":"refs/heads/master"},"91bf59ac378f1352e2c78d844b8174ec18018823":{"kind":"REWORK","_number":7,"created":"2021-04-17 15:37:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/746822/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/746822/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/746822/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/746822/7"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"sdjajadi","email":"stephanie.djajadi@gmail.com","date":"2020-08-19 00:48:01.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-17 15:34:18.000000000","tz":120},"subject":"Create InfluxDB driver for statistics","message":"Create InfluxDB driver for statistics\n\nChange-Id: Ie51ba7c2a7b1895613565d397b0354b2e7e60bd4\nCo-Authored-By: Anushka Singh \u003canushka.singh.2511@gmail.com\u003e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91bf59ac378f1352e2c78d844b8174ec18018823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91bf59ac378f1352e2c78d844b8174ec18018823"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
