)]}'
{"id":"openstack%2Foctavia~747201","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I49e92f81b23e0b306c62c406a45029b96dce20df","project":"openstack/octavia","branch":"stable/ussuri","topic":"sec-grp-rule-quota-stable/ussuri","hashtags":[],"change_id":"I49e92f81b23e0b306c62c406a45029b96dce20df","subject":"Increase the devstack secuirty group rules quota","status":"MERGED","created":"2020-08-20 15:07:24.000000000","updated":"2020-08-20 22:03:53.000000000","submitted":"2020-08-20 22:01:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747201-1597960894154-44e8c23f","meta_rev_id":"bdaa5c987beca7c33f198df0d20db8023720280a","_number":747201,"virtual_id_number":747201,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 22:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-20 15:08:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-20 18:09:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-08-20 15:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-08-20 15:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 15:08:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-20 15:16:59.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-20 18:09:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-20 22:01:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24f32b70ad32707db347d25f6f29f926d5d734d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 15:07:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0c2d4b22e0fd5303232bfdd3009677a3bd33b6ff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 15:07:29.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 40777222b91028412c8b2a96230ec284d99620e1","accounts_in_message":[],"_revision_number":1},{"id":"67e11f55455137ed227f7d1819b9b5f1bd41f5ae","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-20 15:08:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dd681f31f453aafbd92806afb2d04516523f7f8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-20 15:10:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81ed5a1481f7316bf6c1c2086714500196a491ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-20 15:16:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68a68e70296d6992c3d79bda1bdcb6b806820f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:48:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ffab05904194a55a581ac17641d781f : RETRY_LIMIT in 1m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf601c29e0d4538a1602ddcaaef9dcc : RETRY_LIMIT in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29e1050bca44274ae9ba240a9f650f6 : RETRY_LIMIT in 2m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c0895f4481f4fe69b6d48a45aebe7ee : RETRY_LIMIT in 1m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98bb1b833655489fb960f8a1d98505cb : RETRY_LIMIT in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197c8bc79399412eb4299dd11a6b853d : RETRY_LIMIT in 2m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe740322544b43ec9ad0ef5b5421b661 : RETRY_LIMIT in 2m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24fe1c530bca4f4995cbfa5049126673 : RETRY_LIMIT in 2m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06f24ff7c29f44868580ee33d3bf952b : RETRY_LIMIT in 2m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97f0cb7cded141eb9819958c6e33d2d0 : RETRY_LIMIT in 2m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1aeba75f413a449ab854d1660e3295b5 : RETRY_LIMIT in 2m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70a69c65766e4f7e8728a664fe6cbcb4 : RETRY_LIMIT in 1m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d5bd061890004d5f8680817a797d6521 : RETRY_LIMIT in 1m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe59b2fe5449452faafe6f07f9e8aae1 : RETRY_LIMIT in 2m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/39e6052427bc43f7b92031161eff6532 : RETRY_LIMIT in 2m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/615f911c9bcf4b338d17478946830348 : RETRY_LIMIT in 1m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/005c3f0126c34f18a1b99e4c41607063 : RETRY_LIMIT in 2m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dc42ff4ef1d249369c7b94a2331de215 : RETRY_LIMIT in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5237b2ef0275727c1f7ea7a5e423c94367386f0f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-20 18:09:17.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b3d0b36186aa9f831f8a8e82aa9ee618d8208f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:10:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67f2ee156e1b450ea1df7db72e09b981 : SUCCESS in 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9f1e0b4d58d4d9aaea61c01eca9dc05 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1cb15628564e558e6e4c848453e6a7 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/675094f48ef44148950b1895bc2ef360 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/421c1288d0d94db881bb209192720363 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91239b641e24069a15934a12825c718 : SUCCESS in 13m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b562df6972974439a08782b4b7cffa8b : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ba1a53cc807543e394ae9da3b97d1057 : SUCCESS in 9m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6b7855c162c4528a3ea21a06ccd2e68 : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ee59ec2a7f44fe38ad4d72371446cbb : SUCCESS in 42m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a1026f617ab4ba9acf5bd4cc58291e2 : SUCCESS in 43m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d0c920861cf4e3a8ea4bb5961b9019f : SUCCESS in 1h 00m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2efcaea47e5b4df59d57ec79db4fc433 : SUCCESS in 55m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e11823c5a7cf456da1ee637e01457ed4 : SUCCESS in 1h 09m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c26c7f6f746147e29ac9bb9402ed56fd : SUCCESS in 1h 52m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/79cb42fb1dc54e0b828b3a18221a0114 : SUCCESS in 1h 35m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bfc7cd4fcb1946af8e6e4a568718ec48 : FAILURE in 58m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/515f0f4d75924a16b6f91a201118480a : SUCCESS in 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"687dd62039bea3cdea86ef982cba8fdc4b772adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:10:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd8fd96aa142ee3597c9ac490b18245db9923e10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 22:01:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a98af1aba9aa4c62abe11e456b17e4b0 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df2b4e66377404da5ed972fc94a9948 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aee4e2b3ff1e419e808e837bc755406a : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cb67d3e7a4943148e3ec155ebf8c853 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b47f9eedac4396bf1d88d081212913 : SUCCESS in 11m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/013c2fcbe06e4df48fb187dabcdce3bf : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f191f8a647a048bf8a1f11c064d2784a : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/586d01127aad4e39a18e6acb1d27554d : SUCCESS in 49m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bed9a5ee50394bf7971afede8afd46af : SUCCESS in 44m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4730412f7446473b966bc3c2415efe5a : SUCCESS in 42m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02a8378f4baa4d43b053180a7e1b61b0 : SUCCESS in 40m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59ff7a894ca34296bb6b34beec6a5c6a : SUCCESS in 1h 45m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5de1329f25c0fefe53c446c6a88c7a2f6423f4f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 22:01:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"98f9e9733724c2f00ba415556895c1adb7eafc95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 22:03:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06b500b5be04472abfccecb9d864245 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17c22145443ab2f8456e1ab4d738861a6d3b8d7e","revisions":{"17c22145443ab2f8456e1ab4d738861a6d3b8d7e":{"kind":"REWORK","_number":1,"created":"2020-08-20 15:07:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/01/747201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/747201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/747201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/747201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/747201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/747201/1"}}},"commit":{"parents":[{"commit":"f923ef093424cb05e446c490e86f648f51a52e4f","subject":"Switch back to using git for the agent ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f923ef093424cb05e446c490e86f648f51a52e4f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-16 20:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 15:07:24.000000000","tz":0},"subject":"Increase the devstack secuirty group rules quota","message":"Increase the devstack secuirty group rules quota\n\nRecent changes to the Octavia tempest tests has caused our test\nruns to exceed the default neutron security group rule quota.\nThis patch increases that quota for the Octavia project in devstack.\n\nChange-Id: I49e92f81b23e0b306c62c406a45029b96dce20df\n(cherry picked from commit 2e2464d778cc86df161fdf59406aac30770b58f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17c22145443ab2f8456e1ab4d738861a6d3b8d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17c22145443ab2f8456e1ab4d738861a6d3b8d7e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
