)]}'
{"id":"openstack%2Foctavia~747202","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I49e92f81b23e0b306c62c406a45029b96dce20df","project":"openstack/octavia","branch":"stable/train","topic":"sec-grp-rule-quota-stable/ussuri-stable/train","hashtags":[],"change_id":"I49e92f81b23e0b306c62c406a45029b96dce20df","subject":"Increase the devstack secuirty group rules quota","status":"MERGED","created":"2020-08-20 15:07:29.000000000","updated":"2020-08-20 20:31:51.000000000","submitted":"2020-08-20 20:30:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747202-1597955404795-e6cdcfc5","meta_rev_id":"3bcb5a0e6c7ee4d213232ffa63594995afb96d6d","_number":747202,"virtual_id_number":747202,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 20:30:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-08-20 15:07:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-08-20 15:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-20 15:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 15:07:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-20 15:17:02.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-20 20:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7134bfe739c7d4b1b04a208099d0b76d761e3084","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 15:07:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"9f256186151045efcf04b3071b0df426b3056d5c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-20 15:07:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"174bad90465cccd41541f4e55ac2d3472ee46ea9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 15:08:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8","accounts_in_message":[],"_revision_number":1},{"id":"e564ef598f43501f7af5922333093822bb063014","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-20 15:11:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9774f4f6aa414b19197421f7fe77a545565533e8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-20 15:17:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35f6e49b90c20eaf3f2da5c1167da34812d58065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:38:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e78d1e849bf94737a2412df8b275e508 : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/746d5930850d43bba35cf917d8af8503 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bfb5d37ffbe44ea930ed9c62f580260 : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f3e18354e7643fcb823c33141e75fe4 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db6a6ddcd25c4e29b3a51165884b89be : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f57237783884c84a65dd079199249cd : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caf20dd175d46daacfed78496dcb115 : SUCCESS in 13m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5ac171e05b284cb9b89ebaa4da65a1d5 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/47ffdfaad19344cb87df2f27197dbe09 : SUCCESS in 9m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da3e6d848cd449f5b0b7953544706e14 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a987063ad344e559a8b0789646a7cb6 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9a1189537c9409a805e42c530c2256e : SUCCESS in 41m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/84e637a3b63a44438f5cf4d324686804 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/381eca6c086a419990959a76e5eee81b : SUCCESS in 53m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a50948128f1b417e8b00905953c3210d : SUCCESS in 57m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/efd2471349944032b11adc99e0c2814c : FAILURE in 10m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea53b4528a084a0596ef3af3f43b67d4 : FAILURE in 31m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/280713ffb67c4cf49f3e482f878bfee9 : FAILURE in 45m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/296ef4f27fe247dfa666cd60885f7e0d : SUCCESS in 1h 12m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d54c1deaea4e4bf7be23d3e0497706ec : SUCCESS in 45m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7bc0d41406b0413eb8ee38b9eb12c568 : SUCCESS in 58m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/114ac0f09c774df98f2bd2988956b214 : SUCCESS in 41m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b331aeb73b28417a9c58d93d1ed9459b : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c196ceb5bb7848e2bd103c3744bfcd30 : FAILURE in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67328eef5145153805acf4ae0283048280306e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:42:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"770c400c42be2e490d4e0161ffd0978af87b687b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:30:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/407a405dc1234ebda7feccdd2d5a5077 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ff1477ca30426c96b51a6419c94aa0 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/389d01d8d76d49b9843faff8a900ff43 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284e09f618c049baba6b2ea0929c6c1f : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7be58490d9224e84ab68a6451fdb0a4d : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b673f93b75424ec484f93f0c64da6753 : SUCCESS in 12m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8ca223bbaa934808bf786c9baa38f405 : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ff92a3604a34b608293246a1862fb81 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c43b3932b6734cbdb972177ec0c5e9d8 : SUCCESS in 1h 01m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6c18b2742c3841c99dcdc001e07e7fba : SUCCESS in 1h 02m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbb8faeeac09492d88625795ae31c8ab : SUCCESS in 1h 13m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/26c2b05d3e4a40229e81bec33f456f16 : SUCCESS in 1h 18m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6bd6cba6fd024fa6a61c8cbcf8133aad : SUCCESS in 1h 41m 55s","accounts_in_message":[],"_revision_number":1},{"id":"caa7a4b8fb4e3b6681d586f57ff563c3c22529ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:30:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d1833e4db49e9c22888bf0b0afca57fd80d61b9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1393522eca54e28973df32d484b8039 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40777222b91028412c8b2a96230ec284d99620e1","revisions":{"40777222b91028412c8b2a96230ec284d99620e1":{"kind":"REWORK","_number":1,"created":"2020-08-20 15:07:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/747202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/747202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/747202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/747202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/747202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/747202/1"}}},"commit":{"parents":[{"commit":"1bafec3c0465521ebd71012398cb7641be357dc9","subject":"Merge \"Fix the interface filenames for Red Hat amphora images\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bafec3c0465521ebd71012398cb7641be357dc9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-16 20:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 15:07:29.000000000","tz":0},"subject":"Increase the devstack secuirty group rules quota","message":"Increase the devstack secuirty group rules quota\n\nRecent changes to the Octavia tempest tests has caused our test\nruns to exceed the default neutron security group rule quota.\nThis patch increases that quota for the Octavia project in devstack.\n\nChange-Id: I49e92f81b23e0b306c62c406a45029b96dce20df\n(cherry picked from commit 2e2464d778cc86df161fdf59406aac30770b58f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40777222b91028412c8b2a96230ec284d99620e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40777222b91028412c8b2a96230ec284d99620e1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
