)]}'
{"id":"openstack%2Foctavia~747203","triplet_id":"openstack%2Foctavia~stable%2Fstein~I49e92f81b23e0b306c62c406a45029b96dce20df","project":"openstack/octavia","branch":"stable/stein","topic":"sec-grp-rule-quota-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"I49e92f81b23e0b306c62c406a45029b96dce20df","subject":"Increase the devstack secuirty group rules quota","status":"MERGED","created":"2020-08-20 15:08:07.000000000","updated":"2020-08-21 12:09:51.000000000","submitted":"2020-08-21 12:05:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747203-1598011542983-0b925b31","meta_rev_id":"81e255a2eb045c3411c69c60b9af150e4a0101ca","_number":747203,"virtual_id_number":747203,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-21 07:55:30.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 12:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-20 15:08:23.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-08-20 15:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-20 15:26:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 15:08:23.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-20 15:26:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-08-21 07:55:30.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-21 12:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a13f6791dcb2b66257c50dd3c33700167b000b20","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 15:08:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"554c81094662f063340157f4381a4787e3fb5bae","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-20 15:08:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"657b2a0effff0898a84d6d6d1be2fb4d9a4aa561","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-20 15:26:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b13d61aaf3c5c64f3c2d0979a0d8977d5ed9a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:59:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deb0715e6ed149e4abcef1906f1a3437 : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a57f02ab0ea45d985ae4f614500d05d : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd13cf675129413e9e2fee39c308e377 : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e24b399d22614b85964d59ccbe2a93eb : SUCCESS in 8m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c46ac5d6565d4c729064d55e761eeb24 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0df0559311c4d11ac17cca26fbf87e0 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d68ba739f0a4c63a0b0d3edf82764c6 : SUCCESS in 7m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/99d27314f348492baab2ce5a42437938 : SUCCESS in 1h 17m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e1f1eddd206548e29fc44f2301c84bb2 : SUCCESS in 1h 20m 01s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c75aa975ffa84ba7938f971b4ee5b7c7 : SUCCESS in 1h 28m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e578153da68a45fc8e58e2537480793b : SUCCESS in 1h 13m 35s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/038a902387d04f14a5bf4645a353382a : SUCCESS in 9m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4ff392ac36ca4865b90e16cb876cedfa : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e09b623f46344ac79c816b6f2e83c236 : SUCCESS in 7m 20s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a98a4a13574f43d9b7faa0b5cf2d724d : SUCCESS in 1h 06m 53s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4217714765664b74bc9754880bb09a71 : SUCCESS in 39m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/71d80470ce774d9eb5608d54041774dc : SUCCESS in 40m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d9fde55d502542d78aecf689be6f42fd : SUCCESS in 47m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0efe6bd332f1408e85af067d59d5d2df : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/30298f4476f14175aba641bb576d2105 : FAILURE in 1h 43m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d71fd78b2336411c8a636b1cdff0f103 : FAILURE in 11m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f27ce38956194f40ab3273b2103c5933 : FAILURE in 23m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/489e7f4d2c574451a48741b3e909480e : FAILURE in 51m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/63ef2717735a427ebedcdfe64e8c8713 : FAILURE in 46m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/066ebc085edf4bb0ad74e105862454b4 : SUCCESS in 1h 11m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26020df2de9f4c808be08843ca3649ca : SUCCESS in 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e558eb02b48994308a45e6069392b13dc93693d2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-21 07:55:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e6502379d62660499ae633c782ce5152cba255f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:01:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d5f79d769c94a26aca5328a947dcaae : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f58457201044eeca51bd12d340b57f7 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb04ccfa52724f09a2cf465a5b0f038e : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f66eb66fc3ec4ffe86e0e49536f2dfec : SUCCESS in 8m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e38c9ac25f384878b88e55cbd68e51c4 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc2a944c6d854652a1a667aa72eb05ad : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9121664ef4b48ee816a53776c9159cb : SUCCESS in 6m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2fff29659a89422abef127ec2075f3bd : SUCCESS in 1h 16m 58s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/169ad551cfe74f2ebca3abf88127358f : SUCCESS in 1h 17m 52s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41b8e90f2c794999a9e8147654db0520 : SUCCESS in 1h 15m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9f3fcbd8c1924af69539c33a8152bd0a : SUCCESS in 1h 16m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d84224833e44e12b95be4322da9125d : SUCCESS in 7m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/651df9db3a7e42a188dbcfa914c8498c : SUCCESS in 10m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a0f697ec8b444d6a68752e659fb634e : SUCCESS in 8m 15s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bdec7aa51504c44bfa76f7fbb53d593 : SUCCESS in 1h 00m 20s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/78f00664c3a244eb8e5472282172ccba : SUCCESS in 59m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b41ecf4c41314f49934bd437b2908eda : SUCCESS in 44m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9223d13e8327470093598a8be6292a72 : SUCCESS in 46m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d4642d72dac430085f87a1adef82450 : SUCCESS in 1h 10m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/09f6470e51c146f188cd07939af20d9b : SUCCESS in 53m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3bc53f33f8ee48f8b8465e6b5d7939be : FAILURE in 13m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c6a6bfc45334825ad295280dfa801fc : FAILURE in 36m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/174411261e0a4d5f948d1938141a17a0 : FAILURE in 55m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd4c57120373433b9e9b0719ec7b17d8 : FAILURE in 36m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b802f570f0b4cb28e56946f54ddef6a : SUCCESS in 1h 09m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b9fee1b82b54798a3077d5498a238f9 : SUCCESS in 58m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfe672c3f4baa6750a620122a24ede73e59a2e7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:01:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a75792a26a3db6eb1d72847a9d7cb58a12646277","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:05:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"567eb902763920ae4d262a60b704432bf8c715a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:05:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c286bc472b94ff69bd6fb23a6408687 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8af262214a43d2a6907157c7b23ae6 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a948acbbe61e41af8bd538cdbcaaec04 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e430b339b17c403ca11225c9e20f2434 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcbaf1c5efb04a82bc7ed00a2c565877 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738e13deb1fa4edfa8ec4613f177ac56 : SUCCESS in 6m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c23f4d0c231d404e8ab2cf874dc3c6dd : SUCCESS in 7m 27s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/beb15e5bef6a4b3dbbf45a08955398d4 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78d528b576864754a94621e40151c949 : SUCCESS in 7m 03s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d32ff1a665d4a4dbce68890e285014a : SUCCESS in 1h 03m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f8c3ce387be348daad38a1ab32b2f0f3 : SUCCESS in 58m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/775854617f704619a59c951738d6b785 : SUCCESS in 1h 17m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f5d921cf0fec4f8cb0015ef65b3626fa : SUCCESS in 1h 16m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a76b5a80f3a44ed8cb3dc48d8ad6bb0 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3c175c401d7f435daebb79c92edb1466 : SUCCESS in 43m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/702d81e8993041ab9274d6ac73c0f5ba : SUCCESS in 54m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8d4318a11da648b18b03a1725421fe7d : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4649af422618ac81bedd884846c11ef613b19330","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:09:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f15b64df7c0428eb4bac3be39c02374 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8","revisions":{"6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8":{"kind":"REWORK","_number":1,"created":"2020-08-20 15:08:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/747203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/747203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/747203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/747203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/747203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/747203/1"}}},"commit":{"parents":[{"commit":"6b6f73ddf0c41c50164bcb2dbca0ba2ef7f0fbd1","subject":"Disable log offloading scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b6f73ddf0c41c50164bcb2dbca0ba2ef7f0fbd1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-16 20:31:58.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 15:08:07.000000000","tz":0},"subject":"Increase the devstack secuirty group rules quota","message":"Increase the devstack secuirty group rules quota\n\nRecent changes to the Octavia tempest tests has caused our test\nruns to exceed the default neutron security group rule quota.\nThis patch increases that quota for the Octavia project in devstack.\n\nChange-Id: I49e92f81b23e0b306c62c406a45029b96dce20df\n(cherry picked from commit 2e2464d778cc86df161fdf59406aac30770b58f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
