)]}'
{"id":"openstack%2Foctavia~747295","triplet_id":"openstack%2Foctavia~master~Ib4a95483a997734788353b6da2a599ca4c2acbbf","project":"openstack/octavia","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Ib4a95483a997734788353b6da2a599ca4c2acbbf","subject":"[goal] Test out Ubuntu Focal nodepool (20.04)","status":"ABANDONED","created":"2020-08-20 21:32:44.000000000","updated":"2020-09-10 15:00:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"641da9f0e7b71c1e6f09b0d3bc14dede8c911538","_number":747295,"virtual_id_number":747295,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-20 23:18:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-08-20 21:33:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 23:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca73ec18c0343b34a92dbff4f99007a248905aa1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 21:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23871b8c471f70fb264db24f0c3bd711b338c871","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 21:33:05.000000000","message":"Topic set to migrate-to-focal","accounts_in_message":[],"_revision_number":1},{"id":"0ebb1a09a48096059c3d0be37bfdf45a6ec984e4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 21:33:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b49de7c5745d546ccdb77b86a95febf5b7fdc690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:18:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/45372bd9cef549bc8d25b1584c80a6ad : FAILURE in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e85e8e6814049debec0c64dca718ab5 : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f52efeb7ee74467aacc1d0e280483f8d : FAILURE in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a749cc733c4e3b8cc7f5f3eb5d278c : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb8e5795a434fd193eb307a027d2fbd : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/180d98d2c36b46ae87a42a3676a32a43 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c9488bface4a0f8cca3613a5ba2be4 : SUCCESS in 12m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ec60b38455a64784a5f5f3ff326d61b3 : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f5e0356dd6144b69b6281c6ba690e6e9 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b66178c89f248daac16a83eddc26ca5 : SUCCESS in 11m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59a82d7f15784c2e96183fd2cec36c59 : SUCCESS in 49m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d41a87a02c14e6f91d3710f310aa1e4 : SUCCESS in 1h 18m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36ca5da4d5b3409781274f3321a2d6c5 : SUCCESS in 1h 00m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/228a90dec35240cf93981529f6241ae6 : SUCCESS in 58m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23d4390b5f534e87ba807cc995c1a4eb : SUCCESS in 53m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb7385889c114d19bc5eb8922fa6a70f : FAILURE in 1h 38m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e4095b9e4234d76bf8461ded9e9f826 : SUCCESS in 1h 11m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1602552e4d454a379d86a01b8878059e : SUCCESS in 1h 03m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e12ea4f89bdc4018b73eba8b549936d7 : SUCCESS in 1h 17m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/14e083f021124e0f940ddee2610e2bd9 : SUCCESS in 56m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/93a1cd5982e4498ab0de48670e62c5d5 : SUCCESS in 1h 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"641da9f0e7b71c1e6f09b0d3bc14dede8c911538","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 15:00:53.000000000","message":"Abandoned\n\nThis work is now done","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d1a05f3ceaef611294fdb7e363e1053408707ae","revisions":{"4d1a05f3ceaef611294fdb7e363e1053408707ae":{"kind":"REWORK","_number":1,"created":"2020-08-20 21:32:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/747295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/747295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/747295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/747295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/747295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/747295/1"}}},"commit":{"parents":[{"commit":"0a85de4c6d6b5acf95bd99f47810f2ac59b45b22","subject":"Merge \"Switch back to using git for the agent ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a85de4c6d6b5acf95bd99f47810f2ac59b45b22"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 21:30:33.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 21:32:43.000000000","tz":-420},"subject":"[goal] Test out Ubuntu Focal nodepool (20.04)","message":"[goal] Test out Ubuntu Focal nodepool (20.04)\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nBump the lower constraints for required deps which added python3.8 support\nin their later version.\n\nThis only tests the control plane (devstack) as focal does not\nwork for amphora images at this time.\n\nStory: #2007865\nTask: #40201\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: Ib4a95483a997734788353b6da2a599ca4c2acbbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d1a05f3ceaef611294fdb7e363e1053408707ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d1a05f3ceaef611294fdb7e363e1053408707ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
