)]}'
{"id":"openstack%2Foctavia~747662","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ibc525d9a046a5ab7f090a942459d80a2df66ae2e","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ibc525d9a046a5ab7f090a942459d80a2df66ae2e","subject":"Fix memory consumption issues with default connection_limit","status":"MERGED","created":"2020-08-24 09:25:55.000000000","updated":"2020-12-02 00:06:05.000000000","submitted":"2020-12-02 00:04:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747662","meta_rev_id":"6765495063e76028d1f3201aced0a01f8826cfae","_number":747662,"virtual_id_number":747662,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 00:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-12-02 00:06:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-12-02 00:06:05.000000000","post_submit":true,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-01 19:41:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 09:36:34.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-21 09:27:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-24 11:22:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 19:41:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3b5b32ef9e30a0866eb3a11a823a609681a20958","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-08-24 09:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff66acf74083373f8acc7a5f490f1bc8a74ca4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 11:02:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2066a7c441854484816e693e9eee91c2 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf8a609ea1548b68d0025be24f77a44 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a2ffd2497b47e9aaa00f2bc6a9ad46 : FAILURE in 10m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9565aed834404186992e7446067065bb : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a9f9c44bda453a92077d08ea35aef2 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b872b59804cc492b9c2693be166b12de : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4851a66cce824d51a984bf181ae3fad7 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d834a5ce3f144c1c95f570c16a2a2084 : SUCCESS in 6m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6d5b8c0a51a1447c8b039de41a060a6b : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a85e394330c9445cb2414e6d1d129636 : SUCCESS in 6m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4023eea1794043e4a226e1b029d7b5e2 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c89fe4cb341547ec8dac5df33641aa23 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c882e6baf00a447ea9c43efeae9131ea : SUCCESS in 48m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e26ab1482ca749c989eb763727924039 : SUCCESS in 53m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62c17980587f4ffc9b50276dcc535420 : SUCCESS in 1h 13m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6aa4b51a81854f2c8b6ff15ed194ee4c : FAILURE in 1h 05m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/31858ab01708415e8164d1fca6bfe230 : FAILURE in 16m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80725afcc8274487bcf6999a1c349efa : FAILURE in 38m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cbe5291d8864dbca9e55901e1b4791f : FAILURE in 51m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/62275ffca6e74d22955c12fc47aebe34 : SUCCESS in 1h 28m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c136f031f288494dbb029e9a371b784e : SUCCESS in 1h 01m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4eb5b8cfc5eb4c44a091b42df7bcf23c : SUCCESS in 45m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f21fcce8548644b9b531a4a2cceb460e : SUCCESS in 32m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/586ea407bcb3402ab9b51ebcaf61d6b7 : SUCCESS in 1h 01m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/02434e3ce20a48a5bccc457a946337b9 : FAILURE in 15m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/common/types.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_updater.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/v2/controllers/flavor_profiles.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/v2/controllers/provider.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/quotas.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/__init__.py\n  Comments left for invalid file octavia/api/v2/controllers/flavors.py\n  Comments left for invalid file octavia/api/v2/controllers/amphora.py\n  Comments left for invalid file octavia/api/v2/controllers/l7rule.py\n  Comments left for invalid file octavia/api/v2/controllers/l7policy.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/types/quotas.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/retry_tasks.py\n  Comments left for invalid file octavia/controller/queue/v1/consumer.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/cert_task.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/cert_task.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/distributor/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/controller/healthmanager/health_drivers/update_db.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/compute/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_utils.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/exceptions.py\n  Comments left for invalid file octavia/amphorae/drivers/keepalived/vrrp_rest_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/amphorae/drivers/keepalived/jinja/jinja_cfg.py\n  Comments left for invalid file octavia/amphorae/driver_exceptions/exceptions.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/common/keystone.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/stats.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/exceptions.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/volume/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/volume/drivers/cinder_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/db/repositories.py","accounts_in_message":[],"_revision_number":1},{"id":"4bc5d2973452eb559ec0acb01f53253c3385b759","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-09 09:36:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ba31204980816f349cbb9027cd84f92cd72695cc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-09 09:36:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b1eff4575b6abde245609acc7b54320fc4cb9477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:13:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1363af2d5344df4add3bd64eb9545d9 : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf1eb2e020c740e8bd33fee6b7b48d0a : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513093fa1a8c40f791f78bb9864aa47b : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3f296479a3445758361c18ebd2738f3 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9fb36736304a3ca13a1f84b1935bb6 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a1f12985a1a4af1b4ef245005d8ee54 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9fc2559dd84d7f9bd1e3bc212f22d5 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96121b9388b14b48a4dfaadd530e1214 : SUCCESS in 5m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/99e9a52a19ea47baae692c3902dda994 : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b62f446da99e4d18b0f9484d9ff3415b : SUCCESS in 7m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a0d0e8daefcd4ac095117b9f667132bc : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/933b853e5ca34408990280ee804d6b66 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e3b13177b844ab9ba82e81135a32bbe : SUCCESS in 47m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/21c239b0314d4ff3b5a3669e80e57c03 : SUCCESS in 51m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/002821d8d743440b862fe176b7fe24de : SUCCESS in 55m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4221fdd09cff4b46b7f160cd2bdaa202 : SUCCESS in 58m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/abd1aab7022e4637b9bd08dd4af7bfb1 : FAILURE in 14m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f0eb79c7d62498ea60be30b8e1d441e : FAILURE in 24m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a682fdbd04b8459884318df7398cc020 : FAILURE in 49m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4fca1f6a739641b280491f527c95a8aa : SUCCESS in 1h 32m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f06b39d849544ba5ac1c79bf06393368 : SUCCESS in 32m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a8e28907ab9c4ad7a46319f24a807374 : SUCCESS in 47m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d97e31dbee9c4960adae5148d9612681 : SUCCESS in 31m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d17e3e3599054b4083871dc212958270 : SUCCESS in 38m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4784d31098e74eb5a220d1980592f381 : FAILURE in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"006c71c3f7e66bb19eff9604d3d990f29590dbf0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 09:27:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"abf03e33962738a6b6553cf7638888545b1bdb16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:27:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"573766a188566a677d35b02cd9f60fb44158cd33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 14:39:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1f1590bf9441b8b08bd33e612cf8c0 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dab5d8913ea486f968bb08aae8c9606 : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab6e6b7cab934ccbbbba833e029b132f : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f270060026894379961bbbb827bc2c58 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25dacace232c4da9b8e269e343119194 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88dd86c08d14121867e24e25f2ac0a7 : SUCCESS in 17m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee7fd7d3e7e548dbb068ca80c4c5b67c : SUCCESS in 5m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ea320d7ca5814695a8541881a48de772 : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e13b5f2774b8431ca6ba99e9bd90f662 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c455e40453940358c5da5cd6cd65441 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/98d957a0b72f4508ad2f6ff4b93755f8 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db90a53d4d374913a8a7876e20e5f369 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/781dfb3223924352ba67705fc4695a8c : FAILURE in 1h 35m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6916ce6753cf4e64bbf298cb22f55133 : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":2},{"id":"087c6ea425af9a0e540f2e0bc9953ee649a82a51","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 14:40:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c20c50fc494cba3b0cbb2f50b977f432dfb3e526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 21:01:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed79e343f7be4e2fbd8a7641ea69a4df : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c09a20e12146dcae2b2ff2a1cb6342 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b022716a294852a4ccc7df26592ae1 : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d44b0496b8f344b79d77fbae4190e051 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c35c1e1daf49fcbd9653e8f5353d59 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb375dbabf864ebb9f543fd32b54a584 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d72def9d9bb4ff0a706d252ad3d13b0 : SUCCESS in 20m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/006a2a684cd34d43953dd5cd7eee61ee : SUCCESS in 5m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6762a3e2205945829415b9a50645d2fb : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/60f39d52df2547adb663d85622540bd6 : SUCCESS in 7m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/089755c212ad4ac495a77068e1a6da29 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62583152cb724dc9bf94f025ff6ebe60 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1adb512bc774550ba91c49e5cad1de5 : SUCCESS in 1h 40m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8fd61810c57c4a8fb5921c2bdb45316e : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6525f55f3f347708b6d7414dae1730f : SUCCESS in 1h 56m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7e58e421f5e54bb3bad0d8da410c0f7e : FAILURE in 2h 21m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/961792ac086d46979388ae57e51a7eac : FAILURE in 12m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/877c0e29a10a4d96bb658eec17043407 : FAILURE in 26m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c4dd6c19ad694ba091f12ae794afa2ed : FAILURE in 48m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d105e648e50450281461134aedb7b05 : SUCCESS in 1h 24m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da7908e3eb584210b8ddb3d214d5c71d : SUCCESS in 42m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5289dc5c81ce4038a9966ed51e9d7ad8 : SUCCESS in 1h 58m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1eb11bad18774010955915cde1a9e310 : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/71b738d4b7424ed0b1affcfa4ae87fd4 : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3df6356552c34b6196d991eeeee67582 : FAILURE in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ae8e384af47fc0e30b48418c1139ccd9b539abf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-22 07:35:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b88b1c915ad9f9a8e12410bc9cdfd9498015c95f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:08:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ff6ab0f11674b3a860155bde573205f : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f2d2ab00b144ad1a0a64be608c111d6 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d712ad27e54dbc953319f57183365b : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25ef5471e561482980a165dc6db0b2a0 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/883e4ce73efa4ddaaecf7a99487c7397 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f77aff98eee343b7a925e341961b5c10 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17236d7af2244ce7bfb6d4c96c5720df : SUCCESS in 24m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d22e1f10728346dfa39b41858ef20dd8 : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d359e5716e4c45aaa56dfd9f53ecb225 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c59c5269e71249fdb4e9be10c3b91d3f : SUCCESS in 8m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/514e52887fa74b71bdb3de29ac0a68fb : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bf8748fec15448ea416f98f2c7609a0 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17011cc0ce7c448099ad398918bcbe18 : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a0a87da154504c3f8e9fec09fe9df37f : SUCCESS in 1h 46m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a710a4426c184a718cb335be6d298b53 : SUCCESS in 1h 58m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9732adcbe9f6447b8e65f7b9d579ad4d : SUCCESS in 2h 19m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f2e9193127154aba88015da9bbe62d30 : FAILURE in 19m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6783ba7dd4a48f7b68b875913ee437e : FAILURE in 29m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2859d59816574d6482238d4fc71600ef : FAILURE in 44m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b60430af661a45a88f9c39688c8e2bba : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f83aa57a973463292dd451b175dff71 : SUCCESS in 40m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5bf62850b4024460816048b3c78f7f74 : SUCCESS in 2h 17m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d50eb057b46c4824b218ada4b785ddf1 : SUCCESS in 42m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/19cfc47e5b8b4707a91164ccfab75bdf : SUCCESS in 39m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9b98b6cedfff484084e5a1632527daaa : FAILURE in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4578244e9ea3786fcbaaea604bf86551edcea5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:08:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf493d1f37cb7f90eecf4b79ae822c4bf62dfe61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:01:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74f9baa2fac14a01965d0ec53199d350 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18febcc27512491991069937ab1001f8 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6c1e735e6a84a3a8d9eee3b2c58f725 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/852775b31c54449f9f3d88aa022d869e : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03e7aedf2e2d40358d8086b468a16c26 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f21b33228241e8a70bcdb19fad84fa : SUCCESS in 16m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/367ff3cdb9144042929b21eea6f80cdf : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b3310f805abf467d99ff4cf1e7835df5 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2f1927be655483abfd70ef1d40d85c7 : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5131c4bc9e504462a27486257aa96fd2 : SUCCESS in 1h 37m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d0b5bc155da64390a0f19ca07ca13b83 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d39329931e46d9956ddabc18c871cd : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/beb96e7444274ce7991a2c82fc92bc12 : FAILURE in 1h 49m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/25654ddb09654e42a743ab1b40f9d730 : SUCCESS in 1h 26m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f8e2b0839cf0f6883c0a4f37aa70a6cb85291feb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-24 08:30:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36679828cb5daf029129b53c07e3d30d70da7b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 11:22:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34a70131d9ea4f36a7a8e36ebdb1d8b9 : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b35052be82634cefa7ffed360fc17fc0 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28252503a7bc492cb7375abaf60a178f : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/409ca9bf1d9e47b8a8af3063b0bdeacd : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d794f2fa14d4a01b8b25c5528511eea : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/457caff9e56041f7901b64823f9279cc : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2460e3594a44979b18f93448f1ae05c : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e80147dcd854caf96d9e7928d9842e1 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7621414859e24b219f4eeabd3cd9fd96 : SUCCESS in 4m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/453405fc9b1a4775a94e0dff733abab0 : SUCCESS in 8m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e5401cd091c74941badf6ff0387eb3e6 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24da1c645e924d198a0d44d60efcc278 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4346b70effc44da7a8804ab8b6bf7636 : SUCCESS in 1h 38m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ed7fed15664340c0b1c4d7b661743a1e : SUCCESS in 1h 44m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ac16c2e261840c2aaaeeb92f6b409b3 : FAILURE in 2h 00m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/576d7929c5eb40f59f4dd9510336aefb : FAILURE in 1h 29m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5be139258179459f83be9d488799cb2e : FAILURE in 11m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/392c3b5075af42e6aa0d08b63c976c5a : FAILURE in 25m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/71f5e185a0e041b6827d675abc0d503b : FAILURE in 45m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/720f2c526c0f42f8b0e78a67f2ac92ea : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/622fdecdd28f455bb02afee15e817ecc : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dbebe99fff32423297081213d8a05cc3 : SUCCESS in 2h 16m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6d02164a7a94e618b9fb0586c4b3ea4 : SUCCESS in 38m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3692f4e51bf54a7b91aa145ade256a41 : SUCCESS in 39m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/18e863f8eee34d1da82b5b3654d4f55a : FAILURE in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b7d9bb5e879e4a9d62e25f56d36b3d3052bbb43","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-09 11:33:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b7e333079f124456248020df2078ba3d00806187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 16:52:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3f8c429259743809ada0b7681855f25 : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/febbb9f470d54cfc87121f31d69d8e32 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e59d51024e54946a3750343258b0ba9 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/277b045b3dca45e4978d9b90b0300d5f : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a39e73d5a645b7be40cabf70c1ed6a : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a58b5bf5a6624d74b1665d32915a0f2a : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739ba215dfd845b7801bc5e650b48d3d : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/909588c8646049cc9632da5be97b6e2a : SUCCESS in 9m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/777293d6e68b4f518e82027e67ab5a57 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/344691ac41d84940860345f3cd959e42 : SUCCESS in 7m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a4fb73f840394111aa2cb696b5d1a97f : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d004bc8ca0044b57bf58564aa58496a0 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a50f89df1614fc2a1168e55f918d928 : SUCCESS in 2h 08m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/162e1f94a6fb44e4a8782962c587b9e5 : SUCCESS in 1h 36m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b356e7b6bf9e4eebb8a8ddc08bd70182 : SUCCESS in 1h 27m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/35bc5336976c4bdfbe9f266251339646 : POST_FAILURE in 2h 18m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5a37d8b7d92142cfaee08ba4388fbd6a : FAILURE in 22m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/04fbe5defdc741c48f6ef9af3adaddcb : FAILURE in 30m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/128f9cfa983240cf8373eb160c761cf3 : FAILURE in 36m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6f46089513a42d8bb0be6244de7cfab : SUCCESS in 1h 27m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa9c2f138257448ea44fa65f6ad6fed2 : FAILURE in 30m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/50c8cdc424994ca088a3a2e632b61aaf : FAILURE in 26m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/790a7b4bd2934336abd867a2298c601e : FAILURE in 26m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9737a31b19894cb2b4f395d98a6ca4f2 : FAILURE in 31m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/10ae3478bfc7481d8a5e368a6247cce4 : FAILURE in 2h 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c25a895ebd125bbb451f53d9f2347717ad01bc5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-13 07:26:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"21af1e4ab192cda6aa624b598b586fd1738ebd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 12:21:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e295922e04424769a5994679cb259d35 : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e94ce73ad6d44449e80d32aa4c4846e : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7510d0df91f442b79e5c0ca0334ba2e1 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/365c2ed361784376a4946c6c9b5af909 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6b04972fea14db68c8a13173c5c6bbb : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80b2d9f21b9b40f4984dccfc3793d492 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f21cd16c45d4adcad73782c909a8eb2 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a316e47edf649d9b25a645b205dc1dd : SUCCESS in 5m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9fecabae94734595a23335076d067dd9 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61d3a973daa44a3b898f0a3c980d2edd : SUCCESS in 8m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d52448b2799e4e04a2e0851bdfc9c9d8 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c72bf19e5704270b836343b8bff34c3 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8cced117e3864b3285ff17a53c0d98b0 : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1aa6e4bb8a1349eda1be0064ef31c5bc : SUCCESS in 1h 47m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1019cba8a9f44c0ac330baa61c556e1 : TIMED_OUT in 4h 00m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b6368e5bded04e8d8a323b8646179a06 : FAILURE in 22m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/aec63215f658415facbe4d0f068c4899 : FAILURE in 15m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c252fe2b6a6430d86c8ae5f82057088 : FAILURE in 25m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f24174a72d514b7597fdfdeef88135c3 : FAILURE in 1h 08m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d42642b8405455db0a0e22fda0d8f4b : SUCCESS in 1h 31m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97582ed88ca147ea8e9eac382a88be6f : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ccb8db78e118496092c8dc3e059028fa : FAILURE in 25m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f658bb4a34041daa573594ffd8347ec : FAILURE in 23m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/637951b9d978472cb250dd914a2b650a : SUCCESS in 41m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a328688da23340f5941a1b9af1a16989 : FAILURE in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa0e99fd6753e95dd026af003bb7940d5969f5a7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 08:32:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9074895bf8a898ae0ffd0805748f2f61594d44ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:55:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d7ef9bb66174b5fa3d772905e2aa5e2 : SUCCESS in 10m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed7ce262bd354636bc7d440a092d0d1d : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d836338518f4c3f9978afa8d1c34460 : SUCCESS in 9m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9c0c57348fb42e3993437757ccbc141 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0fdd767683425f999be21355ab22a2 : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfaec86dd0cb40d6afad410ed3619420 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72ab1dd04ba4d758ed55fd76604061c : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86e15bc89bdd4c8f8a74fa97d1ed3921 : SUCCESS in 6m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7690cf03775843039d4a59df70829bda : SUCCESS in 9m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3dd8eaf4a7a24d3fa8a22ccb7adfdc84 : SUCCESS in 8m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3b14ecfbd20c40fbafbb0a455349ee7e : SUCCESS in 9m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6147e7c731149c2a850dfc7d4fb0c3a : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3846748abd440469687c3620223a542 : SUCCESS in 1h 46m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7caab376622c42ff9cd23602341a6e9b : SUCCESS in 1h 52m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13b655efff8c422f9df25527dbde0bff : TIMED_OUT in 2h 36m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e320f371b49946c7ae435bb1c6114c1a : SUCCESS in 2h 14m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/50c9d2c18427405da09384daf652e74b : FAILURE in 12m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/930b44815bc14eacaffae9e92ac51e46 : FAILURE in 38m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8122fc62ff14d7e9ddc92b11438e33e : FAILURE in 49m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/943966e931db49b181376cda5ca52fde : SUCCESS in 1h 30m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33042fd704e8422a8897728d87d5d564 : SUCCESS in 47m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e147fa86e034518bb7386cd12a3abd4 : TIMED_OUT in 2h 35m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9cb97df6d5849498a81215134be1204 : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dfee0d4519284185a95ca3c90da798a8 : FAILURE in 31m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1a2928c01b4746d0a45160d308392f1b : FAILURE in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f7c06fa245724f24d1804730a3529c827fc9526","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-23 10:48:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b6c485cc472167d0d03ea28bdffb428329153ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:35:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/551f6db966a44952b073e5c1ed296e12 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5954dfdc28ef48748885a5c320ea500b : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fbafccff0c04243a2800095ae916591 : SUCCESS in 9m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c20cddcae66244598889e53d03692613 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345c4537794b45b1af82cf69cca918d2 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7a1a674a89147c88420409e96ad2998 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1abe767a22f42cda68c3fca62050efc : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/666a80b00dde45a181cb6c792ec9a517 : SUCCESS in 6m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fd69c1f49cf343b2b737e912eb2a1124 : SUCCESS in 7m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/02b8b5d2b6e748ae9a528201b6a0baa9 : SUCCESS in 10m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e5f8cba27384347aa88f884f74996bf : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e28c1961bb1f42b5befdbfdd8283ffec : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eda2180b4a9d4cb08febf77911f76fec : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/df427d569a9c4f988731ad015e9b2c4b : RETRY_LIMIT in 4m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b5e929215e94b56919dec3a30faf329 : RETRY_LIMIT in 4m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f28a2ee64a4b42689fb03c9d05712dea : RETRY_LIMIT in 4m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0986de9e07154927bd623eefcb502765 : RETRY_LIMIT in 5m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/145752be9bb845ed8aa1f51855caaf52 : RETRY_LIMIT in 4m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/515e2b4406e24fe0b82d7543a7e7e38b : FAILURE in 50m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d22406b9eed84ccda34370b42d1a0995 : SUCCESS in 1h 31m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/401d0fe313224157a924e3ad5d288f6b : SUCCESS in 46m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cc3d4cc8b33d4c329d39ad09d833bdc0 : SUCCESS in 2h 33m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3abe128cc3154da7ac5e91d58e6d02fe : SUCCESS in 33m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/832ea0866cc24facb4dfa292074252d0 : SUCCESS in 44m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/67aa49951c6944c48f72cc502092f42d : RETRY_LIMIT in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9d7243d870716e746e0d358b6471326304effc3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-26 09:01:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bfdd0a046c40af982ba942348bf29d19af31f6de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:47:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/071756852ba64a4bbc67109231ce6e3a : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0716ed6ab1f8474b8e2c8f9681214244 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcde0a97587144f28dfdcfe47b4fed1c : SUCCESS in 13m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e9ec420b01c49e4ba5529f30212b353 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f0f2bcd113f49dda9364ec633776cdf : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a6572bb71594448b228c3e786cc43c1 : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72f1236742e438e8c937edf7d79b5b0 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb2288cee0e24855b7216ac9b74a87ad : SUCCESS in 10m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3eed8f79c25347debc529967dfa3d3d7 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/88854345f1374a48be589ce189950535 : SUCCESS in 6m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bb2e4a5387144330926b3692b83c417c : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c53d57e615304c818e3425cfa8fa6050 : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e18de7c7577c4e93a9a514f232b2ffbc : SUCCESS in 1h 47m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ddc227dee286421aaac8f2ec12d29d1b : SUCCESS in 1h 43m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e40a83f0abdf4898bb95b1a10da6ff0f : TIMED_OUT in 2h 37m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f6bde01ec85642b38059eff898369a83 : FAILURE in 1h 43m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3fe1345712694b0f94f2aa4787646d50 : FAILURE in 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af968083cbf54278ac3f5bd0b2026a37 : FAILURE in 45m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2cf7836f5ce48aebb0bb63721aab6ac : FAILURE in 46m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5ac5fc67886421b8a6b542f24afe9fc : SUCCESS in 1h 51m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e339e2b711a4c638f7d76c4d5628f48 : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/24d0139b6b5347c9989279fca86f02f3 : SUCCESS in 1h 52m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8698c246f5c4442da30e6de208419718 : SUCCESS in 40m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7ab53b0c5885416a87cfc966bcc1be1c : SUCCESS in 40m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d59070fdb6ae4872a2e2ce7ec581d3af : FAILURE in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd43f3aea527894301c4aac21ee7b244d19eeb32","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-01 14:05:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1f97237c8fbbc9e24df3199d44dec6f89bffa933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:21:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cbb1b9297bc478c861ce15e37c3d4c2 : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc14c817500e44f89a16179c21fe6f3f : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc065d94430141889e385919fd168aaf : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/073a1a36e0de4d1a9ac9c9d8853f0b59 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a45552475074a4b98d67940e27ced7a : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63a9936c8c52449882ff3e44572ac469 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab831e427804e07883a9d2b2105cc8d : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/430207978bb64278a02001283af4d178 : SUCCESS in 4m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f723ff2cd8bc42b0afe56780a47256e7 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5db51355c0634fbb8305cac91e3bd00e : SUCCESS in 6m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c9b48b25e48e4fd78f90442a7e992b4b : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/882b98ca695b45af8f1a954a737ec20e : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a9c40e4b2fa44be8d1470df140b8e83 : SUCCESS in 1h 24m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/229fea98fb9845f99e7e8e43ed95af46 : SUCCESS in 1h 35m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b20de3abbe4944a3871bdaae8c3df4a5 : SUCCESS in 1h 56m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ba1d772cb686461c90f9f5573f00faa9 : SUCCESS in 1h 43m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dad1219392e24e168e3ed6631f3d37ba : FAILURE in 11m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c70d2d1178643bda9becb30ebd98922 : FAILURE in 29m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/63a6b2620cc642d0b7df2dba4fcf93ed : FAILURE in 48m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/938d533434354b31a7d89072cac0e775 : FAILURE in 1h 28m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3e7686c2034d441bb8154431981237a2 : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfa2ce2999f84c0fb9988d83ef17f1db : SUCCESS in 1h 48m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03c3205076df4d6da7e96da759c4db09 : SUCCESS in 45m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e42bb12c68544acebf32bf154d3b21ac : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/92df016ec7e64b059738ac698fab594e : FAILURE in 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"66dac131020b805426ad40fd680a793c981f7d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-01 19:41:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3927d8815127e907de8e0ce82a7e2c48b0933756","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-01 19:41:57.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d21f93c3347e9df83d5c45062058ba6cce040b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:12:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c50240f1fd7d4996ab2f69ef6809b878 : SUCCESS in 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9b22f21cd26462e8b3e1ecb177b4138 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b182083be864118828d77e1c5e161b1 : SUCCESS in 9m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a527b95c85c440481ab8cb9437189f8 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45fc9f4d29a14dcebc43c5dcdfddcbd4 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/381d16e7cec74ca084cc93e9326a29fa : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81bc5c90ce845b49359f33f5c1e9988 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da2ddae2d72945bd9fc65bb9f5b68195 : SUCCESS in 13m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/665a8b839317432f90bc573828b38c6c : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d4928ea34d8f4999be74547ed84f54e1 : SUCCESS in 8m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9163f3ef03d2424ba1d31f55dff59116 : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fad56f11c0645ea8f52a899406b23aa : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df7498472f6a4ad3b266c6c376cfb7d9 : SUCCESS in 1h 14m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0154ebff96064d7ebc693b2b9955e5df : SUCCESS in 1h 12m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19acab48441f46a3ad057a888becd73b : SUCCESS in 2h 10m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cf1c8e09ec814f48bb26126040e2e2d4 : SUCCESS in 2h 25m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7a2ecf083c874f0bb0b82afa79f0ec94 : FAILURE in 11m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc45d7fd3aca458d9149521f99e533cd : FAILURE in 31m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/19e77cb17677402490d5bac4507b7250 : FAILURE in 50m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a1ab5b7682494a4689272b3c6d60c66a : SUCCESS in 1h 34m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2480c47a6e7a40ad857642647918f5fc : SUCCESS in 53m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fde557a1d1454fd4adddd4eaec9188ea : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e218e538d6ee4cf1bfedfc86d0e32c5b : SUCCESS in 40m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/76886df3e0714f0aaa05792bdfbd169b : SUCCESS in 41m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c4461e01c7c64789868127293a9a95df : FAILURE in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f63305df5a249551e2b0421e78040dbc884d83fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:12:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f1f8adbef094ffafe32a0c50d51bde4a0d460b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:04:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e919450719e84e5fbf3d571377ddcd41 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244c84d3ab2d4bce9e61d57adf5cca3b : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a154aad585d40939337f66a5248b5f3 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d139d98f5e4e09abacc10841fb2de8 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a0e00397d2f4a3da6ce4a445b5a7aa3 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a262eed13a4058a87e6ab614bc0833 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30b7050601464ad6b492ab2fc6ca95b6 : SUCCESS in 14m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ea30d9acd4b3440395d63fd7df12f364 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/093545e00a25476fb8174bd659a65fb6 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc71f0b812c24273ab20ef2b65383314 : SUCCESS in 1h 14m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/738699d990944041875bbb8bc5b667b2 : SUCCESS in 1h 23m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aaa3051753a94922915194d2a56e068c : SUCCESS in 1h 47m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d30587f1812a4029b99cefb24411fda1 : SUCCESS in 1h 39m 29s","accounts_in_message":[],"_revision_number":3},{"id":"fea87f89248605fee6c35d3b3982dc601e27343b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:04:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bec125ba23338a39e739a6fb507d4fe368da8d75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:06:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c812e2df6d6b4dc89a62b8d7765cfd0d : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58335f4d5d242aface2516a4ef8aff2 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ba8dc62b6b9e4aa01f77a39efd5eceb3fdc98a2","revisions":{"fd5a036d407071085e0f6bb7f502dacab70a0eae":{"kind":"REWORK","_number":1,"created":"2020-08-24 09:25:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/747662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/747662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/747662/1"}}},"commit":{"parents":[{"commit":"40777222b91028412c8b2a96230ec284d99620e1","subject":"Increase the devstack secuirty group rules quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40777222b91028412c8b2a96230ec284d99620e1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-10 19:11:23.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-24 09:10:32.000000000","tz":120},"subject":"Fix memory consumption issues with default connection_limit","message":"Fix memory consumption issues with default connection_limit\n\nWith 1.8.x releases, haproxy consumes a lot of memory when\nusing 1,000,000 as default connection_limit.\n\nThis commit introduces a new configuration option for the Amphora\nprovider: [haproxy_amphora].default_connection_limit (defaulted to\n50,000). This value is used when creating a listener with -1 (which is\nthe default) as connection_limit, or when unsetting connection_limit in\na listener.\nUpdating an existing listener by setting connection_limit to -1 also\nsets it to default_connection_limit.\n\nThe global connection_limit for a load balancer is the sum of the\nconnection_limit of the listeners, but it cannot be over\nHAPROXY_MAX_MAXCONN (which is still 1,000,000).\n\nStory: 2007794\nTask: 40046\n\nChange-Id: Ibc525d9a046a5ab7f090a942459d80a2df66ae2e\n(cherry picked from commit f4305e036c30b23588d7d1448448010b95c24069)\n(cherry picked from commit 242239a705b53eac4b5cace51bc5d5a2b4a12df6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd5a036d407071085e0f6bb7f502dacab70a0eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd5a036d407071085e0f6bb7f502dacab70a0eae"}]},"branch":"refs/heads/stable/train"},"ad7dc6c26022fb1c3b4dc0b694f00ef859787fab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-09 09:36:22.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/62/747662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/747662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/747662/2"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-10 19:11:23.000000000","tz":120},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-09 09:36:22.000000000","tz":0},"subject":"Fix memory consumption issues with default connection_limit","message":"Fix memory consumption issues with default connection_limit\n\nWith 1.8.x releases, haproxy consumes a lot of memory when\nusing 1,000,000 as default connection_limit.\n\nThis commit introduces a new configuration option for the Amphora\nprovider: [haproxy_amphora].default_connection_limit (defaulted to\n50,000). This value is used when creating a listener with -1 (which is\nthe default) as connection_limit, or when unsetting connection_limit in\na listener.\nUpdating an existing listener by setting connection_limit to -1 also\nsets it to default_connection_limit.\n\nThe global connection_limit for a load balancer is the sum of the\nconnection_limit of the listeners, but it cannot be over\nHAPROXY_MAX_MAXCONN (which is still 1,000,000).\n\nStory: 2007794\nTask: 40046\n\nChange-Id: Ibc525d9a046a5ab7f090a942459d80a2df66ae2e\n(cherry picked from commit f4305e036c30b23588d7d1448448010b95c24069)\n(cherry picked from commit 242239a705b53eac4b5cace51bc5d5a2b4a12df6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad7dc6c26022fb1c3b4dc0b694f00ef859787fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad7dc6c26022fb1c3b4dc0b694f00ef859787fab"}]},"branch":"refs/heads/stable/train"},"6ba8dc62b6b9e4aa01f77a39efd5eceb3fdc98a2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-01 19:41:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/62/747662/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/747662/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/747662/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/747662/3"}}},"commit":{"parents":[{"commit":"811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe","subject":"Enable octavia-tempest-plugin for grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-10 19:11:23.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-12-01 19:41:26.000000000","tz":0},"subject":"Fix memory consumption issues with default connection_limit","message":"Fix memory consumption issues with default connection_limit\n\nWith 1.8.x releases, haproxy consumes a lot of memory when\nusing 1,000,000 as default connection_limit.\n\nThis commit introduces a new configuration option for the Amphora\nprovider: [haproxy_amphora].default_connection_limit (defaulted to\n50,000). This value is used when creating a listener with -1 (which is\nthe default) as connection_limit, or when unsetting connection_limit in\na listener.\nUpdating an existing listener by setting connection_limit to -1 also\nsets it to default_connection_limit.\n\nThe global connection_limit for a load balancer is the sum of the\nconnection_limit of the listeners, but it cannot be over\nHAPROXY_MAX_MAXCONN (which is still 1,000,000).\n\nStory: 2007794\nTask: 40046\n\nChange-Id: Ibc525d9a046a5ab7f090a942459d80a2df66ae2e\n(cherry picked from commit f4305e036c30b23588d7d1448448010b95c24069)\n(cherry picked from commit 242239a705b53eac4b5cace51bc5d5a2b4a12df6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ba8dc62b6b9e4aa01f77a39efd5eceb3fdc98a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ba8dc62b6b9e4aa01f77a39efd5eceb3fdc98a2"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
