)]}'
{"id":"openstack%2Foctavia~749409","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I0c124e80fec2ac9ad813e78df6ad25da6b8e9668","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I0c124e80fec2ac9ad813e78df6ad25da6b8e9668","subject":"Fix API sort key for complex columns","status":"MERGED","created":"2020-09-01 23:25:39.000000000","updated":"2020-09-04 13:46:28.000000000","submitted":"2020-09-04 13:44:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749409-1599227074242-1189cc01","meta_rev_id":"1b5a957dac9af62b33bff6704fafc720697d0eef","_number":749409,"virtual_id_number":749409,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 13:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-04 11:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-02 07:09:51.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-04 11:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 07:09:51.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-04 11:56:18.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-04 13:44:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da399007a61310e909768ad2038fb8b6e2bd6502","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:25:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c43796486fb79046b69bf0f9a24a3b1a47180b72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:26:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0959b4637abd57a58001b63201310e7fe99976ee","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:27:10.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a9d638deb625ac9fee92ca1a7e83f5b696b60fb8","accounts_in_message":[],"_revision_number":2},{"id":"7dde8894bd9d0185305322807e8645f00fb4140a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 01:27:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29cb154ebf7a4259b983383a7aa10c29 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb86703baa946f9ae9becb0be73fbc3 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba27b42c3dca429d863b50639a63b754 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2206653c8f47488b83c15c6147f57c88 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ca0ff92b03f42e1b14cf869d6a1621a : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456ea7caa62d46a2bf71a7b81b11ea6c : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b66cc50f0714d9fb3a9846577e4440c : SUCCESS in 4m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7259ce30979b4f249074ecdb7648ebe4 : SUCCESS in 4m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/46c2c95afca24be58604647b5b748f05 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98f5255ae85045efa39049b7ca76c9f7 : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/232a818a302442c991a68142d44529b9 : SUCCESS in 46m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/934216ae7131455298e6cf99899c4eff : SUCCESS in 1h 16m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09045e53a65d4906ae622843cd273bbb : SUCCESS in 1h 00m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/89bc81a305ff4bee8dc7353bee6768fb : SUCCESS in 39m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34fadf816e7a4db99decd30644347821 : SUCCESS in 37m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67760c52223946e8b3aea8474525e7dd : SUCCESS in 1h 55m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b1deecd41264986b0c7fbacadcd496f : SUCCESS in 1h 20m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96790b314dca462994941358eb2e46e5 : FAILURE in 43m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96be45c2a79d4fc7b9bd999ac6b62e8d : SUCCESS in 1h 03m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53fc43f36fa9095914e92d64257836871a3d706d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-02 07:09:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"914833f85a5668d816352d66296abfc4f3c9d879","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-04 11:56:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee68fdaaa9a179deb4d0f878da0bf0e0dede59cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:56:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3cf72463aaeac79bfc9e9f8280c016e09148b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:44:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d80e3f5e48a344fead60814f0ec5988c : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb5cf478c714d2d98a4d5733b1a9ac3 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feb490795e6a42228b687db1cbb9e704 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94e30e1cb0954ff6a89b2c3c617f526c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312ec94317634111886d121a1dfbbfe0 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adf3211ea8d746a6ae1cdb95ca2ac436 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36f67d634218458e9e3c5f1623649bf3 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4380f5a579546419794e5bb569a382f : SUCCESS in 43m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6d6a43566fa4ecca6d5950dd8096e95 : SUCCESS in 1h 21m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b508b051d2334906b530d6e20f56fe3d : SUCCESS in 42m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0a98f34b341146c9874cfe45e1909258 : SUCCESS in 41m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab27184382054c46815d0888102aa895 : SUCCESS in 42m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/568652f4e9a84c30ae9918bb4e5d38e8 : SUCCESS in 1h 46m 22s","accounts_in_message":[],"_revision_number":2},{"id":"aa95eb04ffbf7d62c8c0b355edce38861c49dad8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:44:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1b5a957dac9af62b33bff6704fafc720697d0eef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:46:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fd6dd6435a48fb904315d725e8ac89 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c4e2cb1f44408e8b5846ea0e2d3365 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"39eb3286024d1118fd67e80df67bdd3abed7d5b9","revisions":{"c7cbbbb2a385319dd4be902dbf42d4c3e5a572fd":{"kind":"REWORK","_number":1,"created":"2020-09-01 23:25:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/09/749409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/749409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/749409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/749409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/749409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/749409/1"}}},"commit":{"parents":[{"commit":"5e377fa8a73fa2d9c5b6fb83c585abebbaf2b18e","subject":"Pin pylint for stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e377fa8a73fa2d9c5b6fb83c585abebbaf2b18e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-06 22:15:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-01 23:25:38.000000000","tz":-420},"subject":"Fix API sort key for complex columns","message":"Fix API sort key for complex columns\n\nThe Octavia API sort function was not working correctly for some\ncolumns that have different names in the API vs the data model.\nThis patch also handles the case where a column may be a compound\nobject in the data model, such as vip_address-\u003evip.ip_address.\n\nChange-Id: I0c124e80fec2ac9ad813e78df6ad25da6b8e9668\nStory: 2007991\nTask: 40626\n(cherry picked from commit b6f6c14292329b1a977fd7c0831f917cd69cbaaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7cbbbb2a385319dd4be902dbf42d4c3e5a572fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7cbbbb2a385319dd4be902dbf42d4c3e5a572fd"}]},"branch":"refs/heads/stable/ussuri"},"39eb3286024d1118fd67e80df67bdd3abed7d5b9":{"kind":"REWORK","_number":2,"created":"2020-09-01 23:26:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/09/749409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/749409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/749409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/749409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/749409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/749409/2"}}},"commit":{"parents":[{"commit":"5e377fa8a73fa2d9c5b6fb83c585abebbaf2b18e","subject":"Pin pylint for stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e377fa8a73fa2d9c5b6fb83c585abebbaf2b18e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-06 22:15:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-01 23:26:22.000000000","tz":-420},"subject":"Fix API sort key for complex columns","message":"Fix API sort key for complex columns\n\nThe Octavia API sort function was not working correctly for some\ncolumns that have different names in the API vs the data model.\nThis patch also handles the case where a column may be a compound\nobject in the data model, such as vip_address-\u003evip.ip_address.\n\nChange-Id: I0c124e80fec2ac9ad813e78df6ad25da6b8e9668\nStory: 2007991\nTask: 40626\n(cherry picked from commit b6f6c14292329b1a977fd7c0831f917cd69cbaaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39eb3286024d1118fd67e80df67bdd3abed7d5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39eb3286024d1118fd67e80df67bdd3abed7d5b9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
