)]}'
{"id":"openstack%2Foctavia~749410","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I0c124e80fec2ac9ad813e78df6ad25da6b8e9668","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I0c124e80fec2ac9ad813e78df6ad25da6b8e9668","subject":"Fix API sort key for complex columns","status":"MERGED","created":"2020-09-01 23:27:10.000000000","updated":"2020-09-04 14:30:14.000000000","submitted":"2020-09-04 14:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749410-1599229627863-f4e8b138","meta_rev_id":"ced1c0e8ffc45b17b2be004886b80144ecaa0e98","_number":749410,"virtual_id_number":749410,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 14:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-04 11:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-04 12:37:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-09-04 12:37:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 11:56:33.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-04 12:37:37.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-04 14:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77e79617c42039369df92b33f386abaf06b65315","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:27:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"95f7041ea39b79e94e6ca8519f8d64fe8e3bcb38","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:27:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1651c29291bab4bc7272ac8ff5dc448c94bc15b3","accounts_in_message":[],"_revision_number":1},{"id":"84fd3352cfb9f845565b1cfbf4cbb0a48d3c6931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 00:48:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adc58560db71440e9415bb39ed21dba4 : FAILURE in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f62b5d9188c437b84d38785a219cf90 : FAILURE in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b717abcec23943a2bab5b86c15b03d67 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31e1773cae39428a846579dfb20e0c1b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79f398f5c8ab461692b60e232d0ec9c7 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d8f1df790f74991b0470594d298af48 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee23003f97047a8ac9443730f4fe54b : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b4fb11a1fc2400580783ff705705f5f : SUCCESS in 4m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30ef3997dc9c4500a2b3a99f55234d65 : SUCCESS in 4m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24419a1b1b7b4183a8400bea2f3fa281 : FAILURE in 7m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/92efe06558a14bc183f0b80038dc4b0f : FAILURE in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0070e7b0d1184f128b81e979a9cbbe2a : FAILURE in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87d9f214af3d480db8fbc26456913da5 : SUCCESS in 42m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5b5ab85a601043a584c7ce22b165c7b4 : SUCCESS in 41m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c89aad67b4f41baa22348d3a80dca96 : SUCCESS in 55m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/285e0d8fb2c74b5aa22dce2a72e7fa3f : SUCCESS in 49m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8edd3e8a749d4c308855871174fa480b : FAILURE in 15m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c0b655634ca40c68256b5202623c71f : FAILURE in 22m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac4aa3c389634d3ea82b11bb3e59682d : FAILURE in 34m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/21634ee043d342dd8ff63bedd6c19280 : SUCCESS in 1h 19m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08f260a7cfb942cd9f5c3a7f59d04dca : SUCCESS in 1h 01m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1a1dcab55d3c40e390d4589369d93c85 : SUCCESS in 45m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cae34ea66bfe4dcabc5b498995e40b05 : SUCCESS in 40m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5b81c49598ee4fbb808a69de9be354c3 : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d01066045a9144dcbe51bde50605fbae : FAILURE in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0923a77fc4b0c5e41babeebbaea1d2f25201d64","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-02 18:38:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c0a22ed85508ec7817ae1c95db7f3d422ed545a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:30:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eee7f2e2af8c480db58fdc17f289de84 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aa5f2b34d9848b0a1a9f58d35e8a03b : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee79dc20b091454a94c642d1372a29bd : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfa2afe40aa24f59be8780433338f492 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ec08bd6a8a449249fcde8c3ac981ae0 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/beb7d41f543a499eabbd888d2a6ea36d : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e6797e885c496b97552c2cf4dc4e60 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0d582b5cfbe4e77be0ba41941eec199 : SUCCESS in 5m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/40653fc21a824fd9bb6516f0fa5a44df : SUCCESS in 6m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/736e65c3b2de43888a6154bc37bf89ae : SUCCESS in 10m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/41744440c4dc4c7aa5dab92235af612c : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9bcefe41fa1a40aeb84525b485920d9a : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0df1a12188b94907a403923953074a6d : SUCCESS in 50m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/69b9ae295f13436aa65ef6f6cfbd1c81 : SUCCESS in 42m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/008af65e13c74630adce9ae8b941bf6f : SUCCESS in 1h 29m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f1afcb87c3674e718caadfcd92e7e8da : SUCCESS in 1h 14m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5dfbeaa15e2d48869ced7cc3b4a1fc6d : FAILURE in 13m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6079e8f3657445b491075239ff6fac8b : FAILURE in 28m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2396752b91714279a1abb02a275c49e7 : FAILURE in 45m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/899a74a85b7447c28c6d630fe258f341 : SUCCESS in 1h 40m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/402e5e32586a4599930388ce7c526ae1 : SUCCESS in 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b83ae00d4e4747dfab0a19a4a8bb9eb5 : SUCCESS in 54m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88f0fa664e304b80a706fd4d6cf7847d : SUCCESS in 44m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e796d99b7bd54689a3f0094d8d888963 : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9a14285e4bcd48b5a3caf79d87d19b8f : FAILURE in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56b027b866707bbe68fc487723b30e492affce41","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-04 11:56:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c65024d0e743952703a2e104270d4197c5c6099","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-04 12:37:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"08606579315f1b4a638f02b783300501073a393b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:37:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56b75ab0613bd74647c7739bda3ad3c41121bf69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:27:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a575d34f926e4011afd48e86e14b6aeb : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4aabf2a58d446caaa098474f8125528 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f523e48224e40638327666a85650be4 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a900ca1a94d24ef6ab147e26c4fabd4b : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7a2b7887c8b43ffa2732064df2c0706 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9203e091fa47aaac1934e1041ea712 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b177e9cf1bf94d72923f34c14ac0eb8b : SUCCESS in 4m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04a094a99dfd4bc39511505c8f7d80ac : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/992e2cb5902a4fb0926a21a850da835e : SUCCESS in 6m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e4b48fd894c477699b3503cf359b97a : SUCCESS in 45m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5e2ac9214b644ba28d9b35f2b5ab77e7 : SUCCESS in 54m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f33dcaa8e514cb69202c3a7d0a5d3fa : SUCCESS in 56m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/aadfc3aa4378418f8419dbcf2051bd08 : SUCCESS in 1h 09m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f9659f7d7cd04ed998dde255c3dedd8d : SUCCESS in 1h 40m 29s","accounts_in_message":[],"_revision_number":2},{"id":"998e7d1ec51ef4029bb5578302cb410af989c030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ced1c0e8ffc45b17b2be004886b80144ecaa0e98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:30:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d3ae3f8d694ee2b980d758bcdfec29 : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f71ff77709840d5b398c3cf0c95614f : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","revisions":{"a9d638deb625ac9fee92ca1a7e83f5b696b60fb8":{"kind":"REWORK","_number":1,"created":"2020-09-01 23:27:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/10/749410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/749410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/749410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/749410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/749410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/749410/1"}}},"commit":{"parents":[{"commit":"0117d21588db66d4751f95de6d21b8c52e474960","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0117d21588db66d4751f95de6d21b8c52e474960"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-06 22:15:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-01 23:27:10.000000000","tz":0},"subject":"Fix API sort key for complex columns","message":"Fix API sort key for complex columns\n\nThe Octavia API sort function was not working correctly for some\ncolumns that have different names in the API vs the data model.\nThis patch also handles the case where a column may be a compound\nobject in the data model, such as vip_address-\u003evip.ip_address.\n\nChange-Id: I0c124e80fec2ac9ad813e78df6ad25da6b8e9668\nStory: 2007991\nTask: 40626\n(cherry picked from commit b6f6c14292329b1a977fd7c0831f917cd69cbaaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9d638deb625ac9fee92ca1a7e83f5b696b60fb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9d638deb625ac9fee92ca1a7e83f5b696b60fb8"}]},"branch":"refs/heads/stable/train"},"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d":{"kind":"REWORK","_number":2,"created":"2020-09-02 18:38:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/10/749410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/749410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/749410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/749410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/749410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/749410/2"}}},"commit":{"parents":[{"commit":"0117d21588db66d4751f95de6d21b8c52e474960","subject":"Fix accepting \u0027insert_headers\u0027 when unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0117d21588db66d4751f95de6d21b8c52e474960"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-06 22:15:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-02 18:38:34.000000000","tz":-420},"subject":"Fix API sort key for complex columns","message":"Fix API sort key for complex columns\n\nThe Octavia API sort function was not working correctly for some\ncolumns that have different names in the API vs the data model.\nThis patch also handles the case where a column may be a compound\nobject in the data model, such as vip_address-\u003evip.ip_address.\n\nChange-Id: I0c124e80fec2ac9ad813e78df6ad25da6b8e9668\nStory: 2007991\nTask: 40626\n(cherry picked from commit b6f6c14292329b1a977fd7c0831f917cd69cbaaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
