)]}'
{"id":"openstack%2Foctavia~749411","triplet_id":"openstack%2Foctavia~stable%2Fstein~I0c124e80fec2ac9ad813e78df6ad25da6b8e9668","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I0c124e80fec2ac9ad813e78df6ad25da6b8e9668","subject":"Fix API sort key for complex columns","status":"MERGED","created":"2020-09-01 23:27:21.000000000","updated":"2020-09-04 14:30:58.000000000","submitted":"2020-09-04 14:27:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749411-1599229632229-038e877c","meta_rev_id":"b4567ff0f025a8fc24f0236aa595e8b1751c452d","_number":749411,"virtual_id_number":749411,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 14:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-04 11:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-04 12:38:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-09-04 12:38:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 11:56:43.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-04 12:38:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-04 14:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9b48097321ba97fd1c5df6698d2e23907f7bb47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 23:27:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"6ca5c8e0b227f23524adaa946f99d72e651ea3ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 01:15:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8643bad90ee446ebb2e9c76cd88f39f : FAILURE in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acb3900026134234a5a8e9654fd33c51 : FAILURE in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44cc33d2c06445b9b958a5086de7563 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21ff2faad9a447bd94e5fbfd2ff38633 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7a120d089374c768d067da4de18f026 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1402a9d1c1246ddadd5064c1c5a7a25 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed25e9f09714aa7b2ccf29ca7f36f06 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b43171f3f7e480da1f19ffcdb559a68 : SUCCESS in 6m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/38fab91f4cfd44109fd360cdeb8d1f3b : SUCCESS in 1h 27m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2102e396aa9843c5aacc420b75f17fe0 : SUCCESS in 1h 18m 50s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da93178e7e934ff7a283553148d7669f : SUCCESS in 1h 39m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2a848322a8f74fa6a0906fb0cb620a0d : SUCCESS in 1h 07m 31s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7f7b7f43215143f28deab34cd07e9b97 : FAILURE in 6m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e4811d8ef1f84abe98c5cb825cad1a97 : FAILURE in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/879d12ab8737400488979f10ab4e5e92 : FAILURE in 6m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c10fc8ccd36458db0fecc2843556698 : SUCCESS in 48m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2af7598f288a4468bcd60e7b422e5940 : SUCCESS in 1h 10m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e93cc9cf50b4020a569242f440a2760 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/af215f1cc05543fe9c4b1af90a3d681b : SUCCESS in 44m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90001cdd579f41ea8c3bd7db2eb2bdd8 : SUCCESS in 51m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/994d11da3e7644f599b27776d410ab1c : SUCCESS in 1h 04m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b1a95a79cac64c30a9730dbe8e303480 : FAILURE in 12m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28ccb3b273d048d4b552caaf25ef8893 : FAILURE in 28m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/64519715ab444476b43fc6b93ebca487 : FAILURE in 1h 08m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e429916e7d2740e0bcf26efbbc8f6d2e : FAILURE in 59m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a367bf3dabb94f51a3f42251852dfc5d : SUCCESS in 1h 36m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9fbb04a5e73d45acbdbc1d905b27bb30 : SUCCESS in 42m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ecd67b9f4f2501666a4ff7995dcc543786b2fff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-02 18:40:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b331968aad95c63c030ab182133ba48762c2be4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:12:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b437d22441e47eba7e3712070a81145 : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98d30477e30d4abf80c1bb1bb90bffe0 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062942367587438580020e8947ac90d3 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e84044970f22492997e9937c361999b9 : SUCCESS in 8m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a1b4ea96a99e4076ba92775b9b7a67b0 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55ba4f6f71534887b1b4f401f6ed531a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98aab676a8984843b6a4fe4cde5c614c : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270df64ac2844e63ae603d541c2ed883 : SUCCESS in 8m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dee440609bd04ab5831bbcaf9a8dca48 : SUCCESS in 1h 22m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a9ebd318f8df48e9ae45a493b82b55f9 : SUCCESS in 1h 20m 40s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92c15367203a49e79dc3db4c4e4f3b28 : SUCCESS in 1h 21m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/01d6447215e0442fab3fef3c56018aba : SUCCESS in 1h 18m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/642f8f2310c04e97b8a2851e854074b5 : SUCCESS in 12m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d89c5286c3a34cc49f80b56cebbfe964 : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c7a76697aff49df8de93adfb23b5aee : SUCCESS in 7m 17s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78bf3996abb74f18ac6e8e16ea936dd0 : SUCCESS in 58m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/21de3c7daa274825a5391dc22fdd477b : SUCCESS in 1h 04m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58ab6b6e00284f02aa537679418e79be : SUCCESS in 46m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/79a2245b712b40ae9ee8c5af9b2b3353 : SUCCESS in 44m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/002c976ae0b94f9ba42dc4410b16501c : SUCCESS in 56m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cf77a9d56d764bf8812e1b631ee62512 : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/21ce63744a1f4fa5b29a493aa3259bc5 : FAILURE in 10m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e4da6b661274d1b8f48d226e8592ac3 : FAILURE in 1h 07m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/502a7f43715540098d328f9bb810623a : FAILURE in 38m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9245d49fc2f54ca59ce047b99b4504ab : SUCCESS in 1h 21m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0dcd451287d34040a7ee794b551d3f06 : SUCCESS in 1h 03m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53b5ae1563ecf879ef0b33c18b2150e6024efb31","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-04 11:56:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"145a524364a0e7cccbcd314257058fbb04f670fc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-04 12:38:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"acf9a6fcc3fc2963a3dc0ae96d41bc4fc07e93d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:38:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3a27f0962ae1502bbb8f7f7937135561b5ea837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:27:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bff9609a492c4a70807014aba4911bc8 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5534f2bf6a2344be885cd03117709cfd : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3f64ca3fdea453db729b805fa627a8f : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/34745b715882451aaebb101476927de0 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3606cb29eacf4552a3cc12d39f76e626 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10fdd7b7f2304a698c8f580f64f41c9e : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab0085907af426b8e7d0c0588a14fad : SUCCESS in 7m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d13a44852c34c7fa3c8801e60fe7794 : SUCCESS in 8m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/49c53704811941fd9ff455830f94cdea : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60df7ee3dfe54e14b13a3e9a5769d422 : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc636f7971dc4232afcbfe5afd5209ed : SUCCESS in 1h 07m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/006ee482af1b46e5b8f4997c97b6d5d5 : SUCCESS in 1h 07m 22s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e823a9fb99714034a8d8c8df47fa0be4 : SUCCESS in 1h 18m 30s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ea5a893514204318b957e54881e62dc2 : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e398f84ebd13435ea6f0d818204b36a5 : SUCCESS in 56m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e9fed68e9c7b45e88586f713e383499f : SUCCESS in 49m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c828ac32e62545e0b4562f91541f74b6 : SUCCESS in 55m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b56d0304ce7d4048a2c1520880fae240 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1e3d1e22b9ac60cf8ff98dbf4debd5148d613659","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:27:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b4567ff0f025a8fc24f0236aa595e8b1751c452d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:30:58.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f824a5a5743b42e29e61b8f13fd6aceb : SUCCESS in 3m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45fd30620f0d4e298768ea61c7197781 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d85b18cb5c0d0873b919ce8d76cdd3540ba7724c","revisions":{"1651c29291bab4bc7272ac8ff5dc448c94bc15b3":{"kind":"REWORK","_number":1,"created":"2020-09-01 23:27:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/749411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/749411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/749411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/749411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/749411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/749411/1"}}},"commit":{"parents":[{"commit":"6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8","subject":"Increase the devstack secuirty group rules quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-06 22:15:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-01 23:27:21.000000000","tz":0},"subject":"Fix API sort key for complex columns","message":"Fix API sort key for complex columns\n\nThe Octavia API sort function was not working correctly for some\ncolumns that have different names in the API vs the data model.\nThis patch also handles the case where a column may be a compound\nobject in the data model, such as vip_address-\u003evip.ip_address.\n\nChange-Id: I0c124e80fec2ac9ad813e78df6ad25da6b8e9668\nStory: 2007991\nTask: 40626\n(cherry picked from commit b6f6c14292329b1a977fd7c0831f917cd69cbaaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1651c29291bab4bc7272ac8ff5dc448c94bc15b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1651c29291bab4bc7272ac8ff5dc448c94bc15b3"}]},"branch":"refs/heads/stable/stein"},"d85b18cb5c0d0873b919ce8d76cdd3540ba7724c":{"kind":"REWORK","_number":2,"created":"2020-09-02 18:40:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/749411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/749411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/749411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/749411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/749411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/749411/2"}}},"commit":{"parents":[{"commit":"6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8","subject":"Increase the devstack secuirty group rules quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cf72ccbb13e8b0df4e6ccc5588e5db569dbeda8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-06 22:15:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-02 18:40:17.000000000","tz":-420},"subject":"Fix API sort key for complex columns","message":"Fix API sort key for complex columns\n\nThe Octavia API sort function was not working correctly for some\ncolumns that have different names in the API vs the data model.\nThis patch also handles the case where a column may be a compound\nobject in the data model, such as vip_address-\u003evip.ip_address.\n\nChange-Id: I0c124e80fec2ac9ad813e78df6ad25da6b8e9668\nStory: 2007991\nTask: 40626\n(cherry picked from commit b6f6c14292329b1a977fd7c0831f917cd69cbaaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d85b18cb5c0d0873b919ce8d76cdd3540ba7724c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d85b18cb5c0d0873b919ce8d76cdd3540ba7724c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
