)]}'
{"id":"openstack%2Foctavia~750081","triplet_id":"openstack%2Foctavia~master~Ibfb754d57c5c09c2358050160c6d98f09223f6db","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ibfb754d57c5c09c2358050160c6d98f09223f6db","subject":"Add amphora delete support to amphorav2 driver","status":"MERGED","created":"2020-09-05 22:38:02.000000000","updated":"2020-09-16 18:49:15.000000000","submitted":"2020-09-16 18:47:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750081-1600282061739-5dd22bf2","meta_rev_id":"3ac3f18a5cb427d761ceaee56dc22e5caeb92ee7","_number":750081,"virtual_id_number":750081,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 18:47:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-10 14:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-10 14:09:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-10 14:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-16 16:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-2,"date":"2020-09-10 14:09:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 13:59:45.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-08 18:01:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-16 16:24:54.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-16 18:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9625dc92e279766abe656cdf6a8741a5f34bd549","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-05 22:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c56ccb7ac219c308ae113141ec76513251484e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 00:16:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40001d054c8d4241b271fa5d24ac596e : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9be2e65459de49a8921fecf4b925c701 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f5d71590bc4653acdc37c66fde36df : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fced2174ddf442718124aa9d92546d5d : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14d8a286c99845a48b393efa0e5ad570 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e5d2c2115f49468eeddb1b003fa53d : SUCCESS in 11m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d4709ae90d343a895d424164cd2ccc1 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/47572277880443339b2505382d703e46 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f6726d55c3a4952b590e4620c37b407 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42ac9d80e696476bb2d8dca48d1fbd50 : SUCCESS in 45m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1de92ff8d28442f38a038997bcfa465e : SUCCESS in 42m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f0761eca85741ebaac403eb7e30b3a9 : SUCCESS in 30m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/833e6d0182ea44bca4b6f040e6ee9e19 : FAILURE in 30m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eaf3e80e153745d69d12e12d14de8750 : SUCCESS in 29m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3dac18e1845456499f67d9c668f9ee2 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf118ae9d66645d7bedd5b1cf13d8e38 : SUCCESS in 51m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c6ed1dbd75a4742a26cfcb9aec60b28 : SUCCESS in 1h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aaa21ea9458f4405b0cecddc62b8133c : SUCCESS in 43m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a111a950465c4f44b0c23d1780a7ec55 : SUCCESS in 1h 00m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5d5d55cb9c9b4f24b1d033ffdef491c6 : SUCCESS in 43m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/76a445c423894f649fe7c9a53af14964 : SUCCESS in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0b27cd0b97fde36e866b9d3bdc6f671ed536221","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-06 00:59:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d842b9f74fa3a8feaaffdaa1f967b129f59285c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 02:03:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27fd19b7baee4678b7596a6316ce45c5 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2343f0d0b7b4b2d969c7f9d010341f4 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f552ef10e694ee894c4e03e83307088 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7357c865c45477c88d72f53f3ca8382 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb2c5c4a6f4d479aa9aa6a095c6d0ca3 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa1f58718524019829953e035c349a1 : SUCCESS in 8m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/588a49885afb44af84fb5cce1d816f6c : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0d073636a7214873ad0eea1eaeb2f872 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/745fe9f1ab4c42eab89b9d4263beaff7 : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d24f4481eebc4b098db400fa7210c08c : SUCCESS in 52m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b85c91d9072475d8a82ec9d0017e3ce : SUCCESS in 41m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ef8dcc6499341fba9810ae9f55ef6c3 : SUCCESS in 44m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/da35e6401e2f40e5938c5d32dc032931 : SUCCESS in 28m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a9075127547467c83da504a1b2bb46a : SUCCESS in 41m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3bd4144cec3c4b16a9af18c9f4f79e23 : SUCCESS in 1h 02m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ffe2d21eb04f4565a9b9c9619bcf3c35 : SUCCESS in 1h 02m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ee7582e62491473485814e1e64ae11aa : SUCCESS in 50m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10be6323136b4cc49406a1874a77eb9f : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e25b458c46d4e83be43f38f5cfe7a02 : SUCCESS in 44m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/469f2490b76849dbbec13ce9e3897c5c : SUCCESS in 48m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/abe47404924346d8a311e575764cedc5 : SUCCESS in 59m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40a49c2f4e20bad2ea9497f2cd166446d3ba5c09","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-07 13:59:45.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97ec65f6c033f9971da9dc82ed2ffd0eb0c2e945","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-08 18:01:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a9abafd000d8078916095dcf1f484f24d1a28c0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-09 12:11:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91e76ed1c92e981fffed26df8ca150f4e41899fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 12:12:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7190d503d5c46588f21e15798593db3d0f492be0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:42:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5074598d5494e989102d93025c5f1bb : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8923fdd869cf4a27bc7f78603f18dd11 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adb8e015c8c94b2889e8c6e09948944c : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c98514eb1c14493d8468cc6dc25a5907 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da137fedc08847a7b1fad34e22ec6dac : SUCCESS in 12m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f46fe610b52648fd935e36989d47d591 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df3d51e392aa4c78a24c91154c702bbc : SUCCESS in 46m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10064de032014a6982def18d084ab58c : SUCCESS in 2h 17m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c97bbcf9c9194ae598e59505add4fa50 : SUCCESS in 43m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/85f6dfdfadcb4e198c39ce34b6d3ddb4 : SUCCESS in 38m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dec4d3f18a774232b63f6a4229cfb595 : SUCCESS in 32m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aafd82200bd147e38614bcf18335eeb9 : TIMED_OUT in 3h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"85f4b74fb6a6d13b9bc94216ee4eb8c139621caf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 14:09:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5680919acbeae12a9df5e74e4c520ddd1b46b71b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:41:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41739e07877f4aac81cbac3c97251cd7 : SUCCESS in 11m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/077f138253994adca66e35272466f7b1 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5903412f7f844cab3c15d7ca3de4a22 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/624aa6b187964870b42abc086b106ada : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ded5ad56acd1422f8ee45aed957834e1 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cae49648dd4af98527a2c2c5f083a2 : SUCCESS in 28m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e802973dd55147ce97bdbe89272dbd13 : SUCCESS in 7m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/32d0676aa26e4afd8cd9642e78153300 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7190b6bcee134aa48ec123b1359d9cd6 : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/973be7f7fc5646f4a608b1984a5f07b3 : SUCCESS in 48m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d61e4975e09400aaccaf8585e0e994a : SUCCESS in 2h 12m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7eb81e5b4a0f4e8fb167ab9f9acfe68c : SUCCESS in 32m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f7b4f20379a345fc8d381edf86fe0a4f : SUCCESS in 42m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a98604ca7bc4340b889c1f683a9473e : SUCCESS in 31m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/640276c6c2bf447999dcb04cc92bdc9d : SUCCESS in 45m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e667aeeda16f495bb6223141ca63219f : SUCCESS in 1h 43m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/769f7338a2264043b46ae5328d1e3c74 : SUCCESS in 2h 00m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/100e9f788be64f968fe630609aa53c75 : SUCCESS in 2h 16m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e235794f467346878360fb4389fdc197 : FAILURE in 1h 51m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a67693e35516424cb94d2d3d214b064c : SUCCESS in 2h 25m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0d99b529f1bb4efdb6ff9733ebe0afa0 : SUCCESS in 2h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8f7a6836ccfa0ecb3cba7e11258a01464a143a6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-16 16:24:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f54d62676119ea3b2ab729bb63eb1ac9bab313e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 16:25:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"161d2db0e63bc894a5818c16b260d2f24d92b782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:47:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/498a56029e844190b5104c464f5f016e : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a76372662824ea3915ece88f18f5832 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dea91d9ae534d31b308a57706c4dfab : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ede03c45222b45548e95bf05a6d71bf6 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1858803b9042b58f0eeb547e8248ba : SUCCESS in 23m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7d8f00d26ab4eab87e04093ab2c3b94 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/131e2dee3cde427eb3e409c31de8bc6c : SUCCESS in 56m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/840f53e434d04dd5b7ba4be290e1a869 : SUCCESS in 2h 19m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1c4e27ef5634aa195a6e686193bea0a : SUCCESS in 56m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/221033f7d5e24f8b9e1955051a13a9a4 : SUCCESS in 1h 09m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec28c15932c145918364cab2a090a813 : SUCCESS in 1h 10m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/025cf1755c324bea98683eba3b9ffa6e : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e977087125d311e988dc9cce29e84da1192f4245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:47:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3ac3f18a5cb427d761ceaee56dc22e5caeb92ee7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:49:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23068d6bf7994b1d9a02da36b877dfb4 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04b18e4d9019a4a4ffcef81f0f2268040050c141","revisions":{"9dbd4b70399d12bb0fedb1a03867d894fadc4b81":{"kind":"REWORK","_number":1,"created":"2020-09-05 22:38:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/750081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/750081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/750081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/750081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/750081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/750081/1"}}},"commit":{"parents":[{"commit":"46de66b240697b1e6465c06a203fe6dac75b7da1","subject":"Merge \"Update amphora v2 for the failover refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46de66b240697b1e6465c06a203fe6dac75b7da1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-05 22:35:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-05 22:38:01.000000000","tz":-420},"subject":"Add amphora delete support to amphorav2 driver","message":"Add amphora delete support to amphorav2 driver\n\nThis is the follow on patch to the amphora delete patch[1] that\nupdates the amphorav2 driver to also support the amphora delete\nAPI.\n\nStory: 2008014\nTask: 40824\n\n[1] https://review.opendev.org/718293\n\nChange-Id: Ibfb754d57c5c09c2358050160c6d98f09223f6db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9dbd4b70399d12bb0fedb1a03867d894fadc4b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9dbd4b70399d12bb0fedb1a03867d894fadc4b81"}]},"branch":"refs/heads/master"},"04b18e4d9019a4a4ffcef81f0f2268040050c141":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-10 14:09:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/750081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/750081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/750081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/750081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/750081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/750081/2"}}},"commit":{"parents":[{"commit":"4260d8a74bae4a46cd9104462eb98fcf30582d37","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4260d8a74bae4a46cd9104462eb98fcf30582d37"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-05 22:35:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 14:09:50.000000000","tz":0},"subject":"Add amphora delete support to amphorav2 driver","message":"Add amphora delete support to amphorav2 driver\n\nThis is the follow on patch to the amphora delete patch[1] that\nupdates the amphorav2 driver to also support the amphora delete\nAPI.\n\nStory: 2008014\nTask: 40824\n\n[1] https://review.opendev.org/718293\n\nChange-Id: Ibfb754d57c5c09c2358050160c6d98f09223f6db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04b18e4d9019a4a4ffcef81f0f2268040050c141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04b18e4d9019a4a4ffcef81f0f2268040050c141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
