)]}'
{"id":"openstack%2Foctavia~751111","triplet_id":"openstack%2Foctavia~master~I806d220236ad741b638ad9315537334f2d923031","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I806d220236ad741b638ad9315537334f2d923031","subject":"Fix amphora failover when VRRP port is missing","status":"MERGED","created":"2020-09-10 21:57:09.000000000","updated":"2020-09-21 23:30:19.000000000","submitted":"2020-09-18 15:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751111-1600441280633-e4987785","meta_rev_id":"07c28179357379fc6deb197983796821bb59c62e","_number":751111,"virtual_id_number":751111,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 15:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-09-15 06:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-09-18 12:52:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-09-18 12:52:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-09-18 12:52:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-09-11 18:18:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-09-11 21:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 18:18:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-15 06:19:56.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-18 12:52:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-18 15:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b9cf980a98250a9904bd2a396a83ee9b409adfd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 21:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15644235693beb6bbd60058a1d840a0e254f3f0f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 21:57:31.000000000","message":"Patch Set 1: Workflow-1\n\nThis needs more work, so WIP","accounts_in_message":[],"_revision_number":1},{"id":"352dd1a49ef6ff21349c765aac8e5c4351e93adb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 22:29:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a17916f23621b50b0c0f93c6e6375c922b2383c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-10 23:50:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa0e54065075e3495fb421a58332411b6394c82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 02:58:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7598e28b8bf24ff6af2e574e9a2342ae : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59fde3354cd74d78832fb9bca0a22f5f : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2648295ec694061bcd94b3e45f0ad60 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74ebcc7ec3de4673ae93353bd620a2d8 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa2e675534fd45dd853e511a663bdd33 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046d4f87dd064b30a17e5722e119e277 : SUCCESS in 21m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76b081aad4ab4170949019934d3477aa : SUCCESS in 4m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/56e1c8ee432442a58317713d972e7e88 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c4e084fea7c4209874535976e1beb00 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7acba4f9511d497ebafd55bb93ac277f : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/845c8df5d1424cf0ba7e0c45c40f091f : SUCCESS in 46m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ebabd820aa842a4a7b59ed01743870a : SUCCESS in 1h 45m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67e5518d65f846408cd051226762ea49 : SUCCESS in 31m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/74b8deb80c5e45a281336538e2f3bdde : SUCCESS in 43m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aacfdfdf50be4bef88eea55e8c487e9b : SUCCESS in 1h 02m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d97789406fc74528bf6dfc1619d6b480 : SUCCESS in 1h 25m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0fc45842f2ef4147a3de22641df0b665 : SUCCESS in 2h 05m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a8fdfabd456d4e318a27c846f4727a8d : SUCCESS in 1h 37m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/33751f5e082249f58573cadcdbb218b7 : SUCCESS in 2h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1ba7752274b04ae0a5d240615ad094b3 : SUCCESS in 1h 41m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7ebf2202bcc9422895e9c40b74a4aba4 : SUCCESS in 1h 50m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9bba55ac016c48c5b8b553b47995a6a8 : SUCCESS in 2h 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa8c6d88ca5c777d428176d47be89333037966fd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 21:56:24.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"d623037fe8a117509c8dc85560da9eab0ecb32d6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-15 06:19:56.000000000","message":"Patch Set 3: Code-Review+2\n\nSeems to work for me.","accounts_in_message":[],"_revision_number":3},{"id":"094d07e02232f0df7c4febb0e5426f16229840d8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-15 07:04:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cd56f62f6a27d0659cf456150295f02fccfee7a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-15 21:13:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3c61980f4bc9f760936647bab1ded22e6bd49799","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-17 12:11:36.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nGenerally, I am okay with this change as-is as I see it solving the reported issue. Just commenting again to see if I am completely off or there is space for improvements.","accounts_in_message":[],"_revision_number":3},{"id":"b6630c2e7448b6c4f935916dd954e19526524b49","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-17 20:54:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"393aea87083c0e22e9ce49bd59be2615b48e3b5a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-18 12:52:50.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0f01c78bc3a34187227107f19ccbed1a3a4efc62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 12:53:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb08c5d04b7449e94c7d83e6211dec1c1bebff7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 15:01:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"21011879e5dc4d773e5e71d3c59671ddbb9f1e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 15:01:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da602b3979d14038b163aaaed3ed9d61 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b3b838b9614e298eb615e465a5f48a : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24b6df38c9e643f4a120fa0af743e30a : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b356de1c048645a19c75e07c10185533 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a62114b3b3437aa2258d6dfab2cb06 : SUCCESS in 21m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b96cb77415648ee9828bb6dc976405a : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dba926059ec4424da376fec4d368409a : SUCCESS in 9m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca91106115794dea8a259af91ef4e385 : SUCCESS in 1h 37m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26c063e3dbfb46a6b49c807ba7d2545a : SUCCESS in 2h 04m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ece65bbde2a142feaba5fd97c2d4b421 : SUCCESS in 40m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5f7f8f38565744ebabcff9df396e45e6 : SUCCESS in 39m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5449487d70c14197a04dd3d8cf099e29 : SUCCESS in 39m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dbd177587f4b47eb91ec8f94dea82290 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4d93a77d9e6ae81fc0f08ae99ca599e6c596b00d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 15:03:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fead76d4865e4132af280d280def8180 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/895f6cf54ae6434da21faed8ea98f804 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3},{"id":"07c28179357379fc6deb197983796821bb59c62e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-21 23:30:19.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 2dbd8f696d79671a11f7a5a43e282720f8983f31","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"66f3a63f442bb790dea5c529aa56a462d9515363","revisions":{"4a246108ad0d74c5c39ed3569528cda452bbc0e2":{"kind":"REWORK","_number":1,"created":"2020-09-10 21:57:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/751111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/751111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/751111/1"}}},"commit":{"parents":[{"commit":"9a732565e94e1d34067b3b737131a163be0da370","subject":"Merge \"Refactoring amphora stats driver interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a732565e94e1d34067b3b737131a163be0da370"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 21:54:09.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 21:57:08.000000000","tz":-420},"subject":"Fix amphora failover when VRRP port is missing","message":"Fix amphora failover when VRRP port is missing\n\nIn an amphora failover situation, if the VRRP port of the other\namphora in an active/standby pair is missing from neutron, the amphora\nfailover may fail with a \"port not found\" error.\nThis patch corrects this issue by allowing the amphora failover in\nprogress to complete even if the other amphora is also in a failed\nstate.\n\nChange-Id: I806d220236ad741b638ad9315537334f2d923031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a246108ad0d74c5c39ed3569528cda452bbc0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a246108ad0d74c5c39ed3569528cda452bbc0e2"}]},"branch":"refs/heads/master"},"4b21dcf2b9f5d221b88675e75170042cc3c707a6":{"kind":"REWORK","_number":2,"created":"2020-09-10 22:29:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/751111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/751111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/751111/2"}}},"commit":{"parents":[{"commit":"9a732565e94e1d34067b3b737131a163be0da370","subject":"Merge \"Refactoring amphora stats driver interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a732565e94e1d34067b3b737131a163be0da370"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 21:54:09.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 22:28:21.000000000","tz":-420},"subject":"Fix amphora failover when VRRP port is missing","message":"Fix amphora failover when VRRP port is missing\n\nIn an amphora failover situation, if the VRRP port of the other\namphora in an active/standby pair is missing from neutron, the amphora\nfailover may fail with a \"port not found\" error.\nThis patch corrects this issue by allowing the amphora failover in\nprogress to complete even if the other amphora is also in a failed\nstate.\n\nStory: 2008127\nTask: 40851\n\nChange-Id: I806d220236ad741b638ad9315537334f2d923031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b21dcf2b9f5d221b88675e75170042cc3c707a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b21dcf2b9f5d221b88675e75170042cc3c707a6"}]},"branch":"refs/heads/master"},"66f3a63f442bb790dea5c529aa56a462d9515363":{"kind":"REWORK","_number":3,"created":"2020-09-10 23:50:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/751111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/751111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/751111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/751111/3"}}},"commit":{"parents":[{"commit":"9a732565e94e1d34067b3b737131a163be0da370","subject":"Merge \"Refactoring amphora stats driver interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a732565e94e1d34067b3b737131a163be0da370"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 21:54:09.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 23:50:42.000000000","tz":-420},"subject":"Fix amphora failover when VRRP port is missing","message":"Fix amphora failover when VRRP port is missing\n\nIn an amphora failover situation, if the VRRP port of the other\namphora in an active/standby pair is missing from neutron, the amphora\nfailover may fail with a \"port not found\" error.\nThis patch corrects this issue by allowing the amphora failover in\nprogress to complete even if the other amphora is also in a failed\nstate.\n\nStory: 2008127\nTask: 40851\n\nChange-Id: I806d220236ad741b638ad9315537334f2d923031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66f3a63f442bb790dea5c529aa56a462d9515363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66f3a63f442bb790dea5c529aa56a462d9515363"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
