)]}'
{"id":"openstack%2Foctavia~751379","triplet_id":"openstack%2Foctavia~master~Ib2a80afb9223da2c0af6f6add250ffaaa20a316f","project":"openstack/octavia","branch":"master","topic":"fix-udp","hashtags":[],"change_id":"Ib2a80afb9223da2c0af6f6add250ffaaa20a316f","subject":"Remove haproxy_check_script for UDP-only LBs","status":"MERGED","created":"2020-09-11 18:37:15.000000000","updated":"2020-09-14 21:30:43.000000000","submitted":"2020-09-14 20:46:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751379-1600116418591-309c905f","meta_rev_id":"f2bae1b41c70bceee88b969a4c6168886f9be7a7","_number":751379,"virtual_id_number":751379,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 20:46:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-09-14 15:10:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-14 15:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-09-11 21:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-09-14 15:10:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-09-14 15:10:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-09-11 21:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 21:35:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-14 15:01:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-14 15:10:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-14 20:46:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eeb36db2f5f8c23caa78b86f752f5a68a6daa9eb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-11 18:37:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd2f2d6dbe8b6db2c14bae4e167f5ba575b953fe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 19:16:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07cc9f96c022dba670252c9a1d91403571dac871","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 21:35:40.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"6365d8f4e15481a822e9f49b8547548e0dd284b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:54:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b966a3be2b34d0db7c1bee229b0c1d9 : SUCCESS in 7m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3fba1ddd450458187b70c98ff6516b3 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6119aa6b67420f8b92e1dd632db3bb : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/037e5390f962412d95e5a81c66f38283 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/088eb90fbf644535bc78f0c1ff7c96a1 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e0a4fa186b42c998e5cf0fb2d3c4c1 : SUCCESS in 20m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95f6716fb8ee4bd1b39b1ea5157ff003 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/023d907f1c8343d188848982cbfb09b8 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c60790b7fed34ac48db365894ea9f488 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddfaf74e7e7c4b81baa7db0fd07293a2 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b3311a82b2e6445899e6640eea39619e : SUCCESS in 48m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a8e8dc40d7b4817b2a2ae4109250c96 : SUCCESS in 1h 51m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2cd4647acde4fefb61f05c258b2b954 : SUCCESS in 43m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/db3e1b8fcb834dc08b49be602c8dcf7c : SUCCESS in 58m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c601be0867e4e95ab3acd9d4c7cd98c : SUCCESS in 59m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08e12cb0bc174b608eea0da92fedab1d : SUCCESS in 1h 23m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2522d4078cf4103a3ee2a34c95b1d2a : SUCCESS in 2h 00m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d868b135a6d147a3930c2df89689bb27 : TIMED_OUT in 2h 34m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75243e8254384d50b9426ab85d9a4959 : TIMED_OUT in 2h 35m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d1c54a6cdf8f4643b28b0e62c4727a98 : SUCCESS in 1h 57m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ec0f53982b4f4d53beda46dcea930c25 : SUCCESS in 1h 56m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/027213bd4f3a4d2a8136b3c60fcec5a4 : SUCCESS in 2h 07m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02f7b5953b52c6aa19d5a9b0644c234ec0e429cd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-14 15:01:59.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e81d16b9f4a0ae86ababd4ea86fcd3f10ccd9876","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-14 15:10:16.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c83fccfb81b95dc0a2cb489fae8357db8c15e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1f36d7ffd2cc40e8dda33b0191d756aed0acddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:46:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0edec653a374c6292532855f15317bc : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f8eb69bffd4314a30fcfd6aa2c268e : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/473283dd769041c197388562331f063c : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a70c9e11fe0c424aae3511eaa2cf2e55 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e701acfc8fb400d9756afd8bc99c97b : TIMED_OUT in 31m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a81651336b11441a8ff392682c761c7e : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e1264192cc04fbe9953d6bf20e02898 : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5756a6f39fc4774903a17ab5146dddf : CANCELED in 28m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0accfcd4d82e4446904ae06e83636285 : CANCELED in 32m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/acc77e276abd4ca1937ec7da0c4a5e52 : CANCELED in 30m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/de0c58cc89e0463f839d2eaa4014b468 : CANCELED in 32m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2faf2e6afe7c4d46a71335e5ecf7f90f : CANCELED in 32m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43732014cad341009814dd5a22a40c13 : CANCELED in 32m 41s","accounts_in_message":[],"_revision_number":2},{"id":"67942304a24725e0e412986f2cfec27c6ee34a12","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-14 15:52:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d157ae335195ea3bcb28c043e3d5ad2739f27ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:26:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f234a16082544fa5a893aa22a373332e : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d99c1a69cb5e48758ff3f9793608cee5 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b50550f60f334d4c995bfa81b289f430 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51a70bead7fb42c48378a1a41baa74a7 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf3c134726c472ba88ca69f95a2f87c : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be94f73a39b243beb7aaba12dbdee88c : SUCCESS in 18m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e0d9cc1cd6f48448fca43e9be73e92b : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/81a432a1a73f4acfa8d6679a85e9fd00 : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/792f3b435a4741a9a4e2865bc70f344e : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ac10990c398466d8309dba9aa51e832 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c8b99984f17b43188d533bd5ede03fc6 : SUCCESS in 42m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c302e6cc6f1b4585aba71c2558c9c54f : SUCCESS in 1h 47m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9390e3be720146f8bc4d26ee86c0f288 : SUCCESS in 41m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/548eed7250d34aa6b57a54a5e96e2282 : SUCCESS in 39m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fb2f11c41884ad3a713db58fae1052b : SUCCESS in 36m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/670246a91124414b86e3d5f29ac6d9a1 : SUCCESS in 53m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/877e579f0f8f460d977b457412a96d88 : SUCCESS in 2h 01m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1225c4e5939b4943b20a6cacbba9269a : SUCCESS in 1h 54m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc58292c467e49ffb32b6b956626d7fd : SUCCESS in 1h 37m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/79da037903f04d46b67bd1b26449f34b : FAILURE in 1h 51m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f3811e59b7014a7c90c87de664c0aaa0 : SUCCESS in 1h 50m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f074dbe2ef014feeb721ae0aed59b3bd : SUCCESS in 2h 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8167908f7f30661f2e8e742ca9a91f8055bbb74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:26:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81a52a5e350b81b4765499bc822e56a2efa0419b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 20:46:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"389a03442b76b5a8aa98ca05e4362901831d11ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 20:46:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c8ee7229b57425eaae77031881ad551 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00271a53d8e540dbb738e797c1fdebed : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc83e5bba05049d585975065f1704c11 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0c07b79752b421baf36ccb014e737ce : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5d38b0245c4f7abceab5095d7a1d99 : SUCCESS in 26m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6621fccf0174ee581e2110f6b29f9cb : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bb44676d7b643c8835ae512fb17f4b9 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/daa00046a9a34f3cafec9a9268d4963c : SUCCESS in 47m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac27a2dd458f45c59ec84bf260a5c5c2 : SUCCESS in 2h 18m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2c751138e1c4e33a374066d2bd97238 : SUCCESS in 48m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64399ce333194ae08101e4d0319aec8d : SUCCESS in 44m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee5a3282b1b647c69922a217d1c6cf39 : SUCCESS in 34m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/738abe39f73043b3b28c8028606b1efd : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":2},{"id":"54df0f8e431b14622e28201296aefdde5b20ba95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 20:48:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d31c45958154fdaa6482655ed611796 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a26f3e8ee0094dc28fe1c909d50c7e66 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2},{"id":"894af0877abbffd1ab7e5303881bbaecb68451d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-14 21:30:43.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 3106eea20d00ad7d7cf3baa9923dd0d1b0db3c5b","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ce9add9d9c7319104fa3ca2e6852b8166234f06","revisions":{"e9d5a43b903ca05e7fa78a3549d48324dfba9345":{"kind":"REWORK","_number":1,"created":"2020-09-11 18:37:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/79/751379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/751379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/751379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/751379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/751379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/751379/1"}}},"commit":{"parents":[{"commit":"a446c04f779faffa8262d2f1d2c9ce7e506d6525","subject":"Revert pyroute2 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a446c04f779faffa8262d2f1d2c9ce7e506d6525"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-11 18:36:00.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-11 18:36:00.000000000","tz":-420},"subject":"Remove haproxy_check_script for UDP-only LBs","message":"Remove haproxy_check_script for UDP-only LBs\n\nThis file is created when no HAProxy listeners exist so VRRP will fail,\nbut needs to be removed when a UDP listener is added so that VRRP can\nactually correctly start.\n\nChange-Id: Ib2a80afb9223da2c0af6f6add250ffaaa20a316f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9d5a43b903ca05e7fa78a3549d48324dfba9345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9d5a43b903ca05e7fa78a3549d48324dfba9345"}]},"branch":"refs/heads/master"},"0ce9add9d9c7319104fa3ca2e6852b8166234f06":{"kind":"REWORK","_number":2,"created":"2020-09-11 19:16:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/79/751379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/751379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/751379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/751379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/751379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/751379/2"}}},"commit":{"parents":[{"commit":"a446c04f779faffa8262d2f1d2c9ce7e506d6525","subject":"Revert pyroute2 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a446c04f779faffa8262d2f1d2c9ce7e506d6525"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-11 18:36:00.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-11 19:16:23.000000000","tz":-420},"subject":"Remove haproxy_check_script for UDP-only LBs","message":"Remove haproxy_check_script for UDP-only LBs\n\nThis file is created when no HAProxy listeners exist so VRRP will fail,\nbut needs to be removed when a UDP listener is added so that VRRP can\nactually correctly start.\n\nStory: 2008133\nTask: 40859\n\nChange-Id: Ib2a80afb9223da2c0af6f6add250ffaaa20a316f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ce9add9d9c7319104fa3ca2e6852b8166234f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ce9add9d9c7319104fa3ca2e6852b8166234f06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
