)]}'
{"id":"openstack%2Foctavia~751863","triplet_id":"openstack%2Foctavia~master~I46ba7798322f24255ab7d49db4b4df886e92cbf3","project":"openstack/octavia","branch":"master","topic":"story/2008138","hashtags":[],"change_id":"I46ba7798322f24255ab7d49db4b4df886e92cbf3","subject":"Fix image build when installing agent from package","status":"MERGED","created":"2020-09-14 16:57:23.000000000","updated":"2020-09-24 01:52:20.000000000","submitted":"2020-09-24 01:48:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751863-1600912134707-72df3f54","meta_rev_id":"cf8ff628da44f9da7a6985c677701ea8665f8794","_number":751863,"virtual_id_number":751863,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 01:48:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2020-09-23 12:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-09-23 23:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-09-23 23:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 12:50:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-23 23:43:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-24 01:48:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b3555994229df67969b4c7d6fca20e5872ba0d7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-14 16:57:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"912f799a4cb319942283b5c98c1948733f81f8ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 19:33:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c993dc0b549f4223a21e7c638d2072a2 : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0385c42391e4d648246f22754eeeb04 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c37f24a1a8de4e9581e22e27ab22bcca : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91120581b627464280aaac5dac8930fb : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1881885eaf174f8990d5d197665d0a0a : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ddbbfadcb2c4e9895d5c3425bb0d2d5 : SUCCESS in 23m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/52395ce9a1f14907ac0befbf9c5d4019 : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/732646f12c0b4708a971fe1c50ab8f40 : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e0b9447e83a347edbd5935d79d0f1b0a : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a6452d3769e4c11838f398375b4b7e2 : SUCCESS in 42m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bd38e411d1a4d7c8aba69a1362ceb8d : SUCCESS in 2h 25m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/484041fbf0cc4611a7b1def46e310af8 : SUCCESS in 42m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/84202f7ddfe74f779306c0b10780749c : SUCCESS in 44m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80d650ebedd9413fbf1919c4dab75474 : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d88e6675369c47ae85c73d0cb53cbb95 : SUCCESS in 49m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd4f792f95b74bada58acdb54f52d334 : SUCCESS in 2h 22m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ef91e18a19d48178596319e80873fc8 : SUCCESS in 2h 14m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f469865e9b7c410b8d2740e745b76f13 : SUCCESS in 1h 38m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5751cd55c69944fea39e88ddf8ea6bc3 : SUCCESS in 2h 05m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ae445ef0f68949309654e29f2c3fa9c0 : TIMED_OUT in 2h 33m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/56a03bfb9e6145ea95d27234cdb9b9c8 : FAILURE in 2h 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"462a406ce360e658ab60b5f0c875eade7c73fd8d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-23 12:50:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"733c115a3f2037bd7d54312709fc024f24caea96","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-23 23:43:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM\nNot a huge fan of that syntax, but it is valid.","accounts_in_message":[],"_revision_number":1},{"id":"ddd4744a5811a0ef5247187b6878052c7942fd95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 23:43:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8100a7e971a0f0848fbabf6c17e9c03bd75a3920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 01:48:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e3127594daa4b64997dbc296c923882 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3506c2fbd1d84cb2a1bfeb336803c852 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/428089da166c4c65aae78b8c25692c17 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ebc16baab384f0f80fd2d93a07b99bb : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f6326d9fcc450eb5ceb3f0522afe94 : SUCCESS in 10m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf0e24e6b1a94214a1c3f3e011be41d8 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc5bd2ec8a8d43119a33396cfa2a0e05 : SUCCESS in 1h 37m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4751a6de64340c59cb51debff484aee : SUCCESS in 2h 02m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/717bb3b0b1044ca8b6aa713b2a1f137e : SUCCESS in 39m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/16f1e8c5040c4af0843ea51da1c71422 : SUCCESS in 36m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70f7a65f4bb441d58a461c8f10840aea : SUCCESS in 36m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e98cdf8b8fe048aa984d3d928105a8e5 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1f4791f92bdc941e7584e74058c5ba1a7f728f91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 01:48:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b625dd3bd4f3a5140b9b6595614f70dbe6f780eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 01:52:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed6529e45884277905e74f042528aed : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16f9d2ebf8e0ccbb8c26396c8eed8ad3f33bc66e","revisions":{"16f9d2ebf8e0ccbb8c26396c8eed8ad3f33bc66e":{"kind":"REWORK","_number":1,"created":"2020-09-14 16:57:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/63/751863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/751863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/751863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/751863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/751863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/751863/1"}}},"commit":{"parents":[{"commit":"b400620a24e5b19f2551969e5fc02f829d79c9dc","subject":"Merge \"Change tls_ciphers init in Pool API classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b400620a24e5b19f2551969e5fc02f829d79c9dc"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-14 16:51:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-14 16:56:36.000000000","tz":120},"subject":"Fix image build when installing agent from package","message":"Fix image build when installing agent from package\n\nThere is a bug in the octavia-lib DIB element where it runs install\nactions that should only be ran when the amphora-agent is installed from\nsource. This issue causes the amphora image build to fail when\ninstalling the amphora-agent from distribution packages (option -p in\nthe diskimage-create.sh tool).\n\nTask: 40870\nStory: 2008138\n\nChange-Id: I46ba7798322f24255ab7d49db4b4df886e92cbf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16f9d2ebf8e0ccbb8c26396c8eed8ad3f33bc66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16f9d2ebf8e0ccbb8c26396c8eed8ad3f33bc66e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
