)]}'
{"id":"openstack%2Foctavia~751918","triplet_id":"openstack%2Foctavia~master~I5f45e6f0f15bca201c3ef63f2e42a183510f5977","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5f45e6f0f15bca201c3ef63f2e42a183510f5977","subject":"Fix requirements.txt for Victoria","status":"MERGED","created":"2020-09-14 22:44:22.000000000","updated":"2020-09-18 18:58:41.000000000","submitted":"2020-09-18 18:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751918-1600455394794-276e733d","meta_rev_id":"aa60bdeba3b67ba3c7ce483ce4b2ddc93eec277b","_number":751918,"virtual_id_number":751918,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 18:56:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-18 13:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-18 12:54:18.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-18 13:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 16:07:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-18 12:54:18.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-18 13:28:36.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-18 18:56:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1a4582e5c94d180205e144494406f1a7cbd7a84","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-14 22:44:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83912238e5337154ae556cc48965cc20ee600699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 01:21:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/10e11c9bff764150a5b248ce3c206c6d : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/add5ceba7fcf434a9b9d1fa87a086250 : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88e0f1df85b046269d3b2a27d6bc29cd : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82c8b5d71ea4c9cbeeb5a30eb8efe95 : FAILURE in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9898cf43391c46819e92708ebe5bfdc0 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/789500bfd4fb4fcfbbdd638f6b8218b6 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018d187754234376b2a68c5d340e8ebe : SUCCESS in 17m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/69fff3da18944119a52d6d6c6daed847 : FAILURE in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ceaf7c5f97c94904aac5324556909cdb : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f49461e5dab4e15aa2f40a87a0b7684 : FAILURE in 3m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66e93b2c425148dcbe4256d58cee5162 : FAILURE in 11m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/024c72011f804896a6ad769a1655972a : FAILURE in 11m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cdcef610fc4a4ba7b209b996e932eeba : SUCCESS in 42m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bb3834acc2ee476c8e119240d2f4e9b3 : SUCCESS in 53m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a79b15e521c84c149191157be6e1895b : FAILURE in 11m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ef8843b7975f42459bfc40eb8bc8d0d7 : FAILURE in 12m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec2004c0bbc04a268c6c439c9a2d5a9a : FAILURE in 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c818715e64214f5994f7fe78a0880f03 : FAILURE in 12m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2c82292c4c9e4221a4530b4e3d32819d : FAILURE in 2h 31m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f67537703e2740719cb9780535b0fcb4 : SUCCESS in 1h 36m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3549915cfb204d77b275d3bf712f29a6 : TIMED_OUT in 2h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cc712543122151c12a1bf8925fe279fbe9c94c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-15 15:31:16.000000000","message":"Patch Set 1:\n\nrecheck\ninfra cache should be fixed now","accounts_in_message":[],"_revision_number":1},{"id":"a05c81f882e6d3da23349c7ff97b4fb4a7b97071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 18:04:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa2abfe4553b4ae09dd5dc5626896e7a : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72231c4fcddf4c72a46d9f135e4d4bab : FAILURE in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ce17a0a675145879a00f64441ec265a : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b392c993aa4b7d9b9970c7c282fd4c : FAILURE in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa71fd9ca4324749a35737f116d4cd45 : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a419dcd930c54ac78983eef545695a19 : FAILURE in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805b85c554464e9f8dbbc224defc5583 : POST_FAILURE in 18m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/51394c39574f466f9409a40bac89869b : FAILURE in 3m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4dfdb6099828422186cc8c22e63e8162 : FAILURE in 3m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ffaa1cd977940c5ad4b52e9cfa7957e : FAILURE in 3m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48d1f541d2084cbdbe61e94f523ea510 : FAILURE in 17m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45532361037249cbbbd4dbf6ef0861f1 : SUCCESS in 2h 00m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3987620208e7471e8133fededab6d69c : SUCCESS in 46m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7107a62d287c4a32b8adb3acff548b1a : SUCCESS in 51m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5812d1dac60448eaa7eaa1d25db3d6d9 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bb3ce8fd5502421991814145757732e9 : SUCCESS in 2h 16m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49e70c082fc840ed87c4151c0ecb42f4 : FAILURE in 44m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/baadce495ce7482b9bfb67c3a320d287 : SUCCESS in 2h 11m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a333b6a041bd4fb8904fdb83288139bf : SUCCESS in 1h 50m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2b5132fae2474dfeaa894416b69c852d : SUCCESS in 2h 19m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/226d03e2f3564048b9ee157b3f2d704b : SUCCESS in 1h 55m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60d06ae0a7aec5c3365b049c3b9c29710d5a8593","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-15 22:39:04.000000000","message":"Patch Set 1:\n\nrecheck\npypa should be fixed","accounts_in_message":[],"_revision_number":1},{"id":"f8bcf0626d51f7ab0e188acf5bba9391495865de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 00:44:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5224c0302ee7472fbd67eab55a6e9c03 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/332f5983c3b04a58ba8db753940336ae : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbe4582565404ebaab0b96ad5379f2f4 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f5e550ab7749cc9deb03bf54a78ddf : FAILURE in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f34d3b76c44d9881d156852ddd73b7 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8276437cf3a1440f8b735502f3b625c8 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a76c65237941abac7662fb02d4b67e : FAILURE in 14m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/96e4f4ca8e644d75a1c42e8d39a648c9 : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/307e24c302ab449581553e87cfc2dcdd : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/edc93374e91242638197fdbd5e0c63a9 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1ec5bcab36649a58116c6ad5567cc23 : SUCCESS in 47m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/694c0d856ab24e4f8ee131bb45f3ec07 : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e1175fb15b74db081ebacb59cf5d0d5 : SUCCESS in 36m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/46880fecd61a48478312bf5d57b10f56 : SUCCESS in 36m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dc2b22445a74557a3d4d841080ac090 : SUCCESS in 34m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/127e9ac2c394413f81925d723e969716 : SUCCESS in 1h 46m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ef107f2072941f2800436fef80fbbc8 : FAILURE in 44m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/61a980ca69904773ba8c5b4cc8f8ab18 : SUCCESS in 1h 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/57b5ae2eff1546469322687dfbfa94b5 : SUCCESS in 2h 01m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/69a715689b6c4c1786eeb4d90669e9fc : SUCCESS in 1h 37m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ba5b98476f11426e9b6e85233880a5f3 : SUCCESS in 1h 51m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c89cb635ffbc889c99e49ccb159f7637411b670d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-16 11:52:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1eabd65ea04ca8aa5d6096d3d8d7522e3264fb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8f3a2b5ef1f45608f5ec0900ce6db0f : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f98271eeb7042a382cdbbdd24e925a2 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f308a6b6ed84b48894084273196e7b1 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3870f5147547338945faffec2eea61 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f176a2f0feeb4fed9160c38ab02c85f1 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761c8f8f8cfb4fa4bc3cb4d4d814fd39 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e448d29451ad481cb0c98e4719da1d11 : SUCCESS in 20m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c0a599500b2d429bbe31aac25ce083de : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cb8dc1a7e494c73a0b7a04a26823af4 : SUCCESS in 9m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab8a03348fed421784e230979e556754 : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a316abb6815d4b57b5abbc952bb99f5b : SUCCESS in 51m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/924ec839b4794b2c8cc062931ef7fedf : SUCCESS in 1h 56m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6e64c8785d7f4bdf9aba014ec5020e16 : SUCCESS in 47m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0d179e454fa14f9a8a418f71a421fc6f : SUCCESS in 38m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9ca84a4814e4e29b0a8a18455b6a1f2 : SUCCESS in 45m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/88130debd6684672b03060475e2dc553 : TIMED_OUT in 2h 50m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/32be4fcaa5a54f4581f5b1cd8d3fa752 : FAILURE in 1h 15m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/18d335f0639546b8a39eab3b362cb182 : SUCCESS in 1h 57m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7b489d781f6249b9b5940937b388b57d : SUCCESS in 2h 04m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/420094df86fb4275a5fa4d8914a1c942 : SUCCESS in 2h 25m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ac8affcc6b834443a4fd69d3bc962bf2 : SUCCESS in 2h 08m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84a544159c46a61df1cfe24d0906afb3ec719457","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-16 15:58:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"615a706a815ca5305923e959bb9fd8c1f155e275","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-16 16:07:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df04ad0aa443f0bac12b03b06ea1925b21fa2ad5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-17 05:47:54.000000000","message":"Patch Set 1: Code-Review-1\n\nFile lower-constraints.txt should be updated as well.","accounts_in_message":[],"_revision_number":1},{"id":"27b45c43fc62b48bde30ecab31690107de86295f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-17 07:15:04.000000000","message":"Patch Set 1:\n\nIt is already correct in lower-contraints https://github.com/openstack/octavia/blob/master/lower-constraints.txt#L123","accounts_in_message":[],"_revision_number":1},{"id":"180817461b1178d31057ae91f7e96ffad2781064","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-17 07:25:07.000000000","message":"Patch Set 1:\n\nPyMySQL and netifaces should be removed in lower-constraints.txt. I ack there are a bunch of other libs in l-c.txt not being used, however IMO best practice should be to keep both files aligned when touching them.","accounts_in_message":[],"_revision_number":1},{"id":"5e018648af954aa6ffe17bf03aeb5a0164e53b87","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-17 12:08:29.000000000","message":"Patch Set 1: -Code-Review\n\nHm, I missed that, it make sense.","accounts_in_message":[],"_revision_number":1},{"id":"6724bbc225f0e700eac4d51ae9038e7069437da9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-17 21:38:04.000000000","message":"Patch Set 1:\n\nlower-constraints impacts dependencies of our requirements. Just because we don\u0027t have it in our direct requirements does not mean that it should be removed from lower-constraints.\nIf you download this patch and install it in a virtual environment, you will find netifaces is still installed so should remain in the lower-constraints. However, I did see that PyMySQL is not there, which I know we use for sqlalchemy (see the connect string in octavia.conf) so this raised a flag that we are not using the proper \"extras\" for sqlalchemy. I will fix that and you will find PyMySQL is still needed in the lower-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"6c9377d9d4e43ab62fc84cdfc264fafeb4ea1dc4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-17 21:45:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c77cbdc4db2f3e330ce784d9679b683808dd605d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 00:20:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b079846a70434013809c99bf083e559e : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00ac721790734a428558434241a5a2c9 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4c367fdc2394b2ba0f4bef18e75eb10 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97addd5f89cc43718bd9f51639b7b19c : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7496df96097e48d2a2cd298199d948af : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5eaae553b154ea59749f4333f38197c : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14651964849f43b1ba18a83369d0887c : TIMED_OUT in 31m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc9660ce19fc4948a1442d2f7e0102b6 : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e4c3c8238b9b4f089800201c4d645eeb : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97447d2c67cb496e9664a520fdbbfab5 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4a82d03ce914d6b9752668dad605ec4 : SUCCESS in 1h 39m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f2fe12b268d4f6aa4178e917d3e36ec : TIMED_OUT in 2h 32m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3a15ea3c6b44de7a911e6e24b7f9525 : SUCCESS in 1h 00m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aee32c8d71164695af478c788f58fe21 : SUCCESS in 1h 02m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/560e33502bb646d4a0ae14bfb2d93eb3 : SUCCESS in 48m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c6b981ed3b7b4464a767999669a4aee8 : SUCCESS in 2h 11m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46231ee255384a48bab90dc4ef41d3e8 : TIMED_OUT in 2h 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fde752c263344f67b160ac32b27d30bc : SUCCESS in 2h 14m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/18c13ec5cb174e6d8e9bc98f2934e4c1 : SUCCESS in 2h 07m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2841cd4654c84716a7d007f419a38a38 : SUCCESS in 1h 57m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ff003141502b450eb2ed3ea1051aa190 : SUCCESS in 2h 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0a9c271ece7547e960e52eefa088f0fa9a34f06","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-18 12:54:18.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck\n\nThank you for the explanation.","accounts_in_message":[],"_revision_number":2},{"id":"7951f71e8afcee9c28afe13ffda7f82221fcf54b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-18 13:28:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8760e468593a31b6c53ff82d2a5dc2704c725cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 17:03:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1462e6ce125b43d9b5f68d90d9be551e : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66a9857563e744b483c2b1e769c15a98 : SUCCESS in 13m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f5498cdc2ab425986e80b1b191788da : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a013591fa2742aba378a13b3ba9029f : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9733d8fb1dc43a5a3df141ff28dd835 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cf904b149f64bd4b920e86740ce27e7 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d062c75b2ba4c32b68def34747ed0b8 : SUCCESS in 19m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/742ac4e0268f4bf4943fc6e83317b72e : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f03e68589df34725a4f1fba0d2c5237e : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/434df2d6238140649cb860a2d4042ed9 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07632046d498425a99736677675574f4 : SUCCESS in 1h 42m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8c76c5268ac4477adc7b48d3b050a91 : SUCCESS in 2h 04m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78e465b981e443c1be3680912cd39490 : SUCCESS in 38m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12f443733ef6419db5e3f74b516ee3ab : SUCCESS in 35m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8e276e511e840c595ea40a67d57750c : SUCCESS in 1h 32m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f7f543eb2bc4614a87dd5eefde127d2 : SUCCESS in 1h 51m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c19245fd7e734ff1a15d0836f084a215 : TIMED_OUT in 2h 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/059888ed76e84e2689a46a770cec7bf2 : TIMED_OUT in 2h 35m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3fc89bd1192d444ab0b10b0ab82705bd : SUCCESS in 2h 16m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9d240abb520f47b49534e507e175bb0a : SUCCESS in 2h 19m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3472793d58824f8293a029ddf07d7e64 : SUCCESS in 2h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"997950df63f56c157769b156d2f299d726251828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 17:03:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"42f1b5ed9fef07d1c022aad32960748135877cf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:56:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7b4090987d300c1cd98755268ebfdcb088161c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:56:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1b4e3ef4add49b6a7d7362aa9752ba9 : SUCCESS in 2m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a750eaaa030c49f58b173fd12991e842 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af27365e69443349580e3eff79cc7f0 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7acf294f0784028932944211ab619e1 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f495979df30411daa7f7c2fe7b4ca5c : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5f4c2a06d74625b5f86c5263311d65 : SUCCESS in 20m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c2fdf6199c64c1da7a9fa9479759c00 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45cbfd5a581244d5b5b0e12945ad9662 : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eccf69ce472947949e9de1425531e2cf : SUCCESS in 1h 50m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06616f81e1204a4eb2658581ca59a875 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/de7f337cde264ccf98a75d08015a54b0 : SUCCESS in 55m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/567ad0f2abc3411c8396d29c0bc43025 : SUCCESS in 55m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f078090d22fc49f6a84c7ae47f055d57 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":2},{"id":"aa60bdeba3b67ba3c7ce483ce4b2ddc93eec277b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:58:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d38a29ecb2343b48760fc190d48ad8f : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c8c4f18fd9c01824294d5f90af57d2e01213064","revisions":{"ab9f2b514514da5fb04ef1607fe1aa12de4f5dbb":{"kind":"REWORK","_number":1,"created":"2020-09-14 22:44:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/18/751918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/751918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/751918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/751918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/751918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/751918/1"}}},"commit":{"parents":[{"commit":"a28bc2a55a1e61a45c15b0f9bfcd057369423bbb","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28bc2a55a1e61a45c15b0f9bfcd057369423bbb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-14 22:43:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-14 22:44:21.000000000","tz":-420},"subject":"Fix requirements.txt for Victoria","message":"Fix requirements.txt for Victoria\n\nThere are a couple of extra requirements no longer used and one\nthat was missing. This patch corrects that.\n\nChange-Id: I5f45e6f0f15bca201c3ef63f2e42a183510f5977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab9f2b514514da5fb04ef1607fe1aa12de4f5dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab9f2b514514da5fb04ef1607fe1aa12de4f5dbb"}]},"branch":"refs/heads/master"},"6c8c4f18fd9c01824294d5f90af57d2e01213064":{"kind":"REWORK","_number":2,"created":"2020-09-17 21:45:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/18/751918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/751918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/751918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/751918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/751918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/751918/2"}}},"commit":{"parents":[{"commit":"a28bc2a55a1e61a45c15b0f9bfcd057369423bbb","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28bc2a55a1e61a45c15b0f9bfcd057369423bbb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-14 22:43:38.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-17 21:45:18.000000000","tz":0},"subject":"Fix requirements.txt for Victoria","message":"Fix requirements.txt for Victoria\n\nThere are a couple of extra requirements no longer used and one\nthat was missing. This patch corrects that.\n\nChange-Id: I5f45e6f0f15bca201c3ef63f2e42a183510f5977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c8c4f18fd9c01824294d5f90af57d2e01213064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c8c4f18fd9c01824294d5f90af57d2e01213064"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
