)]}'
{"id":"openstack%2Foctavia~752845","triplet_id":"openstack%2Foctavia~master~Ic15f61c64dc0a2caf8cf8d48a26532cc8c2f9fc4","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ic15f61c64dc0a2caf8cf8d48a26532cc8c2f9fc4","subject":"Bump hacking min version to 3.0.1","status":"NEW","created":"2020-09-21 02:50:18.000000000","updated":"2020-09-21 05:12:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca23cfb161d046f245f87e103a8c129d59395df0","_number":752845,"virtual_id_number":752845,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-21 05:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 05:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c30742b5d5ae92311f9059dab34ca867e73f268a","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-09-21 02:50:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca23cfb161d046f245f87e103a8c129d59395df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 05:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bcbb56b00f854a86b37ddbf3abe54931 : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/42e68645ba7c409e91a01dae8933a482 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cea29d2eaf6a4f109022d8d3bd20ef66 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97fa264aa21f419cac4b200872328199 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f99af75e9345f3b32c47c9fa826a29 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/031a8fe433674d09a766b706b96b5fa3 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f60b87fd8e471091b7372a51a7c2ed : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a8b6a6fc284290ab7ce7a85f627dcd : SUCCESS in 18m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bbb45436fec48fb9015e37603bb0890 : SUCCESS in 5m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aaba93a525a94c23a065d6a9ed3de8bc : SUCCESS in 4m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/336bcdeb956f465398147ad3337bfb5e : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/057497998cad430e91f4117bae0a1af0 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ed893c27e4a41f0adb9933662e5bc18 : SUCCESS in 1h 50m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/415b3e94b13b4b45bbeef755b7141e3d : SUCCESS in 1h 48m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d65c651187d34dfe82d5314a26bc7700 : SUCCESS in 38m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dce4bd788ff442579f1fdbc8e7aa60b4 : SUCCESS in 30m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/997285d1c72f413b97ba70ff3ea45795 : SUCCESS in 37m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d348a3370a7a40b8ae3ac423b72c2f89 : SUCCESS in 2h 11m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96048d5fd7f84d5ab07cda9b05fc44c1 : SUCCESS in 2h 05m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c10d24d48e7f4006b5f383908af3cdcb : FAILURE in 1h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ad7e596dd8ab41a38433ff499ddcd4c3 : SUCCESS in 2h 02m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4cdc9c1a51504145a849150b5a40dee3 : SUCCESS in 1h 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/73bc296ce22245dcb38d5db060bf8579 : SUCCESS in 2h 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a82764147739c3cbfe4cbf226009aecaa8e137dc","revisions":{"a82764147739c3cbfe4cbf226009aecaa8e137dc":{"kind":"REWORK","_number":1,"created":"2020-09-21 02:50:18.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/45/752845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/752845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/752845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/752845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/752845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/752845/1"}}},"commit":{"parents":[{"commit":"e45ecc2111efa7c860299809c8d9d629dd9e47ee","subject":"Merge \"Fix requirements.txt for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e45ecc2111efa7c860299809c8d9d629dd9e47ee"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-09-21 02:39:07.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-09-21 02:48:36.000000000","tz":480},"subject":"Bump hacking min version to 3.0.1","message":"Bump hacking min version to 3.0.1\n\nhacking 3.0.1 fix the pinning of flake8 to avoid bringing in a new\nversion with new checks.\n\nbumping the min version for hacking so that any older hacking versions\nwhich auto adopt the new checks are not used.\n\nChange-Id: Ic15f61c64dc0a2caf8cf8d48a26532cc8c2f9fc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a82764147739c3cbfe4cbf226009aecaa8e137dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a82764147739c3cbfe4cbf226009aecaa8e137dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
