)]}'
{"id":"openstack%2Foctavia~752853","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I82d877a41d7339891a388e26d149fd65cb61c727","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I82d877a41d7339891a388e26d149fd65cb61c727","subject":"Set Grub timeout to 0 for fast boot times","status":"MERGED","created":"2020-09-21 06:03:33.000000000","updated":"2020-09-24 06:32:21.000000000","submitted":"2020-09-24 06:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752853-1600928827633-fd99345f","meta_rev_id":"d01f163039876cb1e4f7ac072d45c86f67b8b333","_number":752853,"virtual_id_number":752853,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 06:27:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-22 08:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-24 04:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-09-24 04:10:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 08:01:35.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-24 04:10:12.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-24 06:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87ba9b7b6dfcab4013cb33ba18bdd21f7621e163","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:03:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a194c98ac153728ddfc3ab4a6ede94a65ada0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:29:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf2f13b5465b4fbda8e4cfe956dd3315 : SUCCESS in 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61da84c9ed9f4cccb4cd226e85d267e7 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb8dc0fc09a439195a2ef95c5cc7c10 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/741af991bd8d4087a6e8fa06e10146e5 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/229edf55f6b94fb4b21511d4e5e5b117 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d113681176c94728a70302afc743d92c : SUCCESS in 22m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d955cde491744107a525a4cfdc56f45e : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/497c1b49d98c4f73b399aad9e44014f9 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/895527c1f21c42639d419535d7701860 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7fd6913ff954830997b77b4887a48ee : SUCCESS in 1h 38m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9c185b328a4442096a14959b7e94c94 : SUCCESS in 2h 05m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/560f0f5002574439a12c2ea27bb98d96 : SUCCESS in 37m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ce68f141eedc4dfbbf46b02fc92f27a6 : SUCCESS in 51m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b6bf7e0c0c244c8832ffebb5dcc9143 : SUCCESS in 45m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19143d16a7be49778d7ddc981869602e : SUCCESS in 1h 19m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26760812698148dbaf83b037a4d9fdcd : FAILURE in 1h 53m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c0e89d1f9ee4d20bf8a8c9949e7dc88 : FAILURE in 44m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6760782b5575437b8a04d53619f9d240 : FAILURE in 1h 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d4506be4fee60327d6863753972f6ad67eae1cd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-22 08:01:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83af4ea4b286bcb1ed2bbb3ec5fad8de6ea814f3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-24 04:10:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4943c185cfb91c8a5748d764e0a023eff025e5db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:12:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd704a9caa8fbf6c0ba7f56fce9be0c5cbc63922","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6ef07c26b8a357b989089c2ab3150a83ccac9a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:27:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7b599906d5b452293ee755b3129213a : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2193ed3e6a774c2c95a72ceb278e9077 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee088fcfe4584c76981d1613f6303938 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e5a89b918d44942a759ca2faffea173 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933ac046d8a24dec92e26cda1fafc197 : SUCCESS in 12m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d670f470ccd44424afff218f5aa13ac7 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/805873c1c94742a791f9b7a20f3f7502 : SUCCESS in 1h 44m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01ac4e29ea09472bbea10b75246db59a : SUCCESS in 2h 11m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/297e8b18095741eabbbcdd3263dd50d9 : SUCCESS in 39m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e190476f9e64326b60ead49a64dc6ad : SUCCESS in 49m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3ab7349738a47c1ae97634ebc61f974 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acac5d3d95584a85820b50591dd6015d : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d01f163039876cb1e4f7ac072d45c86f67b8b333","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:32:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca11f4a439b4d858bd7bde0f15542f5 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2fb5e0d3fc57f9ae30f994ea34583d9083b94d9","revisions":{"d2fb5e0d3fc57f9ae30f994ea34583d9083b94d9":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:03:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/53/752853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/752853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/752853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/752853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/752853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/752853/1"}}},"commit":{"parents":[{"commit":"43db34778771ee1af758ca56e9f862724c03e86b","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43db34778771ee1af758ca56e9f862724c03e86b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-12 11:00:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:03:08.000000000","tz":120},"subject":"Set Grub timeout to 0 for fast boot times","message":"Set Grub timeout to 0 for fast boot times\n\nThe default Grub timeout is 5 set by the bootloader element. Setting\ntimeout to zero helps reduce boot times.\n\nChange-Id: I82d877a41d7339891a388e26d149fd65cb61c727\n(cherry picked from commit 611af6008a3b76c0e1686758442b0afd5c56a6f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2fb5e0d3fc57f9ae30f994ea34583d9083b94d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2fb5e0d3fc57f9ae30f994ea34583d9083b94d9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
