)]}'
{"id":"openstack%2Foctavia~752856","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I0154b83cb64952844a28895721694d3e2ff82be2","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I0154b83cb64952844a28895721694d3e2ff82be2","subject":"Update devstack plugin","status":"MERGED","created":"2020-09-21 06:07:20.000000000","updated":"2020-10-23 15:17:51.000000000","submitted":"2020-10-23 15:15:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752856-1603466100830-f4875fba","meta_rev_id":"e9aa448d40ecffd38832b1b07b15696a81600895","_number":752856,"virtual_id_number":752856,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2020-09-21 11:50:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 15:15:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-10-23 10:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-10-19 13:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-10-23 10:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-09-21 06:07:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:07:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-19 13:55:22.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-23 10:51:13.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-23 15:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b19ede09bc6e02134e423fca2053fad029a9aac","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4db4901c86cc2867f44710b0297d11ef7dc643a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:48:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/576870d054c9484bb91d8aa8c7a2dc89 : SUCCESS in 10m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2f13c5404394340aff909e9728d5673 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5da06016b2a045818d9c10ce0c3b8e7d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e280dd029803490e86513f7b8d647f67 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94aa565b67df4e5a86a4ada1aa9dfbfe : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23d713087d14973bff8de2a2071502f : SUCCESS in 20m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/23bd34b93d7a4db48d55b998f74f7d2e : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4067b643b5bb4262a247f0beb58c20f8 : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50f1563104544cf38233be5f96b74ef0 : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4263f1bb763947bb8bffa3e2afd82b72 : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64d0692ee1284fe4aa3f39a1439917b5 : FAILURE in 1h 52m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/88ec61519e2844acbaabce235d2cae21 : SUCCESS in 39m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e8c81e15f53f47559ed2fde3dbf1ab77 : SUCCESS in 37m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e519ee05d5e740e19be39bfa7142ba57 : SUCCESS in 38m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/490279c724054ff0af61dc2fbec31bc9 : SUCCESS in 1h 19m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f218967c9f54bafa4dd4f65a27703e0 : FAILURE in 2h 25m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71d33d18443e4656a04de8667ad99eb1 : FAILURE in 49m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a7457792a60e465cb0e6b96ab7d3c417 : SUCCESS in 2h 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ca18df498a683a5811b8811fc1af44ea9a8f70b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 11:50:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c1cf08fc9c6ee0e2c97291f00faf860be672d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb78441040e74821b327b9b4f0a5897a : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec5093830e1948e0a422545af08c1fcc : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b4edfddff5420eaadee00a06dc4995 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3da9c3d83d2b481e85fcb44a232232be : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c012845ae8d4b7c8747bc248bbf00c0 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8b444db9344d68ba6315c37dc9939a : SUCCESS in 22m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/55b0815a6e214d379b7771bb928367f0 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eba4ebb17c874c7cb6f2ce5fe8a78344 : SUCCESS in 8m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75ae7259784b46e8b202363f443b881d : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cffac6d5868f4816915440eaf4bba7e8 : SUCCESS in 1h 37m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dbdf74faf9942868692a34b788fdd2d : SUCCESS in 2h 08m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2db97b84cfd24712ae2a8f368a0d1bb5 : SUCCESS in 38m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d50d5d6ae0904d4d802e501ff6224a53 : SUCCESS in 34m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99447abc507b4f4f92691512e3797226 : SUCCESS in 58m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d737b58c336b4795be5bb6b4314c842d : SUCCESS in 1h 20m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05cedea84bee4c6daccfedba24db5d2a : FAILURE in 1h 50m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9122f5ea49e446b6be7ba7729b6f714f : FAILURE in 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8ed770e5341c46f8851637aa9fdb1b46 : SUCCESS in 2h 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1da1604c6abcae358848e59285f98a925aa2a56","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 13:55:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"862c18292f281f3e04ef889449487471e5d19df2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-23 10:51:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62b6bb54d9a5e5bc8cbc7199ffcc4618b09a2234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 10:51:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b3b11d667e3abdcee1cd2b1319bf8ffae58f5f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 15:15:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8421c6239094382b8892f64c3f08fef : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ebbc20667b4ca88a0b8bc4845cd144 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b78b27769ab4529a28fa308b1ef5ce7 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9a5c8f4873643b09f68cb168b20ab07 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6b4669a05e41099c196f569a9b1a7d : SUCCESS in 15m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc0926833430489dadb9be8ce4425b9f : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/75deee23b71d4f3dbbe37913981fc400 : SUCCESS in 2h 00m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e87207580baf46bba928c489ccffc24e : SUCCESS in 2h 21m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea513c639a0f486b998aa6ad07608b68 : SUCCESS in 42m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d66091ae4aaf46aaa8e1d282b687a1c9 : SUCCESS in 47m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/765ee03476094e2288df314e3f2629fc : SUCCESS in 48m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6dd5e996b6a441b38f8a0b8b6a569566 : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8a675d40f7147429b08e15a7a07866fb24c64439","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 15:15:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9aa448d40ecffd38832b1b07b15696a81600895","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 15:17:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16c05e0082e4bfbb5c10c658db13055 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5613958a5011a07878c548ccd33b24cdb511e914","revisions":{"5613958a5011a07878c548ccd33b24cdb511e914":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:07:20.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/56/752856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/752856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/752856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/752856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/752856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/752856/1"}}},"commit":{"parents":[{"commit":"43db34778771ee1af758ca56e9f862724c03e86b","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43db34778771ee1af758ca56e9f862724c03e86b"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-08-11 08:19:21.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:07:07.000000000","tz":120},"subject":"Update devstack plugin","message":"Update devstack plugin\n\nThere is no reason to create internal and admin endpoints in DevStack,\nmost other services have stopped doing so a long time ago.\n\nAlso use the global SERVICE_PROTOCOL variable as default for\nOCTAVIA_PROTOCOL instead of \"http\", this will make us automatically\nuse the secure API access when the tls-proxy service is enabled.\n\nFix the instructions for using the devstack plugin, too.\n\nChange-Id: I0154b83cb64952844a28895721694d3e2ff82be2\n(cherry picked from commit dac24cd7ec9d9d6d6a89a9ab3e099e93854807ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5613958a5011a07878c548ccd33b24cdb511e914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5613958a5011a07878c548ccd33b24cdb511e914"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
