)]}'
{"id":"openstack%2Foctavia~752859","triplet_id":"openstack%2Foctavia~stable%2Fstein~I0154b83cb64952844a28895721694d3e2ff82be2","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I0154b83cb64952844a28895721694d3e2ff82be2","subject":"Update devstack plugin","status":"MERGED","created":"2020-09-21 06:08:25.000000000","updated":"2020-10-21 20:52:25.000000000","submitted":"2020-10-21 20:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752859-1603313545901-018b9b8e","meta_rev_id":"00b94caad3f8e3ec48785f060aba83dc789dbcf1","_number":752859,"virtual_id_number":752859,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 20:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2020-10-19 13:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-10-21 18:16:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-10-21 18:16:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-09-21 06:08:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:08:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-19 13:56:39.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-21 18:16:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-21 20:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4403bb9188170e82dcb832ad0508bafe4570d35b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:08:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e9e1d8824b7a0f98f9643555aa4869f0285f844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:25:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03f38f99a1664241b1465bb9ecfa4b26 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d37a1a8da1f42b4a2e4b8e45618feb7 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d9e23c788c43238287b827c670f8e5 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/551b36739eaf4decba45933e3c722ea6 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f269901b031e4f56a158ebec09cd6e0d : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb12b9c8f5ed4b30be5246f55e283c6a : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec9605b1b914702aa6616da6d6dc35b : SUCCESS in 6m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6885d3b6900a478bb2923a3cd331ec8d : SUCCESS in 1h 23m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b8afb78b46064bef84f34fa2d225356b : SUCCESS in 1h 09m 52s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f1eb465153d4c36aeeece69c8949417 : SUCCESS in 1h 09m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/60eb8d233c86495f81ffbcf1ed1caeaa : SUCCESS in 1h 08m 30s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f3dd4fc437f547dd8399c504af7b4f58 : SUCCESS in 8m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/391390ffae47401aa6d4b83662b5c14c : SUCCESS in 9m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c980ab22a5334da18d0552957bde4152 : SUCCESS in 7m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6b89d06411045ff9f2dba9a0d23973d : SUCCESS in 1h 08m 14s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7e16d0fc10c645449bb686014059dc9a : SUCCESS in 54m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38ebb16c37bc47aa8a0463b21ced39a6 : SUCCESS in 1h 20m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1c2d5e97679b40a383677cea1b1bbcca : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39804a9be1dc40cdac24e07bdf508535 : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/aef4b46b7a4f4c1d895e57fd351ab3d3 : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ab62c7d36384e9e9167f93b7c7075cd : FAILURE in 16m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36f0f276d5a641d6aa84454baec7ee7b : FAILURE in 43m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5c47848ec004a86aa48e83ca4a5341c : FAILURE in 46m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a302946df23d469e838c4ea9a7e74b8d : SUCCESS in 1h 10m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3cd73e939e54037b9f15cee0e95da4f : SUCCESS in 55m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d7ace566455f7b80aa1dd20c5d62e7508a2daf4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 13:56:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d81907e15dc2f68d8528ef77a95f4585913b106","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-21 18:16:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNot sure we really need this backported, but it won\u0027t hurt anything so LGTM","accounts_in_message":[],"_revision_number":1},{"id":"cf85c40b3a44187df8d301e694e72f8ca6e60d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:16:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02461f10bb8366c75b2c7ca0ac0d400f1ad20e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a3b7113b12d404eb9f5c20a9c39643f : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ef974b52cc4f0ca79007cbd1ed79e5 : SUCCESS in 9m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e53804e1054c4ead91edf041a23a2636 : SUCCESS in 8m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dd86177316344ab2a53e521b0abc78fb : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d95fc627cf4548b495ddcbb68be7f1 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48dc14e1fa354f51b8a4cd593be884c8 : SUCCESS in 9m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fc7876c35d564ab89c7e1fb66c47dc8c : SUCCESS in 10m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c3065dec5578409a8cda5989f9d7a088 : SUCCESS in 11m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a099ca54e9b409fa8f3e98bba3af873 : SUCCESS in 9m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f937e06754a48009a48e3467cc329ca : SUCCESS in 1h 06m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/09e8fcaf5ac44953bedea8e1ba82c64c : SUCCESS in 1h 04m 30s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ca50da724ef245c9b8f456f31a765189 : SUCCESS in 1h 24m 37s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/42df06468c764bf19e0507e0d7bbb36c : SUCCESS in 1h 33m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a68d3b019e814647b3739ac25e63f277 : SUCCESS in 1h 05m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/29471ca583db42748996ae523b8906ae : SUCCESS in 1h 29m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17e07f3eac6f4bfa9ad63d90031303c0 : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":1},{"id":"00b94caad3f8e3ec48785f060aba83dc789dbcf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5fd3eb16a96dc781b2a8308b83b9ecabdad6696f","revisions":{"5fd3eb16a96dc781b2a8308b83b9ecabdad6696f":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:08:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/752859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/752859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/752859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/752859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/752859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/752859/1"}}},"commit":{"parents":[{"commit":"2ac913cd1946218a42c3209df29cc5e74e811634","subject":"Merge \"Fix API sort key for complex columns\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ac913cd1946218a42c3209df29cc5e74e811634"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-08-11 08:19:21.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:08:15.000000000","tz":120},"subject":"Update devstack plugin","message":"Update devstack plugin\n\nThere is no reason to create internal and admin endpoints in DevStack,\nmost other services have stopped doing so a long time ago.\n\nAlso use the global SERVICE_PROTOCOL variable as default for\nOCTAVIA_PROTOCOL instead of \"http\", this will make us automatically\nuse the secure API access when the tls-proxy service is enabled.\n\nFix the instructions for using the devstack plugin, too.\n\nChange-Id: I0154b83cb64952844a28895721694d3e2ff82be2\n(cherry picked from commit dac24cd7ec9d9d6d6a89a9ab3e099e93854807ff)\n(cherry picked from commit 5613958a5011a07878c548ccd33b24cdb511e914)\n(cherry picked from commit e21028d6d1e8e2cb63711e42251edb4af6e0bad0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fd3eb16a96dc781b2a8308b83b9ecabdad6696f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fd3eb16a96dc781b2a8308b83b9ecabdad6696f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
