)]}'
{"id":"openstack%2Foctavia~752860","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I86add89ed7ea474dcb24962547c78460bc2e260c","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I86add89ed7ea474dcb24962547c78460bc2e260c","subject":"Add missing log line for finishing amp operations","status":"MERGED","created":"2020-09-21 06:09:16.000000000","updated":"2021-01-21 20:25:07.000000000","submitted":"2021-01-21 20:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752860","meta_rev_id":"e32d51de50b9e705d1a45fe330f86e9fccd6491a","_number":752860,"virtual_id_number":752860,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-09-21 11:51:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 20:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-01-21 09:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-01-15 15:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-10-19 13:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-21 07:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-01-21 09:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-09-21 06:09:16.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:09:16.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-21 16:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-19 13:58:46.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-21 07:36:57.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-01-15 15:21:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-21 09:29:03.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"49b13c30022b377bb59ec42a5aefbc648d14fcc1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:09:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c7882afb43343497b73b2a678c5fee8225e1192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 11:27:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac31fd602cdc4768932c6d581155cad6 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/730f50abf96e48b1b2c827d9bee7b27d : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68f85ce8426046f6aeabb4fb06693998 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6807ffff55de4476b0fa19c127d8dcd1 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b69255f8be89444eabc6482cd8c08044 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2afdf0ff67544ba96be76bceef7eec8 : SUCCESS in 18m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8b1552ecdd09423c9d9b6d7dae4471c7 : SUCCESS in 8m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/176df5a0ad8e464586a35e7537a4e6ca : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f377f9c162e345b2aae5e443fdf83076 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28904b4a9a7e47e7adaae1a76471c669 : SUCCESS in 1h 44m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72055a44ca9043c1accc1348d707f509 : FAILURE in 2h 18m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6a61e39e4f14e3985e8adb92ab3f1e9 : SUCCESS in 44m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ffcaec588be14e898dee850dd53c456b : SUCCESS in 43m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3adb44df4b56406cbfef896889344e8a : SUCCESS in 38m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23dd25699a4d4949ac523476025fd307 : SUCCESS in 1h 14m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2178f0a0f3e426e86b60fbc3bb062d0 : SUCCESS in 1h 48m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cb235405baa4ef7b330e3edc906d66d : FAILURE in 41m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5fcbaabef224137a2bdcf8be48a73a6 : FAILURE in 2h 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0184703603745c8ee5ec53850a89585809a89a5b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 11:51:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd67fb760d5908a6adf35138fbe999eeb3a34577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 16:18:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1143e494a5b49bcb2f0db14694c6475 : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9733fc03d8bc47639bd83a808a1dbb93 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2cdb318c1b34b079ba6a8ac84390727 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d9416c7e704de8a4a66f03896b48ca : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b45d7f351bc4567b50ea2a966228588 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9176239ada2a43e780c0de39b6f7617e : SUCCESS in 18m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d40e4b2f47304131b4d466fa55a070fa : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/54687838410b4e6885a165e6bc6499a0 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f5c86bcebce438e94e350239c4f3323 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a707101783b344bd9e0e8cf51f027633 : SUCCESS in 1h 30m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2f5e746814a48f6a0162951c9f57334 : SUCCESS in 1h 54m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef10eb8c09984a3fb78e03f51c78eebe : SUCCESS in 36m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/28d4087212844052bfdb5e6f1fc28a06 : SUCCESS in 34m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f2ef27a9a3b499287f5bd3ce8906a09 : SUCCESS in 40m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5fddae467ef64b06b52a48c4d30a88ee : SUCCESS in 1h 20m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b681bc33c0147d2bd5fe24b21eb415a : SUCCESS in 1h 47m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e0a553b4c0848799bcdc379ebe3b5ec : FAILURE in 45m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd6fef9678d04e30b6a59a07ff09328c : SUCCESS in 2h 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d2595a371cd8b4dd4d8c37f608fb27a0cc6266d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 13:58:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7e255a89368b6bab64fd8f39f2b5a219dfb634f","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-10-21 07:36:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"29b3ab3d41791243976e9d873c039cb34f709709","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-15 15:21:57.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"682be98579d410757ea1cf3db086fd565e0ad4da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 17:32:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb20e213bc684d6ea4f9ca6928ac3a81 : SUCCESS in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57c5fac913144f5f8c11f0e7d33e6db4 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c77c702439964b02b1728b2276b36d6a : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eefca4331fa84459ae241091060971d5 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bc7fd7ca2234641bfdc342b3a1328ff : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13bcfdb42e0d442ea0730a78e46c0898 : SUCCESS in 13m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f120248af1d34e1f833358337fa2faa1 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d589674621c465cbbc851dea13644bf : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f182206c9743473ebb89ff66dca1de0c : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f365d9074c24b7090a01d71dcf7267c : FAILURE in 1h 17m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b141d1f43141487ea2af6b614eb34b03 : SUCCESS in 2h 04m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7a26a7df6b24cab90a9729630ffc1e2 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a1bb023f0e7e442886bb9bfeb16a409e : SUCCESS in 37m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b0611dea29c41079279fb25c0920cb0 : SUCCESS in 40m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/360af42aff1742518f6a96cbdbf9d9b6 : SUCCESS in 1h 25m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8620beed31f47e49bfcc988b11e2bbe : SUCCESS in 2h 06m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/092551ccc209480196afaf17f800bae6 : FAILURE in 56m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d7fcc2156875419aa3254add7da74eef : FAILURE in 17m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9baaadac2c6f9cea457bda5400fefec8eb669de0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-21 09:29:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18a5e5d8596628d877556eb3fc55b7d65ad68dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:36:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f54871d60c0d4f8b8bdcdea2a7491227 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a1f31b8aaf141c894c4f9053a198135 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e137f9df414ecfbb97c69e400f849e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a34f9278184d3eb6221048272399f7 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27bd758b4d854295b4aa737ece69fa0b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f583c1326ceb4d009ba1808c39270ce3 : SUCCESS in 12m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2cca8da9cc034550a2e3a10df5fbafc5 : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/63464f6e71ff494e834f36dc9b07a72b : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e53bdd278c384dd6b910ee6f91f80436 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ca69d53a5a74f468c7fafb9e4d87a6b : SUCCESS in 1h 20m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69659f6c72d441bdbd0e9b8035baca9a : SUCCESS in 1h 40m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cb76505808f43a6b4225b1310060fa3 : SUCCESS in 43m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/40c8851de5ba40f08070c0cb57333234 : SUCCESS in 46m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/074da6dac73f4e70a4efed0e647c27fe : SUCCESS in 41m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ff88bfdb1cf4c25a11392b6586d0e86 : SUCCESS in 1h 13m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b97a878b2f374f938eed0d2d3740e98c : FAILURE in 1h 56m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/93322929cb2341619ab740131de3672c : FAILURE in 53m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55115a03d9dc4da4b92ccbc78560f80b : SUCCESS in 1h 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e8b45cd4de4802021cc3c96306afb283e5471eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:37:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff572bd1898d8ee4f2dd02f9f1bd0411176a7bed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 13:30:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d5f53988c89493581c5222c94b8da1f : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f41535218b34c338d052d4c68b5da1b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28228dbf20dc41179320c069224e5f60 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f45c2a86cd5e4ae59e7ab9d6bbe85850 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e39c521cd84f2483321ff9936ad4e1 : SUCCESS in 10m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a556daf0a9c5428896e7adf3806dbd3d : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4b57044407f49c7b82a692a500597cc : SUCCESS in 1h 26m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87844cbbce064cbf88c441a0512774dc : FAILURE in 1h 47m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d402512ea6da4bcca621d5c75138631d : SUCCESS in 39m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b1170334ef948a58e33dce6ced40e41 : SUCCESS in 38m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d725bdb7ca24abab42abb17a755b0f0 : SUCCESS in 38m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0bfe5fd78e08438492cc89ee26482b33 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7bf2c30120755159efce242f6bb92ab3452799e1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-21 14:55:21.000000000","message":"Patch Set 1:\n\nrecheck\n\nDon\u0027t know if we merged something that broke this.","accounts_in_message":[],"_revision_number":1},{"id":"b68330e00316f3cb9b7dd0c752134cbaf1f94e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:27:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5d8b9cea3b1428589d03e75db17a662 : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9225e31789a47f89e6c2b2ff919ff21 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc0f053a7f74e7e87a1970cef536db8 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/173e015bd1d5471493ece7c4fd2e2446 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95784f72f0bd4776893a0d29c2aeba23 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4a14cde1d14a02a7a835fbe6ec56b2 : SUCCESS in 12m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/367931218a4b4940bab81b69be0095cf : SUCCESS in 7m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/786e1de017d4428081fc12cfebfd2c43 : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1f4f0f3062c4e539ec788d1adf401b1 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2825c06d69f94d02b52630bcef2fc5c5 : SUCCESS in 1h 12m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3079b29641f949fe85160c3f47b1a219 : SUCCESS in 1h 45m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03bdf8e341014878a95f12e04132f65a : SUCCESS in 42m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ecf8f3c3447c4e7891d5160b00962157 : SUCCESS in 1h 06m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e50390eb94b45e59b893d5f5cdb0e11 : SUCCESS in 42m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b823459848548498b51ef2c8951e15c : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77055119610844319352dc5c182a4aeb : SUCCESS in 2h 08m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a6db2e61537417bbe7bbf004ab3b000 : FAILURE in 48m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39b296f3a4e34a989e8f200a13ee3e4c : SUCCESS in 1h 59m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8aff529db5286ad4b460639624786f04e394ff74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:28:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4ea5b584323fa8528137c367f21c9b27d1ec1bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:22:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92569e42f5ca40acac94178621c90c3f : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d5e5d05dff45a19164b3c398340559 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ea2c9e7338c49618b9e07db3d5569fb : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a3d2d90da6d436d96cd14351e78254e : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73eaf166b8a141978c6d7a440980d4f5 : SUCCESS in 11m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d598c200fc68495ba190089d3fa05223 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4402c4bb8c874328b3e1c32dd86d14e9 : SUCCESS in 1h 18m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11784663ae1648a18b7d48b4a2d4f48b : SUCCESS in 1h 46m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a12a71d014645d59753fefda910576b : SUCCESS in 39m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/58459761b0164d4ca9efe9508af405f3 : SUCCESS in 39m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a458612c84574bd28096c9251b2a3141 : SUCCESS in 39m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72987a55bf1240759caef653c91ecdd0 : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"91c43910ad0eed96b83c39503a3cf2c6ba00bd20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:22:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e32d51de50b9e705d1a45fe330f86e9fccd6491a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:25:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea037a81ea04d219687f95097e74693 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0afb17c434d1c385eb3b7f4859060de35a0d6f67","revisions":{"0afb17c434d1c385eb3b7f4859060de35a0d6f67":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:09:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/60/752860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/752860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/752860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/752860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/752860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/752860/1"}}},"commit":{"parents":[{"commit":"43db34778771ee1af758ca56e9f862724c03e86b","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43db34778771ee1af758ca56e9f862724c03e86b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-11 23:36:49.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:08:59.000000000","tz":120},"subject":"Add missing log line for finishing amp operations","message":"Add missing log line for finishing amp operations\n\nBoth amp config and amp cert rotate were missing Finished log lines.\n\nChange-Id: I86add89ed7ea474dcb24962547c78460bc2e260c\n(cherry picked from commit 6b182b50cda83bc83aa5cc1edd72e3dd6609de17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0afb17c434d1c385eb3b7f4859060de35a0d6f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0afb17c434d1c385eb3b7f4859060de35a0d6f67"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
