)]}'
{"id":"openstack%2Foctavia~752862","triplet_id":"openstack%2Foctavia~stable%2Fstein~I86add89ed7ea474dcb24962547c78460bc2e260c","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I86add89ed7ea474dcb24962547c78460bc2e260c","subject":"Add missing log line for finishing amp operations","status":"MERGED","created":"2020-09-21 06:12:55.000000000","updated":"2020-10-21 20:54:01.000000000","submitted":"2020-10-21 20:52:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752862-1603313544603-4e3cf6b5","meta_rev_id":"fcdc197ade707baa8b1e488354ad4d3be45a14fb","_number":752862,"virtual_id_number":752862,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 20:52:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-10-19 14:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-10-21 18:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-10-21 18:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-09-21 06:12:55.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:12:55.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-10-19 14:00:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-21 18:14:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-21 20:52:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e0ab48236cadf839fd9dcd320acfb7f70f75d8b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:12:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f97cdbeea9ec60958e625c59137adc9ea9080df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:56:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55660ed2a9824b14a5597bc336d4246e : SUCCESS in 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d981a67e53b41b8a1797b3beefd52af : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/423f510879024c468bc11fba17a8ffbe : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f517b07bded4a7d98cb2d8850ad7498 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/54b83b721b3645009d4867004754de5b : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ba4dd3c2854c68ae1975e622c7e189 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cca2bfb109e406aac2331628fa964ca : SUCCESS in 7m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9db2b0516aa94a54b4d76a8999e906d9 : SUCCESS in 1h 19m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0a65222ba97b48dea6a7b72a3961cdd5 : SUCCESS in 1h 19m 55s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d40695a5919c4adea50a60ec4c69c910 : SUCCESS in 1h 18m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/db6e228e1d724e23983a72f92d233dad : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6e0d1053c31c4edf82cc795deea41a2b : SUCCESS in 7m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/32f23c6555dc410c97a0fd91c3d7996f : SUCCESS in 9m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e40c61c1853d41249149ea3bbdcf6ce2 : SUCCESS in 7m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c22224a6925c44cba2676c658be359f1 : SUCCESS in 50m 42s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6a0846899dae4eeebe43ec498d826812 : SUCCESS in 56m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6037a943d69c48809c9a02bb6f1e3b65 : SUCCESS in 1h 31m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/13f0a39800eb4ce4b5af280b014e7856 : SUCCESS in 1h 22m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b59d84a98b374e30b6c6bddd46326d59 : SUCCESS in 1h 29m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/869d73024a5a423e8d04e98c6a30023a : SUCCESS in 1h 19m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/da6663a1a5474f3a927b66602bee4802 : FAILURE in 12m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8a7a74f29d9a4552b96f11040d9c2c1d : FAILURE in 52m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4417f6a67f4743b89a7025671a622bd5 : FAILURE in 54m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ffd69c9700b46e88365edb2963c2714 : SUCCESS in 1h 15m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7a5e92290dd41ed8aeb526720a467d9 : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4deccff2d2eae20e9e8b48a3e19c2831c5db7de9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 14:00:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54dba1bf6a43288a3188d63dbf54367bb6a5e7b3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-21 18:14:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b5a1806168dfd1314489c029c1244ab4a942d724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:14:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2eef836cbb1ec187090d8a480e8747cc51427a34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b61d5009b00f9e5d1b64390d25e69f2a03ddec38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e57f7a9e19364662816fa623f55f25a9 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/081d2894b1b74069a367b7d0a1be4cbe : SUCCESS in 9m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd9708b5ceed479c8a5850cddfdab72e : SUCCESS in 9m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/43a0da0705394ecb99e973c3618ab102 : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/269e95d2966546c699dfd8a264e876de : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f06fae9177b48249a07ef4efb943310 : SUCCESS in 10m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/22e12b078c314703ba523e24a50bb33a : SUCCESS in 10m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/992b653b0b5d4416886318338410cf9d : SUCCESS in 12m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73471ec8c8bd4b81a5d8d1d14e2a8c7d : SUCCESS in 10m 11s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a58c52e9f900487594da40ebca0db823 : SUCCESS in 1h 03m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6228c996c136451b9c644694b63bc795 : SUCCESS in 1h 11m 15s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6b1d8737c2c94ebd8eb1fd3d5cf0bc1c : SUCCESS in 1h 18m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/12900e05effe443cb36a089990417bd4 : SUCCESS in 1h 09m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f9ed6d0a94246f5997c43bd365b9760 : SUCCESS in 1h 48m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/289a6b27cf3b4155a0942d3b1d501f4d : SUCCESS in 1h 29m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e194d4a61467412f844c122a1a636696 : SUCCESS in 1h 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fcdc197ade707baa8b1e488354ad4d3be45a14fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649622d8dd834087a9d233900fcad10d : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0916c77c62822abbafb61a0334187f8e364a7d1","revisions":{"e0916c77c62822abbafb61a0334187f8e364a7d1":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:12:55.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/752862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/752862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/752862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/752862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/752862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/752862/1"}}},"commit":{"parents":[{"commit":"2ac913cd1946218a42c3209df29cc5e74e811634","subject":"Merge \"Fix API sort key for complex columns\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ac913cd1946218a42c3209df29cc5e74e811634"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-08-11 23:36:49.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:12:37.000000000","tz":120},"subject":"Add missing log line for finishing amp operations","message":"Add missing log line for finishing amp operations\n\nBoth amp config and amp cert rotate were missing Finished log lines.\n\nChange-Id: I86add89ed7ea474dcb24962547c78460bc2e260c\n(cherry picked from commit 6b182b50cda83bc83aa5cc1edd72e3dd6609de17)\n(cherry picked from commit 0afb17c434d1c385eb3b7f4859060de35a0d6f67)\n(cherry picked from commit 4aa563325055e05291049118f9cdfa3dd6b1a8c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0916c77c62822abbafb61a0334187f8e364a7d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0916c77c62822abbafb61a0334187f8e364a7d1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
