)]}'
{"id":"openstack%2Foctavia~752863","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I477d2d12c64c1e38e3917333e5d47f612cab93b0","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I477d2d12c64c1e38e3917333e5d47f612cab93b0","subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","status":"MERGED","created":"2020-09-21 06:13:42.000000000","updated":"2021-01-14 19:01:00.000000000","submitted":"2021-01-14 18:58:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752863","meta_rev_id":"0b63ba457140201cb9cff085b0b965d5acc377c0","_number":752863,"virtual_id_number":752863,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-09-21 12:00:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 18:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":2,"date":"2020-12-14 09:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-02 00:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":1,"date":"2020-12-14 09:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-09-21 06:13:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:13:42.000000000","updated_by":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"reviewer":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"state":"REVIEWER"},{"updated":"2020-09-21 19:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 00:09:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-12-14 09:04:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-14 09:25:12.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"1b8ffd709aab66a9af2fc8d82e3d0c0a490f0df0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e05c91fea805b587918f01ee443a9fdb09c4ede0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 11:55:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc2e1bb0b5f43e1b9c11c75046f4b7e : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f4e03ce4ce74929a1f7dbab7d3e5f14 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582cc59a06754b7c94da18c9d7ba4413 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57f1f84a12f544cbaa10b8c0d701c3f0 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a16b01d45d047d39078e78133da0620 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c247f631f24c53b0492d757bcb0583 : SUCCESS in 16m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b479ebf902c443319227f43843322b86 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e92977fb88a646b28858f45b17cf340c : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd7ae151a7394fb8a2473d3b41b76793 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8a6960393c841018bd5c2bd1fc33258 : SUCCESS in 1h 58m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd065fa6541f4de59c25c259a4e6a8ed : FAILURE in 2h 26m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48a881398b7740a9b6b91e97a88d0967 : SUCCESS in 46m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9cd8f71c43734b49b39d48397d0dc730 : SUCCESS in 1h 16m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a359a44b138459881266f3c8bbbdad0 : SUCCESS in 36m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f208f0e427b4b248db60e4832ae655c : SUCCESS in 1h 17m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e58a9063060047f3a00437a7851bf330 : SUCCESS in 2h 14m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ea39cc1b4ed4a5a907a063b0a41cf71 : FAILURE in 45m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d204883313124bc6b58e0c85493fb0a9 : SUCCESS in 1h 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90c08c6aa05d2455f473af50c55940b92ec43eaf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 12:00:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6efebd0dce2870ce2d221f71e4b2f17a321915bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d96b6e93ca4ad6bcfa784237815ef6 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a96a38ca85b74827a764f5bd7106d001 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b188215579447c382ab0b4709238d12 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fdeffc70b37457fa33d35756392f345 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/806da2b2f9a5433e846f065be9887408 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d911faff84dd4963aba8633c50c6c9cf : SUCCESS in 22m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/be810c50ee054087bb11a4ec99404f29 : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5e5f341908dd43d79ff6a6c48ac274e8 : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b63cd4099c149229c7f4cdc5e4f4091 : SUCCESS in 10m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7765928ca499408294f1c114c82cd862 : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b04fbcb727c848f9bc02a7d5d8615270 : SUCCESS in 2h 18m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80cd163838654c6e82ee19eb1e0dda6f : SUCCESS in 38m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/de40af0fa53b42afb775b957fcf9f72c : SUCCESS in 44m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff2c5fd262984939bf34ebcbdf9f0093 : SUCCESS in 39m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3e1c712d74404df4b73eef540044decc : SUCCESS in 1h 19m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d70163cd0c34192b4461666f750689a : TIMED_OUT in 2h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f93ccab1b634528adde1240bed1509a : FAILURE in 43m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/57a1cfe3d3ab4c63b2d05cade474188b : FAILURE in 2h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a84818b8dbbf7853dacdd39b3de65298d0c3914","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:09:06.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"7d730dab283b208b8a5ada1715e81fddb336ddbd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-14 09:04:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd38a61da003bbbbacf0be1e704ff72f63c17eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:04:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"531a48cbe767c9eda77db7c53326f1385623e52a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:20:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3124ce46b52c47aca848b2f0abfaf094 : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f4e8af3f5e4e66aa3ccba6e6083990 : CANCELED in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd1b18e2b034e80a7c8d29e6ca92b3d : CANCELED in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e4555b0b3d948e3bec23ab4229ea375 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f91b5b52dab74b7f8dc9952deb779d45 : CANCELED in 5m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ade4aec89c543ba8c02a4591572f62e : CANCELED in 2m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6143b0881ca64d6d86a86665e4290a1a : CANCELED in 4m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66ab2390ed194103ad1d224d5935ef45 : CANCELED in 4m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/None : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"e4daed9c2df8162558d99bc63d3d6d54d9c18415","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 09:25:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"931ca6ad878f03e9745e7632a619bfebffc56997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:13:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c548def33caa493b80163519d20d9d9c : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c60a1de78ff460e966e3eb79584e065 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5fba158997548c493b5cbe2cc01236c : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce71d65e8a4949b08fa7792e36fbd957 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c839471d1c3435581d562ad22b2a6cd : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc207af54cc45a597e8e70af89e85b3 : SUCCESS in 13m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6bc552a40c854394b90f533065c7a066 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/13f347ec2d7c427e85df8a2ec503a821 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b0a8eda3b6c42f5a1f5945a1a5a7e55 : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cdfcdf024c2545c9b8aa048dabd2c0d5 : SUCCESS in 1h 25m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd608d354a664b339bd078a6a0cbfbb1 : SUCCESS in 1h 19m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/593ce31621aa48de84b70c3071962643 : SUCCESS in 41m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4144a8e2a1974c218c93432a76a49847 : SUCCESS in 42m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/341cb82bc7a24023b5007d9e0d69ec22 : SUCCESS in 38m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/91699098d81e446e82d4f08f3df1dcca : SUCCESS in 1h 20m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb60c95653ef4b03ada13dab05a38059 : SUCCESS in 1h 41m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/def7599279ee477d8b510c5331503241 : FAILURE in 1h 05m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/386b32189dd44b49855f00262c5c993b : SUCCESS in 1h 53m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eead6b8926125c8e98fa112c5c564d232515db47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:14:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf20fcc7351451e0f9e0debf6d957f5437556dec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:58:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a7bfefa90224d1ab2d90fc6071a90f8 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4ac2c7815d49d0a9b4f5a0759f7bf0 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6294ebc714a486abe7a412c1316b449 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22833433a5a447c0a4e92c66e60e7859 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687b94bbf2a24b818f3c5ded59d3b2b9 : SUCCESS in 16m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bf930be48bf4ffc9d1d00c012c1200a : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28b1bfbbd5414eebadfdc484d6b959e6 : SUCCESS in 1h 28m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae5cb0b4fc9641538d32928f50115b26 : SUCCESS in 1h 38m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1557d89d83b14035b1cf76e25a98ce6e : SUCCESS in 47m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a45eeb1efae0400b94875accf14d3df9 : SUCCESS in 1h 34m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcc2cf89a92e4176913ee42ee0e85d61 : SUCCESS in 40m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/22ba744802484052bc8e823246d64849 : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8d6218cfc2baf25bfb2c37089fab9738821a9f45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:58:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b63ba457140201cb9cff085b0b965d5acc377c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:01:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d155003196604be4adddbbae97af206b : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"966adf0f84ae3c67c91f2e5840fa059bd4995c8f","revisions":{"966adf0f84ae3c67c91f2e5840fa059bd4995c8f":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:13:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/63/752863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/752863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/752863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/752863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/752863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/752863/1"}}},"commit":{"parents":[{"commit":"43db34778771ee1af758ca56e9f862724c03e86b","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43db34778771ee1af758ca56e9f862724c03e86b"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:43:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:13:27.000000000","tz":120},"subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","message":"Fix user permission for WSGI configuration\nTask: 35692\nStory: 2006172\n\nChange-Id: I477d2d12c64c1e38e3917333e5d47f612cab93b0\n(cherry picked from commit 6df1ce8e28eb76cb782dc06f913598cfa2ecc156)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/966adf0f84ae3c67c91f2e5840fa059bd4995c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/966adf0f84ae3c67c91f2e5840fa059bd4995c8f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
